)]}'
{"id":"openstack%2Fhorizon~810388","triplet_id":"openstack%2Fhorizon~master~Ie22865995d14fa60c16cf2cea582aa0eec46b65d","project":"openstack/horizon","branch":"master","topic":"django3-support","hashtags":[],"change_id":"Ie22865995d14fa60c16cf2cea582aa0eec46b65d","subject":"workflow: Do not touch dict during iteration","status":"MERGED","created":"2021-09-22 10:43:11.000000000","updated":"2021-09-22 20:38:56.000000000","submitted":"2021-09-22 20:37:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"810388-django3-support","meta_rev_id":"17b6f7e53ddbe0f41ef26ae713f06b077dfd8682","_number":810388,"virtual_id_number":810388,"owner":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-22 20:37:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":2,"date":"2021-09-22 12:09:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":-1,"date":"2021-09-22 15:15:04.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":2,"date":"2021-09-22 15:15:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":1,"date":"2021-09-22 17:58:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"}],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-22 12:09:54.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2021-09-22 12:25:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-22 12:32:01.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"state":"REVIEWER"},{"updated":"2021-09-22 12:32:01.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2021-09-22 12:32:01.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"state":"REVIEWER"}],"messages":[{"id":"b4f8ad050f507663e193c4d7e9e40fe40b619b62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-09-22 10:43:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"10b75831579869e24e2c738c4e860b28247c2cff","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-09-22 12:09:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a66f0b4382ad884c12326cbca269f260c3784c35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 12:25:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/0727c45c8afb49e2b6d27cb397d586a4 : SUCCESS in 11m 06s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/7421fd408648475992521ade33524cf4 : SUCCESS in 6m 02s (non-voting)\n- horizon-nodejs14-run-lint https://zuul.opendev.org/t/openstack/build/70f2faa7838e4e2298cacb0cbf055eee : SUCCESS in 7m 13s\n- horizon-nodejs14-run-test https://zuul.opendev.org/t/openstack/build/3fa1d6c8f3964f29b0bd58cbefc075d5 : SUCCESS in 10m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5085140a3162429c81a139120f1bf40e : SUCCESS in 10m 11s\n- horizon-tox-python3-django32 https://zuul.opendev.org/t/openstack/build/24e2c0a535be4946abb3fb1017122155 : SUCCESS in 10m 55s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/28b1273793064cb1ba1d09dad3acedb8 : SUCCESS in 11m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/987e2fbd37b44f0dbfde037c957fdf8e : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bedca55371fd4b20817efc4a64cd3bb7 : SUCCESS in 10m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/980603345d7a4f8182a94aca0fb3c27d : SUCCESS in 10m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/091b866e3e344351a4d13607c51e9128 : SUCCESS in 7m 12s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/2ea923ddc37f4a5d821b52115622c59b : SUCCESS in 7m 58s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/e07adc72958a4f70b11e5e27aa328c21 : SUCCESS in 1h 19m 50s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/68f9a106d2a44eec92f0bc95f2bacb68 : SUCCESS in 31m 05s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/54aebf59778745aca5a49e324db08ab0 : SUCCESS in 5m 31s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/741a882740124d22a646666b83cacf12 : SUCCESS in 26m 36s","accounts_in_message":[],"_revision_number":1},{"id":"6453b4233dbd1301c5b83199d79d136d73434a33","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-09-22 13:29:01.000000000","message":"Topic bug/1944548 removed","accounts_in_message":[],"_revision_number":1},{"id":"ee1648516a4f3b17bcf650baa10862159f3dbd46","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-09-22 13:29:04.000000000","message":"Topic set to django3-support","accounts_in_message":[],"_revision_number":1},{"id":"c4d57c007d4d1132a6cd0a5063010a96b855abed","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2021-09-22 15:15:04.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7b2f12742fcb06b180712eddf7dd49827b13239f","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2021-09-22 15:15:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3fe04e325f532ebdfc894dc3cd2844dc59675b23","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-09-22 16:48:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dbc46b8731c8aade2a60d33eb4e5cbacb459ecf3","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-09-22 17:58:10.000000000","message":"Patch Set 1: Workflow+1\n\napproving to include this patch hash in the release patch.","accounts_in_message":[],"_revision_number":1},{"id":"9a2b6ee76677431f08d806c5a351b9fdb101a53a","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-09-22 19:09:32.000000000","message":"Patch Set 1:\n\nrecheck.","accounts_in_message":[],"_revision_number":1},{"id":"a4be090165739a7d306ec88c7ce1a047e1ce1bbf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 19:10:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e64645d021182654b1bb4bbfc9d9b92503c87f4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 20:36:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/dbb713e61f5e4c5383e63bf0505aba43 : SUCCESS in 7m 36s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/d1c3a4b3cae041268bf5ea4aa150e715 : SUCCESS in 6m 30s (non-voting)\n- horizon-nodejs14-run-lint https://zuul.opendev.org/t/openstack/build/6b30afede2d7405ab342586d546d5602 : SUCCESS in 6m 17s\n- horizon-nodejs14-run-test https://zuul.opendev.org/t/openstack/build/f6f779d76a9d4634af117fb007968127 : SUCCESS in 11m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d032861094964be5978af33ef94469d3 : SUCCESS in 9m 38s\n- horizon-tox-python3-django32 https://zuul.opendev.org/t/openstack/build/914348c0f51c4bccb4b40c5acfe7ef9f : SUCCESS in 10m 50s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/947064e0c6e4473e8fb51afa23dfbe8c : SUCCESS in 11m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23f3e5ba7c9848eeb78595cd7cdf4533 : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0518fbd975a74b75a1302f4d16a55d3a : SUCCESS in 10m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a131ad0fdf22454185a8a55fa4c9266d : SUCCESS in 9m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b36e0ba645d4ca8a7b7e3e5f0452615 : SUCCESS in 7m 39s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/a0670a82faa8490298e9dee804666671 : SUCCESS in 7m 59s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/a7cd439f8deb45adb9c525b938e9e568 : SUCCESS in 1h 24m 59s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/39a9465b8e3344e689ddc1e1d3555fd8 : SUCCESS in 25m 18s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/086f53d604df42f4a0c74af24094cc02 : SUCCESS in 5m 27s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/e27124ae93254f4e901c84c10aced94a : SUCCESS in 28m 40s","accounts_in_message":[],"_revision_number":1},{"id":"c72e0872595ff77ca225c858ca9035fae1b70955","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 20:37:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- horizon-nodejs14-run-lint https://zuul.opendev.org/t/openstack/build/182b3a54d782489ead05c08efbaca05f : SUCCESS in 6m 10s\n- horizon-nodejs14-run-test https://zuul.opendev.org/t/openstack/build/36042ae795a44d74882ba61e93196806 : SUCCESS in 10m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e184d2187b754a669b951453276e27bc : SUCCESS in 11m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/895ef31668a74deeab71a942f330ab0d : SUCCESS in 11m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecf821283b0a4661bdd7a2210565e30f : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81d3d487190741d694c61b1a6c7ba425 : SUCCESS in 12m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb7f41adc7a6481db34ec36d9503221c : SUCCESS in 6m 59s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/d2203e49d8dd4eba9e482db4a1b52fbf : SUCCESS in 33m 01s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/924573368d6b4480bcd435b7ed451e4a : SUCCESS in 6m 35s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/f8c5fd7a51534213bbcab8480a02b0b6 : SUCCESS in 1h 22m 05s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/2375540554af40f48c01cd4e9676b4b3 : SUCCESS in 34m 01s","accounts_in_message":[],"_revision_number":1},{"id":"55aeb9d60cb8cf1fff6c81e81abc773a2bb3946c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 20:37:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"17b6f7e53ddbe0f41ef26ae713f06b077dfd8682","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 20:38:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b18be5fe39f4d2b9650b6bc373f1bc2 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a20c4f6f2218255597c719aa7505295f689abf9a","revisions":{"a20c4f6f2218255597c719aa7505295f689abf9a":{"kind":"REWORK","_number":1,"created":"2021-09-22 10:43:11.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/88/810388/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/88/810388/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/88/810388/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/88/810388/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/88/810388/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/88/810388/1"}}},"commit":{"parents":[{"commit":"0f34f6e7edff7195519dea3f1298efae2f8abd51","subject":"Add a release note on Django 3.2 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/0f34f6e7edff7195519dea3f1298efae2f8abd51"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2021-09-22 10:37:44.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2021-09-22 10:41:55.000000000","tz":540},"subject":"workflow: Do not touch dict during iteration","message":"workflow: Do not touch dict during iteration\n\nDjango 3.2 stored the field information as dict instead OrderedDict\nbecause python 3.7+ ensure the field order of dict as the language spec.\nWe cannot touch dict itself during iteration. We need to pass a list\ninstead of a value from .items() itself to avoid the error.\n\nChange-Id: Ie22865995d14fa60c16cf2cea582aa0eec46b65d\nCloses-Bug: #1944548\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/a20c4f6f2218255597c719aa7505295f689abf9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/a20c4f6f2218255597c719aa7505295f689abf9a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
