)]}'
{"id":"openstack%2Fhorizon~810693","triplet_id":"openstack%2Fhorizon~master~I1e39e7abea536b5054557eb725069583b7504e0f","project":"openstack/horizon","branch":"master","topic":"add-yoga-python-jobtemplates","hashtags":[],"change_id":"I1e39e7abea536b5054557eb725069583b7504e0f","subject":"Add Python3 yoga unit tests","status":"MERGED","created":"2021-09-23 14:14:38.000000000","updated":"2021-09-24 00:04:04.000000000","submitted":"2021-09-23 23:59:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"810693-add-yoga-python-jobtemplates","meta_rev_id":"530bd45ea646fb39193082e83c2fe1a324286ade","_number":810693,"virtual_id_number":810693,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-23 23:59:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":2,"date":"2021-09-23 14:37:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":2,"date":"2021-09-23 22:29:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":1,"date":"2021-09-23 22:29:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-23 14:37:11.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2021-09-23 16:12:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-23 22:29:29.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"}],"messages":[{"id":"7106347f2286b71e51b8980479ce07e3c37cc93e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2021-09-23 14:14:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0cc17ae9f31ffd6ebb8a15b88f75bdfcd06191c1","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-09-23 14:37:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"918740c3b5cbf018694c862558c05e9b786e3028","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 16:12:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/0281efd8effa435a8435e14453afd524 : SUCCESS in 7m 30s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/676bc7fef70744eab94a35d44d225fc9 : SUCCESS in 7m 28s (non-voting)\n- horizon-nodejs14-run-lint https://zuul.opendev.org/t/openstack/build/020a8d9288da44efa8a211e4cc4f9019 : SUCCESS in 8m 23s\n- horizon-nodejs14-run-test https://zuul.opendev.org/t/openstack/build/c93b0ea81b8d41888588da378710d802 : SUCCESS in 12m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42268f938fce4034b53005f50040f09b : SUCCESS in 12m 17s\n- horizon-tox-python3-django32 https://zuul.opendev.org/t/openstack/build/3e7f7640136647199944c33f41aacc70 : SUCCESS in 13m 12s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f8732b76b34493994d728e08da552f7 : SUCCESS in 13m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f550aff03a3b408d88fc14f3e5478a1d : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96c67b006bbd4c01ad0ac8588ff52379 : SUCCESS in 11m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af13ff8eb99744e4a32da2cf2f64245e : SUCCESS in 13m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86eca7b178664c6aab363d9a62706c29 : SUCCESS in 9m 09s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/bbd726b2442448d8a417ed73d5bd58a8 : SUCCESS in 9m 59s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/64a5e541766f49a993007ffcfa5c510e : FAILURE in 1h 47m 01s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/acb8a49e26c54dd5973090b144a6c628 : SUCCESS in 33m 28s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/debe9af661714beba8e4f4c26d3915d1 : SUCCESS in 7m 23s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/09bcd90a0f13486dae496fd361d08bc8 : SUCCESS in 32m 29s","accounts_in_message":[],"_revision_number":1},{"id":"ceb2e46c2e87cc1feea49caf1767300d5a3e1319","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-09-23 16:26:47.000000000","message":"Patch Set 1:\n\nrecheck.","accounts_in_message":[],"_revision_number":1},{"id":"291655a6b0358b71b43b616cb86d1823dcda6270","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 17:59:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/ad8a2948081d44ea972d2acc6dfe867f : SUCCESS in 8m 42s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/aa2ca6e50b6d44ab93df5a48eefdab92 : SUCCESS in 7m 03s (non-voting)\n- horizon-nodejs14-run-lint https://zuul.opendev.org/t/openstack/build/f8de80b0cf4d43c3986bf9d154045fee : SUCCESS in 7m 35s\n- horizon-nodejs14-run-test https://zuul.opendev.org/t/openstack/build/fd588934fb17423797a0868b89024df0 : SUCCESS in 11m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e675d9fc149b4b5d949354ea9f4d6a01 : SUCCESS in 10m 45s\n- horizon-tox-python3-django32 https://zuul.opendev.org/t/openstack/build/e279f26bc2f94c27971e68ea1019d2e8 : SUCCESS in 11m 29s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a4c0f27d4910410990be73d03a65d35f : SUCCESS in 13m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4f15fab261b47c18989034d373fb380 : SUCCESS in 7m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be3896e2f0f14c83be5ecba767a346d4 : SUCCESS in 11m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e251c531e1974efa8e62d20f4ea03cbb : SUCCESS in 11m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b204437de65244dc93bcb1de8662e572 : SUCCESS in 8m 54s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/5b1b0782f8334126899597df0f8c1fbc : SUCCESS in 8m 23s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/22ebfca8840d4e30b0bf86a9e4449488 : SUCCESS in 1h 30m 09s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/e547317f66214630ad0d8bcda8955e85 : SUCCESS in 34m 55s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/924628fb315b41f1b599f3359d84c342 : SUCCESS in 6m 54s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/857fc7db4b054b2fa1c66c61c3a282c1 : SUCCESS in 33m 48s","accounts_in_message":[],"_revision_number":1},{"id":"c1f062f01a3cbbcd505a2496c0da2b50c89e5ca9","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-09-23 22:29:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"23aca8bf9f97fbab2681e53041c62aeb7e6f2cf9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 22:31:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c59ef06a917b73c8966d12bbec77e0a62dc40b32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 23:59:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- horizon-nodejs14-run-lint https://zuul.opendev.org/t/openstack/build/c78b3468ab554638a4c16bfdbbc7005e : SUCCESS in 7m 02s\n- horizon-nodejs14-run-test https://zuul.opendev.org/t/openstack/build/028784d5a3cb4daa920c59fd51e3f7da : SUCCESS in 10m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d747303def434d82adf093e8a8d2cd93 : SUCCESS in 10m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/538ca8fdfefd4346b42ba5612aaac0e6 : SUCCESS in 10m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08261ff6cc6a4b3a83cbf3edb62a796d : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7918de5cda2e4e7f84a805bd0a94e154 : SUCCESS in 9m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5837823a2ab64023af3ef1be64e2c3dc : SUCCESS in 7m 59s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/570357372a6d4c3c8c7a62c73bf5f965 : SUCCESS in 28m 56s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/ece38e388db14cc98231c979843712df : SUCCESS in 8m 30s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/a027ec0cafb44e2f914b3618392db591 : SUCCESS in 1h 27m 24s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/5a26403720684084a8b4f5319d9e6b8f : SUCCESS in 29m 29s","accounts_in_message":[],"_revision_number":1},{"id":"10f526cec4720600964d26e93487a5b242843873","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 23:59:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"530bd45ea646fb39193082e83c2fe1a324286ade","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 00:04:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01a8e0969f1b497a9e151b0df8666c25 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"66091d7b2f42c4177171c096fdda41a50b39852e","revisions":{"66091d7b2f42c4177171c096fdda41a50b39852e":{"kind":"REWORK","_number":1,"created":"2021-09-23 14:14:38.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/93/810693/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/93/810693/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/93/810693/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/93/810693/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/93/810693/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/93/810693/1"}}},"commit":{"parents":[{"commit":"c8dc5cf7099aa2a4a76a757043283e94be92254f","subject":"Update master for stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/c8dc5cf7099aa2a4a76a757043283e94be92254f"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-09-23 14:14:34.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-09-23 14:14:34.000000000","tz":0},"subject":"Add Python3 yoga unit tests","message":"Add Python3 yoga unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for yoga.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: I1e39e7abea536b5054557eb725069583b7504e0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/66091d7b2f42c4177171c096fdda41a50b39852e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/66091d7b2f42c4177171c096fdda41a50b39852e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
