)]}'
{"id":"openstack%2Fhorizon~830692","triplet_id":"openstack%2Fhorizon~master~I6d6048f78f56d0f3ca0f5804f1059bf31cde6a9e","project":"openstack/horizon","branch":"master","attention_set":{},"removed_from_attention_set":{"3031":{"account":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"last_update":"2022-08-22 15:03:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6d6048f78f56d0f3ca0f5804f1059bf31cde6a9e","subject":"Bugfix: Add missing closing \u003c/div\u003e tags to network topology","status":"MERGED","created":"2022-02-23 21:36:15.000000000","updated":"2022-08-22 15:04:25.000000000","submitted":"2022-08-22 15:03:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830692","meta_rev_id":"ea664a14938fa4336a942459160e04de96024f40","_number":830692,"virtual_id_number":830692,"owner":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-22 15:03:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":2,"date":"2022-08-22 04:50:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":1,"date":"2022-08-22 13:57:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":2,"date":"2022-02-25 04:42:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":1,"date":"2022-08-22 04:50:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-23 23:06:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-25 04:42:32.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2022-08-22 04:50:14.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2022-08-22 13:57:23.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"259bd7a5b12c547eafb0ed097ffab9c9ec5d5985","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2022-02-23 21:36:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7a3af7e98f89f14c60b03ba574ccc81a572180b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 23:06:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/f20df7d2e93c41229517ddc930b664c6 : SUCCESS in 5m 27s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/d44f9fe76d934ffdb105987cdd0031ef : SUCCESS in 4m 21s (non-voting)\n- horizon-nodejs14-run-lint https://zuul.opendev.org/t/openstack/build/9ea10496d4054e19b8816f731e3dfdb0 : SUCCESS in 5m 39s\n- horizon-nodejs14-run-test https://zuul.opendev.org/t/openstack/build/96923d9763904217a8d3c624e016dc73 : SUCCESS in 7m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4bd774a64c0f431cb5851c3adf7a2596 : SUCCESS in 10m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/57c223bad37b464db5de7438191dc8f6 : SUCCESS in 10m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0a5ac5988204b42b9c020767ed3a60e : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a6ce3531bcfc49c9a61f79affc0f4ad7 : SUCCESS in 9m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62f06c4081604827993a7b6db8af58ec : SUCCESS in 5m 55s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/2fe07cf5660a479e88fffbac2bec74ee : SUCCESS in 6m 45s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/86d392d69b0f462d8bd7d6e56f821fcd : FAILURE in 1h 26m 50s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/84e7d3100d4748f68f6049b9e9d5cf39 : SUCCESS in 21m 06s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5200aecf1ea64179804c97651a00828d : SUCCESS in 4m 48s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/45dc9714aaf3416b959228b29e824fb8 : SUCCESS in 22m 29s","accounts_in_message":[],"_revision_number":1},{"id":"a824cf98f5f111ae1d7904cfa3248d21cec49bda","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2022-02-24 21:51:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2cf4bc705fa177f634c5b9077f564e4f5a262172","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 23:48:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/5b24b2afc223458080467759a8b568fc : SUCCESS in 8m 35s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/795bdae31f064b17a15d828c88b1248e : SUCCESS in 5m 19s (non-voting)\n- horizon-nodejs14-run-lint https://zuul.opendev.org/t/openstack/build/9035fdd8d0014237ad98dd1032b550ee : SUCCESS in 5m 09s\n- horizon-nodejs14-run-test https://zuul.opendev.org/t/openstack/build/69d75f70ebb84e898a80413666857073 : SUCCESS in 9m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1aeb9e4b8ca547038166f3e4dc6e2313 : SUCCESS in 11m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/908eb9fe3e4d4ac89c49f00f78d9a700 : SUCCESS in 11m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28fc34d7bd1843b89ca802fc37b8c058 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e8cbb8273e543f98cf00c5f6561a625 : SUCCESS in 11m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3591e859ed14f829e57273ee6c61cab : SUCCESS in 6m 33s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/844468cf8e074463ac36757b6ce33915 : SUCCESS in 6m 44s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/df4a7e1e92de4631815c80dbfb44611b : SUCCESS in 1h 51m 48s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/e0528dbf4e68455da0d5693d26e591e5 : SUCCESS in 34m 57s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c5d3a581ec65414b908f0b0bb1b133cc : SUCCESS in 5m 19s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/97b77fff2bf74e1999d4b5926366643d : SUCCESS in 33m 21s","accounts_in_message":[],"_revision_number":1},{"id":"c406816ec26fa1098424592dbf9552c7886cd019","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2022-02-25 04:42:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"63c0c6bded48c306ef297070aa537a369518b36f","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2022-08-22 04:50:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"adcf291cecb0c761ba576c2b4bc04f091158e937","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 04:50:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e82e5ec75c4b760af1b32aec587262ea0eb02754","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 06:29:37.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9304a943af154e6eb4419453e6ba07c3\n\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/803816988a8043deb33e9a4088ab3ffa : SUCCESS in 5m 43s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/5eeb8249a66943be8df6595283c8b156 : SUCCESS in 8m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5891dac4384e47de9fa026bbf928b3c7 : SUCCESS in 8m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37b617e8f20447838f16f9e23ece2405 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6bb2ff9d1d4b4beab05c1469f925e1fc : SUCCESS in 8m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c2be034b72641dc91ea91e7a9539ea6 : SUCCESS in 6m 37s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/9f47f69afa224d33972c39ddc4690237 : SUCCESS in 25m 26s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/cd6080bd838146d6abe3fa9c426dfd06 : SUCCESS in 6m 57s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/eaed4576062d40ce8a0f873b2224bd2c : FAILURE in 1h 07m 36s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/0db4d8ebfd174238866b698d16a522bb : SUCCESS in 25m 24s","accounts_in_message":[],"_revision_number":1},{"id":"c27b09d6194eef9debbfe9ad78b6f4578d8c30aa","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2022-08-22 12:13:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"52c7e941a0d2fbbbb92a203ce76f17f79ff3149f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 13:41:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6b4ed929889433599cce8cc8ae77878\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/ec8ff49c63fd48359ce4cea3023c427a : SUCCESS in 6m 29s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/0bd4c9b2d72e468ebc290ecac2ea9add : SUCCESS in 8m 58s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/745bd8073f744f3391f8d1d4d3e52153 : SUCCESS in 4m 37s (non-voting)\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/70e7b8ada30f4b0681e632ba38ab8bc8 : SUCCESS in 5m 50s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/4fd37b4a592b4a8c86631a9dd7fe8fb1 : SUCCESS in 10m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9c6f428ce57f4cd5a24b6a38f654a2ec : SUCCESS in 9m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51def9f780484a7cbdee9d77ec20c221 : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/888449a3ed3b417bbf4ac13128c3149e : SUCCESS in 9m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/87a16fba3c3a44d68f507d208f14a1c8 : SUCCESS in 10m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89f4b4caddcb463888158312e9242914 : SUCCESS in 7m 09s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/f6fbe194e3ba47e3aa0272ea09ea36f5 : SUCCESS in 6m 36s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/3b1a94dfbb284aa2b3a4fd45966fbb41 : SUCCESS in 1h 26m 13s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/5ba8d20a91684409b079c55c737f47b2 : SUCCESS in 26m 47s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e295df57740149be909b535146545e27 : SUCCESS in 4m 44s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/57d291c4fe7f45d7b2b3a0bc87667232 : SUCCESS in 31m 32s","accounts_in_message":[],"_revision_number":1},{"id":"e3f0bdd6cf224b70141c2c4e03a47fc920f4a1e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 13:42:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b324ebf14cdc312e09f49c27bb0b9511699b28ab","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2022-08-22 13:57:23.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a0cb2bab0ea749b67fb78810bf662c886d3b51bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 15:03:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fc95ca036724c48af2e27a53315cd56\n\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/4e76656617594c36ac2248f7109a3f67 : SUCCESS in 5m 47s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/71d3d3609c884eb08917385c10f7961a : SUCCESS in 9m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/68f609f90a0d41fd8a46357c7af19084 : SUCCESS in 10m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df2c954a52cb4e05a53b652aa626e3d6 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fcfd3e316d2040beb30991defb504b12 : SUCCESS in 10m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11e09121afd0494caddf2c39535f708c : SUCCESS in 7m 26s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/622a045fea474809b71d9211d6d81df4 : SUCCESS in 32m 15s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/5d833a08d5034ed6ac83f5077b5e239e : SUCCESS in 7m 35s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/6245d754a93b4107bc20485b0e902155 : SUCCESS in 1h 19m 46s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/cca2b2eef00d4b78b42ab3945a479fd3 : SUCCESS in 32m 21s","accounts_in_message":[],"_revision_number":1},{"id":"5c23e6c6d8a3fdcd316d23d351830f78f983e99d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 15:03:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ea664a14938fa4336a942459160e04de96024f40","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 15:04:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31292f0084de46ab8ad17b55fa1d4988\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7697ddcd0ebc4319bba3d35464afe5fb : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f9eb7c8d999323b14dce8d9d50689ff1faff3a05","revisions":{"f9eb7c8d999323b14dce8d9d50689ff1faff3a05":{"kind":"REWORK","_number":1,"created":"2022-02-23 21:36:15.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/92/830692/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/92/830692/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/92/830692/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/92/830692/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/92/830692/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/92/830692/1"}}},"commit":{"parents":[{"commit":"94e4e2df0106293ae0c7880b8fb687523bf9c436","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/94e4e2df0106293ae0c7880b8fb687523bf9c436"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2022-02-23 21:35:21.000000000","tz":660},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2022-02-23 21:35:25.000000000","tz":660},"subject":"Bugfix: Add missing closing \u003c/div\u003e tags to network topology","message":"Bugfix: Add missing closing \u003c/div\u003e tags to network topology\n\nChange-Id: I6d6048f78f56d0f3ca0f5804f1059bf31cde6a9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/f9eb7c8d999323b14dce8d9d50689ff1faff3a05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/f9eb7c8d999323b14dce8d9d50689ff1faff3a05"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
