)]}'
{"id":"openstack%2Fhorizon~847985","triplet_id":"openstack%2Fhorizon~master~I013da554f58e1638178d3655459c9742fd7f85ba","project":"openstack/horizon","branch":"master","topic":"volumes-pagination-setup","attention_set":{},"removed_from_attention_set":{"8648":{"account":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"last_update":"2022-07-22 16:57:49.000000000","reason":"Change was submitted"},"841":{"account":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"last_update":"2022-07-22 16:57:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I013da554f58e1638178d3655459c9742fd7f85ba","subject":"Add setup and teardown to the volumes pagination tests","status":"MERGED","created":"2022-06-28 15:18:09.000000000","updated":"2022-07-22 16:58:58.000000000","submitted":"2022-07-22 16:57:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"847985-volumes-pagination-setup","meta_rev_id":"9e814477cadc1e5f66597fdb7d398aab17c0bc88","_number":847985,"virtual_id_number":847985,"owner":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},{"value":0,"_account_id":14151,"name":"ziyu032","email":"chen.qiaomin@99cloud.net","username":"ziyu032"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-22 16:57:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":2,"date":"2022-07-06 14:09:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":2,"date":"2022-07-22 15:31:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":1,"date":"2022-07-08 14:03:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},{"value":1,"date":"2022-07-08 10:24:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14151,"name":"ziyu032","email":"chen.qiaomin@99cloud.net","username":"ziyu032"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":1,"date":"2022-07-22 15:31:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},{"value":0,"_account_id":14151,"name":"ziyu032","email":"chen.qiaomin@99cloud.net","username":"ziyu032"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},{"value":0,"_account_id":14151,"name":"ziyu032","email":"chen.qiaomin@99cloud.net","username":"ziyu032"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"}],"REVIEWER":[{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":14151,"name":"ziyu032","email":"chen.qiaomin@99cloud.net","username":"ziyu032"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-28 15:27:12.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"CC"},{"updated":"2022-06-28 16:41:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-04 17:38:30.000000000","updated_by":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"reviewer":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"state":"CC"},{"updated":"2022-07-06 14:09:19.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2022-07-08 10:24:10.000000000","updated_by":{"_account_id":14151,"name":"ziyu032","email":"chen.qiaomin@99cloud.net","username":"ziyu032"},"reviewer":{"_account_id":14151,"name":"ziyu032","email":"chen.qiaomin@99cloud.net","username":"ziyu032"},"state":"REVIEWER"},{"updated":"2022-07-08 14:03:53.000000000","updated_by":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"reviewer":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"state":"REVIEWER"},{"updated":"2022-07-22 15:31:54.000000000","updated_by":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"reviewer":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"state":"REVIEWER"}],"messages":[{"id":"f40f49e7a46c17eb348b0b0d005a7771b90f61c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2022-06-28 15:18:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8aa18b1dc02d97f3482fb91639a3b88d7a3dabed","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2022-06-28 15:26:49.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"849484b84f53f621eca0db6a7423ed9427b1b3d8","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2022-06-28 15:27:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5707fce1cb20d96f89cd6de0f8c282ae0de25c6c","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2022-06-28 15:42:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e511e22598131e65133d76cf4661b5663507e723","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2022-06-28 15:48:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"df1288b0b745b0c22455e975d74f88a2f6153ae0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 16:41:30.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/7d77f5a854914c86a57b4bf38c2dbe69 : SUCCESS in 5m 22s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/98cd27e29c67441ba906247f2c41aab5 : SUCCESS in 5m 07s (non-voting)\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/4bc64f644be449a1b58d1dc73e7611c0 : SUCCESS in 5m 29s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/8c08e5af53d0471eb476f7332a4dae19 : SUCCESS in 28m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f65424baebb46f09be5e574a1ced785 : SUCCESS in 9m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd9e574274d948d0b6b600f7b68c8ec8 : SUCCESS in 10m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d382cf9203264c3b991f09adb642cd98 : FAILURE in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9dfb47a3e2564914aa49d3cffbef4bc7 : SUCCESS in 9m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/464a14da28c442488cc776b0d7b67ed4 : SUCCESS in 9m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/734da01b480742249a1e54e20fd2151b : SUCCESS in 7m 00s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/812b2efa892342f9ad15ac2335e9e0d1 : SUCCESS in 6m 57s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/6128b8da86c948149e883590653beb34 : FAILURE in 1h 21m 08s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d3f7bf4661914add8e9d326446f727a9 : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":1},{"id":"3c75c1afb2ca3ec6e376a2432f36f6f538f00b88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2022-06-29 07:12:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"664bc2a35d0528094f376d0eb59b5b35eae872c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 08:38:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/4033c0d97bc44e7fb606309d68d79767 : SUCCESS in 6m 09s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/19906688c967494288d0b865f9590f67 : SUCCESS in 5m 16s (non-voting)\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/3649e43e1b8a488696759465e0f70c75 : SUCCESS in 4m 53s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/6b493a11e13b467daf35e1e528685d04 : SUCCESS in 8m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5427616d2d5c429d85dfb4f9ed56f875 : SUCCESS in 8m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/140ad319d4914bdd99b5d5978e7491b9 : SUCCESS in 10m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed7f39bc6ee447f2ac6e7a390c309d60 : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/129c3a717c2b48e6b0ff5ee603e3107a : SUCCESS in 9m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1c0b88a7cde84752a428990e9ee27da6 : SUCCESS in 12m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d5383e403f449599b73c80c996d1ba7 : SUCCESS in 6m 37s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/b8a22355a75b4f57b33738af007b9e60 : SUCCESS in 7m 10s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/bb4458f4d1d142d19664551fb335725d : FAILURE in 1h 23m 17s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/811c5124706f4c90ba479e6bcdc8e07b : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":2},{"id":"e793af2bfb2fd99839006799bb400e510cfb384b","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2022-07-04 17:38:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"41b0357b52d920cad83043e5f0e8de3a42266554","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 19:07:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab8336b254df4f348ebe26c59f02641f\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/73a84e5d30ca40e8a117200d4ae93a36 : SUCCESS in 6m 51s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/8c6d196b4b304045b10c0addc08285ab : SUCCESS in 5m 01s (non-voting)\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/3be5ae9fb86c4e8687f69362a74f0b43 : SUCCESS in 5m 03s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/26014973dce547e6bb26d9948980faad : SUCCESS in 8m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d4a37bcdf7f474f85b4de63a7cac213 : SUCCESS in 10m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e40c776da4834dafb0775d2c6c6f0b05 : SUCCESS in 12m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f6d8d4355e24251b1c8b779e0dad07e : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d38585abc7942b2b9c8a6001df848f8 : SUCCESS in 10m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/78198a08a9e34b5d9b4378cb667f8f2b : SUCCESS in 11m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce3b10d2dc174bdf87d6d91b95b5cf9d : SUCCESS in 6m 31s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/ddfab6e96773481eb2591f602c1767fc : SUCCESS in 6m 28s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/e0b061f909374574b01ec64d4b3e3826 : SUCCESS in 1h 21m 46s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3fbab3e757c94b60b35c0196ca5f13ab : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":2},{"id":"f0e610377c93845c2afc7d03bfde5d624acb79c3","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2022-07-06 14:09:19.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e9b1caed98d1eed54e042704cdb16257fadbe091","author":{"_account_id":14151,"name":"ziyu032","email":"chen.qiaomin@99cloud.net","username":"ziyu032"},"date":"2022-07-08 10:24:10.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"7e7227e1166a99fda2117dc6cf7526612cfbec02","author":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"date":"2022-07-08 14:03:53.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"6490961294828ab07c9fd0f19d70213f38150ba4","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2022-07-22 15:31:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c59b8cc298f8e73610f60ae7ab8144b1cfbbe845","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 15:32:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"64649d8024d14c08a86b003ac3c8c551726c494a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 16:57:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/477de68da5c344d3a5685d1ed2ddb965\n\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/b2988940d6a7428f94a86220866b7ae9 : SUCCESS in 5m 39s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/768c17cf0b0a40cab66771bb4fb27f81 : SUCCESS in 8m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e66699681a7a4312b2f739c9ceb455ae : SUCCESS in 12m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6abc7113574481fbeac8c73828daf40 : SUCCESS in 10m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8425b91c477d4217905b87c9329f8899 : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/02416de2eb65434183d311697cfd46ed : SUCCESS in 9m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91509b3753f143f59dd1fb5dee8d119f : SUCCESS in 7m 49s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/585fa68f25174c9bb5c1fb307b7df2a3 : SUCCESS in 8m 10s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/0c3f741c39e54954ad72741cbbe37750 : SUCCESS in 1h 24m 27s","accounts_in_message":[],"_revision_number":2},{"id":"c749f51afb2cd43ccd5d7180fd0b6388caa96d01","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 16:57:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9e814477cadc1e5f66597fdb7d398aab17c0bc88","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 16:58:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbf8fad52bf442639551656f7496ff05\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/904b6cc7e37144b7b45424b1afa5ce0b : SUCCESS in 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"81edf08676f2651f4343498c23d93b000fca7127","revisions":{"0115ca7745e910b4ff2cb6c3dd24919414040ee6":{"kind":"REWORK","_number":1,"created":"2022-06-28 15:18:09.000000000","uploader":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"ref":"refs/changes/85/847985/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/85/847985/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/85/847985/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/85/847985/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/85/847985/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/85/847985/1"}}},"commit":{"parents":[{"commit":"f7a55fab69388a329cec4bea664c6bd8fa8d8138","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/f7a55fab69388a329cec4bea664c6bd8fa8d8138"}]}],"author":{"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","date":"2022-06-28 15:17:31.000000000","tz":120},"committer":{"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","date":"2022-06-28 15:17:31.000000000","tz":120},"subject":"Add setup and teardown to the volumes pagination tests","message":"Add setup and teardown to the volumes pagination tests\n\nFor now just delete the old volumes at the start\n\nChange-Id: I013da554f58e1638178d3655459c9742fd7f85ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/0115ca7745e910b4ff2cb6c3dd24919414040ee6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/0115ca7745e910b4ff2cb6c3dd24919414040ee6"}]},"branch":"refs/heads/master"},"81edf08676f2651f4343498c23d93b000fca7127":{"kind":"REWORK","_number":2,"created":"2022-06-29 07:12:47.000000000","uploader":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"ref":"refs/changes/85/847985/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/85/847985/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/85/847985/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/85/847985/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/85/847985/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/85/847985/2"}}},"commit":{"parents":[{"commit":"f7a55fab69388a329cec4bea664c6bd8fa8d8138","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/f7a55fab69388a329cec4bea664c6bd8fa8d8138"}]}],"author":{"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","date":"2022-06-28 15:17:31.000000000","tz":120},"committer":{"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","date":"2022-06-29 07:12:36.000000000","tz":120},"subject":"Add setup and teardown to the volumes pagination tests","message":"Add setup and teardown to the volumes pagination tests\n\nFor now just delete the old volumes at the start\n\nChange-Id: I013da554f58e1638178d3655459c9742fd7f85ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/81edf08676f2651f4343498c23d93b000fca7127"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/81edf08676f2651f4343498c23d93b000fca7127"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
