)]}'
{"id":"openstack%2Fhorizon~854005","triplet_id":"openstack%2Fhorizon~master~I14aa1f1ee78f123bdbb6e3c4db314f96eec96f1b","project":"openstack/horizon","branch":"master","attention_set":{},"removed_from_attention_set":{"31746":{"account":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"last_update":"2023-01-03 09:13:12.000000000","reason":"Change was abandoned"},"1004":{"account":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"last_update":"2023-01-03 09:13:12.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I14aa1f1ee78f123bdbb6e3c4db314f96eec96f1b","subject":"Fix endpoint type in user_verify_admin_password","status":"ABANDONED","created":"2022-08-22 15:15:38.000000000","updated":"2023-01-03 09:13:12.000000000","total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0dc87f16452cfa9fc276e764d6654401a5139510","_number":854005,"virtual_id_number":854005,"owner":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-09-27 16:31:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-22 16:45:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-23 04:12:05.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2022-10-11 12:52:36.000000000","updated_by":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"reviewer":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"state":"CC"},{"updated":"2022-11-28 09:28:04.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"CC"}],"messages":[{"id":"36e092b6ef3b1f27e35a96da9196bdc09f3db2d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2022-08-22 15:15:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8fcb2a0b4cfb68582e48a7fb4875e536860be2c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 16:45:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf4a5719ae9c41c2baa3ff420a6e43f2\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/9357f90442424b568683940e1c9e50dc : SUCCESS in 7m 05s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/e5c790660b3a482ea89d3488b97d995d : SUCCESS in 8m 04s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/e66e4f5414714b56aa31fccb6a6bd589 : SUCCESS in 5m 03s (non-voting)\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/296d581c26054961830200ae67a17460 : SUCCESS in 5m 45s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/4b3d5bc50b3c4d6f937caa2b74409ddb : SUCCESS in 9m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ab3989680d44acfbbb31c213200072c : SUCCESS in 10m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26f8e3050e3d4246b792fb737a07de88 : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f9423c1c93e64696beb16f4b2a698eaf : SUCCESS in 9m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/14bf30f678e14bab890f5d27768abb03 : SUCCESS in 9m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd3c80863cf442afb7f54afaaa83950a : SUCCESS in 7m 48s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/c72d59b4128c488b8f389cee2045909a : SUCCESS in 7m 34s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/bc1681834d114a088e938ac3793c6ceb : SUCCESS in 1h 27m 46s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/eafa1f43f5554495b8d6916eafd5c6fe : SUCCESS in 36m 07s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9480e23b66b44899bd466a4136fdabd9 : SUCCESS in 5m 03s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/359c3a4f1f7a425f916cd8e8f21c89b0 : SUCCESS in 35m 45s","accounts_in_message":[],"_revision_number":1},{"id":"19fd9a8e8bbe16e302872ac6c5c7a9a76df33284","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2022-08-23 04:12:05.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"421a4b17763d449167f428fd2057b0101d03b715","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2022-08-23 07:11:24.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f3d3661428945d47707a824450ef935879eca353","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2022-08-23 07:12:00.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"8e40ef93bb82b15b3ae0af61094b5d014cadfcaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 08:32:33.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/86c7af2352bf4036b93f64f9511da228\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/4ed3622aa99f418f89fee9b8785b1caf : SUCCESS in 7m 33s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/3f63904d50ce48e3aec07e5c9092f834 : SUCCESS in 9m 49s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/55156898da4f415397a859d4bad228f2 : SUCCESS in 5m 30s (non-voting)\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/ea114b27bcaf42a0b110a7b182a79189 : SUCCESS in 5m 56s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/7ada9a71d421407e8234599db0947b34 : SUCCESS in 10m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/82a42b809f1945f7babb1c42c0ebf330 : SUCCESS in 10m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54848e01676a4b928e0394ee07409e21 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb2ae83d8e2d47f6a76ce25f6bdad20c : SUCCESS in 9m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e8ae007be04448839861810d2a7549a7 : SUCCESS in 10m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97ba37da8c8b47fbbae980c9a4a00f24 : SUCCESS in 8m 05s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/81d000b4eac946dca6ee27463ac047bb : SUCCESS in 9m 15s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/175d742529d741cbb42d4280860d3c87 : FAILURE in 1h 18m 59s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/52164fc4ce9b4ad2867ab491353d13f6 : SUCCESS in 40m 17s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e69b19dfea804db48231e823a526d6fe : SUCCESS in 6m 43s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/7543e2d4e7df430fb7b29436f2c5e62d : SUCCESS in 41m 20s","accounts_in_message":[],"_revision_number":3},{"id":"315efd5a8a9e0bf58056c02d3181e4d8189c498e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2022-08-25 15:54:26.000000000","message":"Abandoned\n\nFalse negative","accounts_in_message":[],"_revision_number":3},{"id":"c041d21f9f6510057a376cbfe5badae2a4ae4916","tag":"autogenerated:gerrit:restore","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2022-09-27 14:44:17.000000000","message":"Restored","accounts_in_message":[],"_revision_number":3},{"id":"0117e32c039f5cad22d52fd74931b4014b396b19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2022-09-27 14:48:13.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"009b0ac42353083b87ffb1053deaaa9388628f1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2022-09-27 14:55:49.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"163610818e72d5bcb6b34eba59c46ba3e88bc917","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2022-09-27 14:57:07.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"b16886f5064907176182c1c02431ba4914dda0aa","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2022-09-27 14:58:17.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"68c1a491a504588e3e56771d38efe413b6ac30d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 16:31:02.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5315c1060b5c420db9d4bc2de2950f1f\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/327de42a26144d849ad1c0d04ba88896 : SUCCESS in 6m 52s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/38282ebf290940de8819a25a669f8a23 : SUCCESS in 28m 27s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/c1da816ca0044dc2a1f1144cedbb39b4 : SUCCESS in 5m 28s (non-voting)\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/bfa621d7d4614bc787bdf59d11b936ed : SUCCESS in 5m 43s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/5263d7ea0f0643f4a32bcd2637cbd9d9 : SUCCESS in 19m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dbd930dd5698441fb368ebb693bbcbca : SUCCESS in 10m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/535a02e0094c4f0aa78b5f42f4c19422 : SUCCESS in 7m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d3dd8203fec8426787a43fb051cf6267 : SUCCESS in 11m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0817861241e5498980161afed2fbf41c : SUCCESS in 7m 03s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/5723b5a686144b2d93c9c96a51eede04 : SUCCESS in 6m 30s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/005a9b2e5a2c4311a62fffd35ce2c5da : FAILURE in 1h 26m 10s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/89fcef29fd054ec6b2c9a9ecc749ca52 : SUCCESS in 27m 23s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/76d04a49ed7a41dda7326159fdb2a1c7 : SUCCESS in 4m 25s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/e16978a2246e4d6db02dce544d988cff : SUCCESS in 27m 22s","accounts_in_message":[],"_revision_number":6},{"id":"3ec4a74bc1c570a8953f1f889c3e54b36a532620","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2022-10-11 12:52:36.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9e31b2f373a866d8931c8e27d032fa095ab5b283","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 14:48:07.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/320e339616c741fb9e187501befb6385\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/9fa2a21973ce4e7a881cb458318632f7 : SUCCESS in 5m 20s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/b9de761075424cf08914c02188961d86 : SUCCESS in 7m 13s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/e6d4aac21bd14ff4a007b6a30791363b : SUCCESS in 3m 56s (non-voting)\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/f0a400ab52094d22ba7a3c2b44e6d9f0 : SUCCESS in 4m 05s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/1cd3d35b8ef045b4aba0cf6320ff0a93 : SUCCESS in 7m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/40cb56f99be5429c91794d6f926299b6 : SUCCESS in 10m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccd29e6ae7094e029bcfdf67262bdf6b : SUCCESS in 5m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0afe4eb909cc447ca01acfc8ac005d0c : SUCCESS in 10m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c052990db3eb456bac19b358f58cffd8 : SUCCESS in 5m 55s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/2d7d8fcaab6d4d06ac017db335d7a73f : SUCCESS in 6m 18s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/62774f61d33740c485c686bdbcf4f258 : FAILURE in 1h 50m 32s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/2c14ae7d168a49849b30cfaa858698c0 : SUCCESS in 34m 14s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/57097e6480424f969561273044039446 : SUCCESS in 3m 44s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/45db2cb44f744c6998c0d2d7cb3e696c : SUCCESS in 33m 20s","accounts_in_message":[],"_revision_number":6},{"id":"47c39e2447c65e2ad890a4d5ef05f44388290909","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2022-11-02 14:02:51.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"cc745ce3cc81efc1520f0a72a234b95721e1fa69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 15:55:24.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/803fa8b0935c49e180ac0918de11afa8\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/aca23528998d4a7bac3a64264028efc5 : SUCCESS in 6m 01s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/f1349df0707c4a7f95a46b57686f60aa : SUCCESS in 7m 23s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/9088c50833774f299de546c8172ce057 : SUCCESS in 3m 41s (non-voting)\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/c5af7fa6fdc54c689dc7cfb9e501d154 : SUCCESS in 5m 21s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/0627aaa8ebb5428aa4ecd20491279e86 : SUCCESS in 6m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/50b10bf4a59d484d9d6e88ddc9ab4e7a : SUCCESS in 9m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65b0112c7e6e40e6989e3e495a405838 : SUCCESS in 6m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8591c642f2f04727a054e5fe94d8e0b1 : SUCCESS in 7m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51cbb9e433db49a1a0b63d606ced3b88 : SUCCESS in 6m 41s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/bd12012f59714dd7ac0d24967620bcd3 : SUCCESS in 6m 17s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/e67928849f56412a9bfce207936aefab : FAILURE in 1h 48m 35s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/16bac468cc254816bad9cb3a41915560 : POST_FAILURE in 25m 27s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ae5b03fd1097449dbc3b1a59cbc3472a : SUCCESS in 4m 04s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/b87a22a23c78461188e1f2772456bc9c : SUCCESS in 26m 15s","accounts_in_message":[],"_revision_number":6},{"id":"0758fd6bf7ad2196fcd6434bb97ca99cd68a0bda","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2022-11-28 09:28:04.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0dc87f16452cfa9fc276e764d6654401a5139510","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2023-01-03 09:13:12.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"73b60c3f8ffa190d5f98a82ddaa4f59483a0c356","revisions":{"3dcd51060dd6cde7aa50a9b7fda40b27d5f27ad0":{"kind":"REWORK","_number":1,"created":"2022-08-22 15:15:38.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/05/854005/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/05/854005/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/05/854005/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/05/854005/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/05/854005/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/05/854005/1"}}},"commit":{"parents":[{"commit":"9bcf50f3cc049dcdee1c90680947459acde6de1a","subject":"Merge \"Bugfix: Add missing closing \u003c/div\u003e tags to network topology\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/9bcf50f3cc049dcdee1c90680947459acde6de1a"}]}],"author":{"name":"okozachenko","email":"okozachenko1203@gmail.com","date":"2022-08-22 15:12:59.000000000","tz":600},"committer":{"name":"okozachenko","email":"okozachenko1203@gmail.com","date":"2022-08-22 15:12:59.000000000","tz":600},"subject":"Fix user_verify_admin_password function","message":"Fix user_verify_admin_password function\n\nWhen ENFORCE_PASSWORD_CHECK is set True, password change is failing.\nBecause user_domain_id or user_domain_name are not provided.\n\nChange-Id: I14aa1f1ee78f123bdbb6e3c4db314f96eec96f1b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/3dcd51060dd6cde7aa50a9b7fda40b27d5f27ad0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/3dcd51060dd6cde7aa50a9b7fda40b27d5f27ad0"}]},"branch":"refs/heads/master"},"078062e48c84cd1b77fc04c81074182fdefa95d9":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-08-23 07:11:24.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/05/854005/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/05/854005/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/05/854005/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/05/854005/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/05/854005/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/05/854005/2"}}},"commit":{"parents":[{"commit":"9bcf50f3cc049dcdee1c90680947459acde6de1a","subject":"Merge \"Bugfix: Add missing closing \u003c/div\u003e tags to network topology\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/9bcf50f3cc049dcdee1c90680947459acde6de1a"}]}],"author":{"name":"okozachenko","email":"okozachenko1203@gmail.com","date":"2022-08-22 15:12:59.000000000","tz":600},"committer":{"name":"okozachenko","email":"okozachenko1203@gmail.com","date":"2022-08-23 07:08:11.000000000","tz":600},"subject":"Fix user_verify_admin_password function","message":"Fix user_verify_admin_password function\n\nWhen ENFORCE_PASSWORD_CHECK is set True, password change is failing.\nBecause user_domain_id or user_domain_name are not provided.\n\nCloses-Bug: #1728031\n\nChange-Id: I14aa1f1ee78f123bdbb6e3c4db314f96eec96f1b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/078062e48c84cd1b77fc04c81074182fdefa95d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/078062e48c84cd1b77fc04c81074182fdefa95d9"}]},"branch":"refs/heads/master"},"db14870fbeac7173300c3f76afd0fd81c355e6fe":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-08-23 07:12:00.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/05/854005/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/05/854005/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/05/854005/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/05/854005/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/05/854005/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/05/854005/3"}}},"commit":{"parents":[{"commit":"9bcf50f3cc049dcdee1c90680947459acde6de1a","subject":"Merge \"Bugfix: Add missing closing \u003c/div\u003e tags to network topology\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/9bcf50f3cc049dcdee1c90680947459acde6de1a"}]}],"author":{"name":"okozachenko","email":"okozachenko1203@gmail.com","date":"2022-08-22 15:12:59.000000000","tz":600},"committer":{"name":"okozachenko","email":"okozachenko1203@gmail.com","date":"2022-08-23 07:11:32.000000000","tz":600},"subject":"Fix user_verify_admin_password function","message":"Fix user_verify_admin_password function\n\nWhen ENFORCE_PASSWORD_CHECK is set True, password change is\nfailing. Because user_domain_id or user_domain_name are not\nprovided.\n\nCloses-Bug: #1728031\n\nChange-Id: I14aa1f1ee78f123bdbb6e3c4db314f96eec96f1b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/db14870fbeac7173300c3f76afd0fd81c355e6fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/db14870fbeac7173300c3f76afd0fd81c355e6fe"}]},"branch":"refs/heads/master"},"4478b904ddc20003552ef31aa592870a436af532":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":4,"created":"2022-09-27 14:48:13.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/05/854005/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/05/854005/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/05/854005/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/05/854005/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/05/854005/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/05/854005/4"}}},"commit":{"parents":[{"commit":"a0560b4bf5396988844520f76bc2fe457b6da8ce","subject":"Merge \"Fix \"Set admin password\" checkbox visibility\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/a0560b4bf5396988844520f76bc2fe457b6da8ce"}]}],"author":{"name":"okozachenko","email":"okozachenko1203@gmail.com","date":"2022-08-22 15:12:59.000000000","tz":600},"committer":{"name":"okozachenko","email":"okozachenko1203@gmail.com","date":"2022-09-27 14:44:59.000000000","tz":600},"subject":"[WIP] Fix user_verify_admin_password function","message":"[WIP] Fix user_verify_admin_password function\n\nEndpoint type is hard-coded in this function.\nhttps://github.com/openstack/horizon/blob/master/openstack_dashboard/api/keystone.py#L466\n\nChange-Id: I14aa1f1ee78f123bdbb6e3c4db314f96eec96f1b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/4478b904ddc20003552ef31aa592870a436af532"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/4478b904ddc20003552ef31aa592870a436af532"}]},"branch":"refs/heads/master"},"39945408de977c18b9eb0868e0638417d3ab9ff4":{"kind":"REWORK","_number":5,"created":"2022-09-27 14:55:49.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/05/854005/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/05/854005/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/05/854005/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/05/854005/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/05/854005/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/05/854005/5"}}},"commit":{"parents":[{"commit":"a0560b4bf5396988844520f76bc2fe457b6da8ce","subject":"Merge \"Fix \"Set admin password\" checkbox visibility\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/a0560b4bf5396988844520f76bc2fe457b6da8ce"}]}],"author":{"name":"okozachenko","email":"okozachenko1203@gmail.com","date":"2022-08-22 15:12:59.000000000","tz":600},"committer":{"name":"okozachenko","email":"okozachenko1203@gmail.com","date":"2022-09-27 14:55:21.000000000","tz":600},"subject":"[WIP] Fix user_verify_admin_password function","message":"[WIP] Fix user_verify_admin_password function\n\nEndpoint type is hard-coded as public in this function.\nhttps://github.com/openstack/horizon/blob/master/openstack_dashboard/api/keystone.py#L466\n\nChange-Id: I14aa1f1ee78f123bdbb6e3c4db314f96eec96f1b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/39945408de977c18b9eb0868e0638417d3ab9ff4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/39945408de977c18b9eb0868e0638417d3ab9ff4"}]},"branch":"refs/heads/master"},"73b60c3f8ffa190d5f98a82ddaa4f59483a0c356":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2022-09-27 14:57:07.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/05/854005/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/05/854005/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/05/854005/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/05/854005/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/05/854005/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/05/854005/6"}}},"commit":{"parents":[{"commit":"a0560b4bf5396988844520f76bc2fe457b6da8ce","subject":"Merge \"Fix \"Set admin password\" checkbox visibility\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/a0560b4bf5396988844520f76bc2fe457b6da8ce"}]}],"author":{"name":"okozachenko","email":"okozachenko1203@gmail.com","date":"2022-08-22 15:12:59.000000000","tz":600},"committer":{"name":"okozachenko","email":"okozachenko1203@gmail.com","date":"2022-09-27 14:56:30.000000000","tz":600},"subject":"Fix endpoint type in user_verify_admin_password","message":"Fix endpoint type in user_verify_admin_password\n\nEndpoint type is hard-coded as public in this function.\nhttps://github.com/openstack/horizon/blob/master/openstack_dashboard/api/keystone.py#L466\n\nChange-Id: I14aa1f1ee78f123bdbb6e3c4db314f96eec96f1b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/73b60c3f8ffa190d5f98a82ddaa4f59483a0c356"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/73b60c3f8ffa190d5f98a82ddaa4f59483a0c356"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
