)]}'
{"id":"openstack%2Fhorizon~855586","triplet_id":"openstack%2Fhorizon~master~I4ac098693bbac4894cfabf895303d0125fa37995","project":"openstack/horizon","branch":"master","topic":"zanata/translations","hashtags":[],"change_id":"I4ac098693bbac4894cfabf895303d0125fa37995","subject":"Imported Translations from Zanata","status":"MERGED","created":"2022-09-02 04:50:02.000000000","updated":"2022-09-06 03:57:13.000000000","submitted":"2022-09-06 03:56:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"855586-zanata/translations","meta_rev_id":"fa37df60b0d08dc0ff362cc0330509b299e21508","_number":855586,"virtual_id_number":855586,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-06 03:56:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":2,"date":"2022-09-06 03:42:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":1,"date":"2022-09-06 03:42:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-02 05:01:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-02 07:06:00.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"}],"messages":[{"id":"a6a507b18609ebee38fc136487f8e1bdd64615e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-09-02 04:50:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e1bc338b7d9546acbefc9114eec367e50c222ea5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 05:01:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c65c6d220f640889e03c50d31477187\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/4ece4b2bf07c4a6a8348eb19aca67483 : SUCCESS in 6m 40s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/68c1c5f5dd804a82b034a5b9d94f921a : SUCCESS in 6m 54s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/adeac4b022114ea39f0070d8c1c048dd : SUCCESS in 5m 18s (non-voting)\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/cd4cf7716e7d4c048c339421a956d154 : SUCCESS in 5m 15s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/e3a078158f264548830658365495ebc9 : SUCCESS in 7m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c3ce7f198de444729d945788ea98da6c : SUCCESS in 9m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9378817a59e4831935b5492d1b910f4 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8de2d6ca53704460bb86c540a6b33c51 : SUCCESS in 8m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e10fb958e7b14e76a6b35dbcc43e1ff0 : SUCCESS in 10m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/224b8b340d434460a824ece8537e9f48 : SUCCESS in 6m 11s","accounts_in_message":[],"_revision_number":1},{"id":"593b1f488eb46c68d9552c9eee3b765d61c90e7d","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2022-09-02 07:06:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fb31b11420bbcc3e64654f4cba44b2362c2907ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-09-04 03:54:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7cf9cc7a5c6b2b78dfa1550a41c660cd73c406a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-04 04:05:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b40c434f46bf4a37bf6cb5e3665af80f\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/6dab99f3711947cdaa7f5007d27ff9e8 : SUCCESS in 5m 57s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/d9d85320a7b24cabbfdfb6e1f17008d6 : SUCCESS in 7m 27s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/f30c2d6774b848768706f366996173b3 : SUCCESS in 4m 15s (non-voting)\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/dbc2ccdf11884b57aaba90999ac8503d : SUCCESS in 5m 20s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/1a4edc30317d487d8b40fe62d8096a2e : SUCCESS in 8m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3a3011a9003a428994158013641541f1 : SUCCESS in 8m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bdd11ce75e4419eaa450f7e8a27fcec : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/34fe23f1889d464497c5ff7b1ed248c9 : SUCCESS in 8m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/39ae13032cf545fcade5e2f0248b039e : SUCCESS in 9m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4edb6469893740f694f9a930a90082c1 : SUCCESS in 6m 29s","accounts_in_message":[],"_revision_number":2},{"id":"3999c3b2d86723306617707568adb9a3a830ce28","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2022-09-06 03:42:28.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e05462f7901dc5e0b53de969d6415cbeadfb88fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 03:42:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4a82848888deb926c3bebc213c60dc2407ff8837","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 03:56:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bc8c9e88dcc4a31b861254ecdd57df6\n\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/c12582063925441892f5ce1c67e04043 : SUCCESS in 8m 32s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/4b1a3a9e2b374e80a526a18cd306df38 : SUCCESS in 11m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7dd43850135a4fe69f2af4bf605a3149 : SUCCESS in 11m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87bb5576fec94c17ae1db9f4ac936f04 : SUCCESS in 7m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8572c7213e4e4f17b1474e383483266c : SUCCESS in 8m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/836972c95f774ed5b4da8fb01277e585 : SUCCESS in 6m 19s","accounts_in_message":[],"_revision_number":2},{"id":"01cbb38c67ed2cd29bc53b8821af5006a13cd416","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 03:56:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"fa37df60b0d08dc0ff362cc0330509b299e21508","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 03:57:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f372d871b2a64891aed317f418591e2d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79093ac518d94deb9a1bdaf29ed30297 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"58f83295377f566a764b23e954a3eb402650a416","revisions":{"a38cff7260db35b456496cd572ff858037f09e7b":{"kind":"REWORK","_number":1,"created":"2022-09-02 04:50:02.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/86/855586/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/86/855586/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/86/855586/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/86/855586/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/86/855586/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/86/855586/1"}}},"commit":{"parents":[{"commit":"0a77f4215667c524ca3912ea532f5fa1e0c9adb9","subject":"Merge \"Add pagination to Networks table in Launch Instance wizard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/0a77f4215667c524ca3912ea532f5fa1e0c9adb9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-09-02 04:49:58.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-09-02 04:49:58.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I4ac098693bbac4894cfabf895303d0125fa37995\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/a38cff7260db35b456496cd572ff858037f09e7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/a38cff7260db35b456496cd572ff858037f09e7b"}]},"branch":"refs/heads/master"},"58f83295377f566a764b23e954a3eb402650a416":{"kind":"REWORK","_number":2,"created":"2022-09-04 03:54:28.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/86/855586/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/86/855586/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/86/855586/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/86/855586/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/86/855586/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/86/855586/2"}}},"commit":{"parents":[{"commit":"3a2e82ab75c605cfd6fd8d26d7dddd3cce1f812d","subject":"Extend settings with OPENSTACK_SERVER_DEFAULT_USER_DATA","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/3a2e82ab75c605cfd6fd8d26d7dddd3cce1f812d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-09-04 03:54:25.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-09-04 03:54:25.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I4ac098693bbac4894cfabf895303d0125fa37995\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/58f83295377f566a764b23e954a3eb402650a416"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/58f83295377f566a764b23e954a3eb402650a416"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
