)]}'
{"id":"openstack%2Fhorizon~856308","triplet_id":"openstack%2Fhorizon~master~I076a97a6f943ab74a2db8bc5179a7db194009db4","project":"openstack/horizon","branch":"master","topic":"bug/1827120","attention_set":{},"removed_from_attention_set":{"29313":{"account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"last_update":"2022-10-13 05:35:07.000000000","reason":"Change was submitted"},"14567":{"account":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"last_update":"2022-10-13 02:15:55.000000000","reason":"\u003cGERRIT_ACCOUNT_14567\u003e replied on the change","reason_account":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"}}},"hashtags":[],"change_id":"I076a97a6f943ab74a2db8bc5179a7db194009db4","subject":"Fix app cred create without project_id for domain admins","status":"MERGED","created":"2022-09-07 17:34:15.000000000","updated":"2022-10-13 05:36:12.000000000","submitted":"2022-10-13 05:35:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":22,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"856308-bug/1827120","meta_rev_id":"fe30b7909417d6037481ce8ed0443ff6253bbaef","_number":856308,"virtual_id_number":856308,"owner":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-13 05:35:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":2,"date":"2022-10-11 07:44:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},{"value":2,"date":"2022-10-11 07:43:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-10-11 07:43:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":1,"date":"2022-10-11 07:44:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-07 19:17:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-21 20:52:31.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2022-09-28 14:00:06.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2022-09-30 08:15:04.000000000","updated_by":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"reviewer":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"state":"REVIEWER"},{"updated":"2022-10-10 11:31:07.000000000","updated_by":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"reviewer":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"state":"REVIEWER"}],"messages":[{"id":"4678df78f5275068132e094fc3670e5ddbd8e1b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2022-09-07 17:34:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ccb85a03c191835654b4f37ec66408eb222cda6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-07 19:17:33.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fd067b42215d4f29b583b4caf9f9e6c7\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/1967ce3860c647399e326a87f53e8972 : SUCCESS in 6m 40s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/1083351107a54b3ea7195fbab83a5e8e : SUCCESS in 7m 16s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/6970da2bbdc14e49893bc48a7906a8fe : SUCCESS in 4m 35s (non-voting)\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/930a168ea26346039d389dc784958556 : SUCCESS in 5m 26s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/ef99f1ccd23b4ef8900bc661432d0c8a : SUCCESS in 9m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b7d0dfac8d7c4111ac5d498d15dfe858 : FAILURE in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a344fdbece84733823285769e8c1d54 : FAILURE in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f61169f7cf24575a426c29f33ddd91c : FAILURE in 6m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cf0703c26f634851ba3d7349760751d2 : FAILURE in 6m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b41fea7d394466a93c933f541ea0d26 : SUCCESS in 7m 08s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/e54174348cbb41b7ae535fba8180db83 : FAILURE in 6m 41s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/2df76ec279d747c188afc58c882c3fac : FAILURE in 1h 36m 11s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/b9b75301ac4b4d7a883b7d98034de0bf : SUCCESS in 40m 59s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2c35ff37a7984202971d29c588822a2c : SUCCESS in 5m 04s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/8c81c5f74d4e416e932fad8e8e40f542 : SUCCESS in 38m 11s","accounts_in_message":[],"_revision_number":1},{"id":"7fa8fcbdbeb9565f76fc36eb37f246e2962891e9","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2022-09-07 20:32:54.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6f58d77f4360db6e5094fbf702f895b7c64c335d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2022-09-07 20:41:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6d6f2fa0ee90a193787972fd26137711415c257b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-07 22:07:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c983af99d09a4c378bf8a9643dac7c6a\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/3eef90a713bc41b3845f2497c704f098 : SUCCESS in 7m 27s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/80fba8c1cc884cde986db37fbf353bec : SUCCESS in 8m 28s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/e779c5779f92493cb66e71edfc371d8e : SUCCESS in 3m 51s (non-voting)\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/4e30d5cc1f514d6da58e6af0ae109110 : SUCCESS in 8m 26s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/b0208a3e205641ad99a29498317d3273 : SUCCESS in 14m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d076536e2f664d728cc187da137fd629 : SUCCESS in 12m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa05650be48740c3b11997236e3c32db : SUCCESS in 7m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/736b0702a37b4fda8390791cbf831daf : SUCCESS in 12m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/99e7719cff314c30a2e62974d91a5aff : SUCCESS in 12m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b03027170b244c39500a8af38372861 : SUCCESS in 9m 49s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/aa735fb7ca874d08b24058394e61f0ab : SUCCESS in 10m 03s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/c8d8ee9a5d534189bd46338987c6682b : SUCCESS in 1h 23m 59s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/3dd03e4d4e19439ca2e23e2359346c83 : SUCCESS in 26m 49s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f0d3d5cd8d2c40eb91c67531706c96ba : SUCCESS in 7m 50s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/d5912371e1ae48f0932cd4f1a80f35b6 : SUCCESS in 27m 29s","accounts_in_message":[],"_revision_number":2},{"id":"dec9a37092c56168031c14e76a20bc04f704f203","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2022-09-21 20:52:31.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"0c57f7e8e2b66fabc93622fbfba091e434b8d5fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2022-09-28 21:18:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5628ce9a1f813e7b122b7568047324ea782c86e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-28 23:02:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7e025a768934e0fa5f5f6283636254a\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/c8bc36cfba714ec98cfbd652297b33d6 : SUCCESS in 8m 29s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/382175de9543431faa1176db7ec8e740 : SUCCESS in 11m 14s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/71ff15cd9c324b0287a866fe8fb57dce : SUCCESS in 7m 19s (non-voting)\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/8aba87d64c2b442ca143dafeaf1909dd : SUCCESS in 8m 08s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/0c6c5908f91e47cfafe86547deb2652f : SUCCESS in 10m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c058db83591843a5b7524a95b5a7c237 : SUCCESS in 11m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f127bf4836ae40d7954b08c05bceb866 : SUCCESS in 7m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/80f32b945156457dad58692bebb60812 : SUCCESS in 11m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00f4b01b95dd4325bcf17060a23b52df : SUCCESS in 9m 09s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/1c63dd6c33e745c3a523f72d218a1630 : SUCCESS in 6m 50s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/74d1be0559c346ed930853206bbef22f : SUCCESS in 1h 41m 40s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/330a1cff703b4633b08d36c3881a554f : SUCCESS in 38m 24s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0737b8c4c0144b6087fca36a8e634fb6 : SUCCESS in 6m 26s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/e179c997b52e4bf6b273356e501b8a4e : SUCCESS in 38m 04s","accounts_in_message":[],"_revision_number":3},{"id":"e0fd4ff51ca9d33572b2af0b9a9ca65fa1c5ca88","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2022-09-29 12:28:59.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"01dc06da309502b70cafee4ecbf72764535f7c95","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2022-09-30 08:15:04.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"abcec0126a998253ecbb2c6362dce2a50d7bf9fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2022-10-04 19:28:51.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"af9c91a1ba0c83ee7121d6c02ac8f2385d2d003c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 21:16:25.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3beb253f47804c44aabe87ab8923cffd\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/f84b1e482a26442f8ed9f519726a538c : SUCCESS in 6m 31s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/cf350cf50e474705bcf5cd8f8b8a6c71 : SUCCESS in 8m 13s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/0f93ad50d90c44e9894e363f7e318667 : SUCCESS in 4m 43s (non-voting)\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/dace97c8ccd84f1c88734cfe22d6e5ad : SUCCESS in 5m 16s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/e9049a06a1ba4ed1a8e0de13d6af2e9f : SUCCESS in 9m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/388e2d55beb94532a8db6f74fb5ed9bf : SUCCESS in 11m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbe933a373fd4a1c8b8dd5c3d2e2419e : SUCCESS in 5m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5cf00a1731b544049ed54dea57851e19 : SUCCESS in 8m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7616daec35774a29885268ec59cabf0a : SUCCESS in 6m 12s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/5d821e1b571f4b6bbaab1381827016eb : SUCCESS in 6m 34s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/794245935d83433ea53a728dbb37da1a : SUCCESS in 1h 46m 15s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/13d4a57f704c4351aa7f8f0013d793ba : SUCCESS in 22m 49s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f453eaed128b416eb0941cc81bd117c4 : SUCCESS in 4m 45s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/a2050c5b852344aa8ebdd559330fe3aa : SUCCESS in 39m 56s","accounts_in_message":[],"_revision_number":4},{"id":"90828527c722f7228b308e778f77bb0dfce9f01e","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2022-10-05 15:01:23.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"a487bb4e3f02c0f67dd8ddb80f776ce416cc28c0","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2022-10-06 11:23:43.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"ecda99291b9d0b8f79e4f757645bc25f9efd1e0f","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2022-10-06 13:55:08.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"60c97ddc2fb2f7d53ccf7d1811b461b3dfc1e116","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 13:55:41.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"77c7f34e93fc584751804b1a648e1d66bebb85c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 15:13:56.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/37c08eb7ea734d24a2c847b3de28a00d\n\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/c5d4ff064caa4662b5e0af23594df0c7 : SUCCESS in 5m 13s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/45ab0b9d585b42338306f478a5ca29c5 : SUCCESS in 8m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e7e8d4206ce5499a8b805367423e196d : SUCCESS in 9m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bae094fde0e43519d23a79784e05fa7 : SUCCESS in 5m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cf567313855d43e0b53e2a8815986d2f : SUCCESS in 10m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3174ce6df3b424dbf158ffd399e54f4 : SUCCESS in 6m 44s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/052f9ea7aa1b42498cae649998ed586a : FAILURE in 20m 48s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/ceb190204b49429594ae2d4f8f7b0e21 : SUCCESS in 7m 28s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/732abd2c58fa4015868f4b0fd99b6b55 : RETRY_LIMIT in 20m 31s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/737d104c90e74116aa029fb0793937d8 : FAILURE in 21m 52s","accounts_in_message":[],"_revision_number":4},{"id":"70823620420a8ea626046d36c929dd1b01aa44c3","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2022-10-06 18:12:48.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3fa5c20857c2c339a15f5fe14f2f27fe7089ee23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 19:34:09.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ff836d1f8e83442fa6f1a0b59819b395\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/94483b85a48c4e98ba1f02e12c9b89a4 : SUCCESS in 6m 17s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/d1e458ffca1e42ddac75c0ef510be374 : SUCCESS in 7m 26s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/ac8184e9d8dc451092d94bafe993aa50 : SUCCESS in 3m 58s (non-voting)\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/39558b9d9ee04ffca3f992a8a59d7098 : SUCCESS in 4m 38s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/53d87582ac7d48ea9619fe95d2266995 : SUCCESS in 8m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/13e85429be2d44218f44942bcf0b3244 : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d46ac14d40414e678a4d2323ae3c6a12 : SUCCESS in 4m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1c786b9afbdc41058c68f273b8d67c93 : SUCCESS in 8m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3042ce6874748a7991c338db8b5bdbb : SUCCESS in 6m 42s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/78157afeb49c47e38bee682414b9e65b : SUCCESS in 6m 29s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/ccc5fec3e13c41f892cf7ea5b4aec254 : RETRY_LIMIT in 17m 20s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/a729407bcab645cb9a3e82dcdc58d744 : FAILURE in 18m 55s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cbfac28f8590447590150d8889bab62a : SUCCESS in 4m 17s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/7a2ef1e909744f25a5dba6cd20beb922 : FAILURE in 18m 54s","accounts_in_message":[],"_revision_number":4},{"id":"8f1a12e92a3116be97f4e80cc318a28583c17f66","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2022-10-06 21:49:29.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1e96792e2f43889933b76edc5eac4294f33d2a74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 23:53:46.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/544b2766ca8e41bb981f1377f3a711d2\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/cb776ba25fdd4761b4c4112021d68e39 : SUCCESS in 5m 39s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/aab6e595f93a4290a989209a6d04ab07 : SUCCESS in 7m 55s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/35baa06097b14d3980165cdbf7e60eac : SUCCESS in 3m 57s (non-voting)\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/e272bca1ac404fff827141eb5d92b32b : SUCCESS in 4m 45s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/7a30b4f4332a462a8bedcca9b66825c7 : SUCCESS in 8m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/38032c1d25634f599c610a4c273b1f48 : SUCCESS in 11m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29005347020a4cc78c35233b5c520f73 : SUCCESS in 5m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5f2f4a9e584d46b0966ccfaa442f84e3 : SUCCESS in 10m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7235af3c61184b96902fd7d626b3928e : SUCCESS in 6m 29s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/a17ba637f7734cbb8979fc018cbec99f : SUCCESS in 6m 01s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/845f3a8edc07496da198888834ff113b : FAILURE in 1h 57m 46s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/10cbd03188a64d348f65634099a77c65 : SUCCESS in 40m 52s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c8baf94f60c045ccb0f7e5f1b0ea7902 : SUCCESS in 4m 34s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/2cd3db16d48b4688ae44f186c3077f9b : SUCCESS in 32m 46s","accounts_in_message":[],"_revision_number":4},{"id":"f34d33a54908ee3b898dee81e9cd047d2bdcf6c0","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2022-10-07 04:56:14.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"126b8584afca3cdc64798e486c1e983b295401d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-07 06:40:45.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b1492d3164e3457dad571a1ede6e157f\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/c2e0332646d94841b408380166e511a9 : SUCCESS in 6m 34s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/196c7c479f96441ca61d385d711ce0d1 : SUCCESS in 7m 23s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/fb2e5e586b7844fc9b1e7341966eac32 : SUCCESS in 5m 03s (non-voting)\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/d990ce62d3bb44f2980c67bac7e419e9 : SUCCESS in 4m 23s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/07ab2dac5c44481198e7aad4619ce49b : SUCCESS in 9m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/88f8b12cada942059469b5a069c71259 : SUCCESS in 10m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/805b83a21b024179aa8f3a026e90ec25 : SUCCESS in 4m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e3040680b05a4d2cb14897d2796e917b : SUCCESS in 8m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6d45f81aaa64198a8ba72878ccabfc4 : SUCCESS in 5m 10s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/73b9f29b47994981b78741b9b465ea01 : SUCCESS in 7m 03s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/8e9abcc647664e4da4a9f06ea14e22a7 : FAILURE in 1h 43m 21s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/ab88713f6a8a420f860366459e0ffa55 : SUCCESS in 22m 20s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/720e600f6a86423f93eaf6a699e3fbab : SUCCESS in 4m 20s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/be6242a9fc164a95963bb22f262af999 : SUCCESS in 20m 58s","accounts_in_message":[],"_revision_number":4},{"id":"9a7e4c797bf90d8e8c306b99eae2705240f1a73e","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2022-10-07 12:35:28.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"552db31676a18378fa6efcfab43901b8ee95d9ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-07 14:04:07.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ac80169c079e4c78bf214f8ae9c108a4\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/032cd97cbec14fd785f9c17ba91dfb26 : SUCCESS in 6m 34s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/577fe125d594492db60d255463ce5b50 : SUCCESS in 6m 52s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/cab924960b254b57afebbfa10d3f3063 : SUCCESS in 5m 08s (non-voting)\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/c0d92ad60b2245eaa9246e0f14cb8729 : SUCCESS in 4m 49s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/b6846bc5e68c493c84dec23fb32e7649 : SUCCESS in 9m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b53753d0708b43e78566341b28c12293 : SUCCESS in 10m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6d8a93c73d2435397d8a46679bbd258 : SUCCESS in 5m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5c32e86445954f5bab92c2025e8f6a4c : SUCCESS in 11m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9540b8fbd1c743fea8e3f62e3b669679 : SUCCESS in 7m 34s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/17acf2ca448342a09a72592063b1f697 : SUCCESS in 6m 35s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/f17dc48da41a461089cc18308a9fcf7c : FAILURE in 1h 26m 26s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/2edf4850856c443dbc46c80194e5eb7b : SUCCESS in 35m 55s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7af371a977534cf4a088ed244adc36d7 : SUCCESS in 4m 01s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/26d37c748ee04687b9b5cd1fb46889d7 : SUCCESS in 25m 08s","accounts_in_message":[],"_revision_number":4},{"id":"df5e1442eda028469f6f33a1ba4e147ddf4d2a2c","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2022-10-07 16:53:18.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d64b22eebf7d174090b38cc34e1a22f850246d37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-07 18:30:04.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bfdeb1bf08bb40be979c40bd6118ae84\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/00d50897fd8a4aec8944d755da441d87 : SUCCESS in 5m 53s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/2e47670adbc84ab787c0a591a73e6dad : SUCCESS in 8m 10s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/edc7918b318843c890ac4d3656dad8ba : SUCCESS in 3m 35s (non-voting)\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/1c76364d0f9d4fc3af048f076144d243 : SUCCESS in 4m 17s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/634f806f4c4b4af490dbdaada8b016ac : SUCCESS in 9m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/548d029aeac34fc0bfcb2dc108ea61b8 : SUCCESS in 10m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36f4a972c47645798fc4d80c20bf5ca2 : SUCCESS in 5m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f0d1aaadb75c45efb05de48127176d48 : SUCCESS in 9m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/050287ed8f864c08b334bdcde3ea69c8 : SUCCESS in 5m 54s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/2dd60c8bcfcb4aa99079d49e6f354406 : SUCCESS in 5m 49s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/67d3a3dbd7d54a29bb27711de30e80f2 : FAILURE in 1h 34m 52s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/ee5eecf7fb154d94859486a1c26d4818 : SUCCESS in 31m 59s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d611a79da0c049cc89fea7b09527ed4e : SUCCESS in 4m 48s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/6c059b732eaf4ff6bcb85ba47981fdac : SUCCESS in 35m 14s","accounts_in_message":[],"_revision_number":4},{"id":"b099be024e7683db5d6970931c6044f7c46aa432","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2022-10-07 19:14:21.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ddc45545f2acd87926d58dc10d8f4dc1eae3aeb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-07 20:56:10.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a4808fc001944dd0be5669d1b5cb908c\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/6e7c0ec5176b4f37aff8dd8cbb8b3960 : SUCCESS in 5m 10s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/a340cac439624f4ebcbb5311ca19697c : SUCCESS in 7m 01s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/3c57a36c9b404399aa87d8b927f54a20 : SUCCESS in 3m 57s (non-voting)\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/dcebae2c5d8248b4821b97e74baa558a : SUCCESS in 4m 42s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/a6ff5f0e85874cc1923b34ad48bc7438 : SUCCESS in 9m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/361e76817e4f4cc5aa2004d496c17f39 : SUCCESS in 9m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63bc68a7cf6d496195de50127d8047f9 : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cae2bc8b23044c6da38735d8564ae344 : SUCCESS in 11m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53b5689763434267a7057def5a53e514 : SUCCESS in 6m 00s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/90b891b166114a57acbc3e20622a416d : SUCCESS in 6m 28s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/eb1638d5881c4a71be83542c69274b5c : FAILURE in 1h 37m 13s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/cab016de7b7d4d2280423627504ea39c : SUCCESS in 37m 35s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d03f7615dc694a118b566ed08b415dc2 : SUCCESS in 5m 07s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/85e88f02abf54534b48e78f364c9a1d2 : SUCCESS in 34m 58s","accounts_in_message":[],"_revision_number":4},{"id":"146b9809159095b87a4512a3a4136364dc82ee51","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2022-10-07 21:21:55.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7973107ea830f0f29b07f95306414b954892ae53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-07 22:58:07.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3d7b7856e765428c8d3bacfe1fc12a69\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/5dc107d3dc8d49f3bdb07881dc9fc25e : SUCCESS in 5m 59s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/7f69670b72854890a8b34d9294d53501 : SUCCESS in 7m 36s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/4445ef990dd846b6b358d6aead05d2d4 : SUCCESS in 4m 16s (non-voting)\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/6e67e3776727431681e738374cee8c06 : SUCCESS in 4m 57s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/1b596786853d4973910f3c196f35c76b : SUCCESS in 8m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd92db0661b249c9b68c46d23ffb98fd : SUCCESS in 10m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ef5a242d6ae4c1cbbc73b96da3c6e79 : SUCCESS in 5m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/738ae395ae52483faa4efc2a5b23de78 : SUCCESS in 9m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36a5ba293a5743b4bae653c660c8dcc2 : SUCCESS in 7m 45s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/50fd6f5dd08244c897ce70a4b2d916dc : SUCCESS in 7m 13s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/1c68c9587ac74836a4be358345235128 : FAILURE in 1h 34m 57s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/1e82ab24028a43dc9f71d0778c57a90b : SUCCESS in 29m 37s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d1e1e576e8cf45cb85d8e2703c43a8d1 : SUCCESS in 4m 42s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/52ebbc38795f41e3821842d89e7194ee : SUCCESS in 27m 51s","accounts_in_message":[],"_revision_number":4},{"id":"229f2ffc55504a1ae4cc32fc0f2c12e9649c4841","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2022-10-10 06:58:05.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"af915aa5718289e0d94522a48a1db0844826910e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-10 08:38:25.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c90b5895dec841c8a5502269269c6f7a\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/f0281d66b4ca4cf582b9614c35b5d79d : SUCCESS in 6m 13s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/cc0495c77d044d928ff1052af43667a7 : SUCCESS in 7m 19s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/618b01f7a89a4e1faa1ec55145d211e8 : SUCCESS in 3m 52s (non-voting)\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/a15fbdb8f3da427f9e77ca2fa729330c : SUCCESS in 5m 00s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/6cf5f79e81384072a84fd6f51c87058e : SUCCESS in 9m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a97b59740ed469394716391b787dbf3 : SUCCESS in 8m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adfe435555d24faf8247608b9a3c5ae6 : SUCCESS in 5m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/91f9efb9de7e4efbbe09d00256619f22 : SUCCESS in 8m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d788492cc874a4789e083547ff3b2ba : SUCCESS in 6m 49s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/315b6abd81bd49db81a14105f8cf2e97 : SUCCESS in 6m 37s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/cb0d7d33d26b47e287741e67e245d980 : FAILURE in 1h 38m 57s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/5beedc7183df4951adbcfad49002405d : SUCCESS in 23m 33s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c82a519df7174f9d93532bcec30d8cbb : SUCCESS in 4m 38s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/9d34b03232bb420eb725b6e4024afe5d : SUCCESS in 22m 11s","accounts_in_message":[],"_revision_number":4},{"id":"f74eca5202482d5a070caeee423c1373b1bd0495","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2022-10-10 10:57:38.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a2a68aa2ef1cc4e6137de6597d35218bddb5f77f","author":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"date":"2022-10-10 11:31:07.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"d42f02f9f2ad619bf8206dde907db12795fa2501","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-10 12:43:10.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fca467dda97a4033ba4bec640eb0003e\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/c297159353b248be851ee6609f8ded8a : SUCCESS in 6m 33s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/8c857483e79445b5863059d4a4dddfe8 : SUCCESS in 7m 33s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/1477b411208b43f4adfa3c22f339637c : SUCCESS in 5m 11s (non-voting)\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/fdfa0f8cbcc241f995e1b07c95a1e084 : SUCCESS in 5m 41s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/b03f9093d0c84aea844d31ce237e02c7 : SUCCESS in 6m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/319f28dd8bda4dfd81d15769ac963120 : SUCCESS in 8m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af590d9b0d394aa69472234a0129002f : SUCCESS in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/645f3069b8084de08e357deced047c12 : SUCCESS in 8m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e72ecc41d614a999239bb0893c7cbe5 : SUCCESS in 6m 52s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/1efa53645ec94911a919d657a6eba829 : SUCCESS in 7m 47s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/963a374c643b42148730aa5210d333c4 : SUCCESS in 1h 43m 39s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/065a240d05c34ab096b046170484d6d9 : SUCCESS in 29m 43s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0450dedada534d299c8314ffd891ef49 : SUCCESS in 4m 32s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/9ed8b414a267450298c274f9eb1da14a : SUCCESS in 27m 54s","accounts_in_message":[],"_revision_number":4},{"id":"0e12d98924875984534231a64f7ace9e5596e359","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-10 12:43:44.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"40a5dbc396bd76d64300e4f34560d41f856a611b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-10 14:23:12.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1256a5e05a0a4b88bda492402db99287\n\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/877aabebcbf34feb83db89103c394058 : SUCCESS in 6m 18s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/d25d3f3c23e5488880e0d1a7a45e7e37 : SUCCESS in 8m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af22abd54bac46f1a4c882c80d07b6e9 : SUCCESS in 11m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ff1e6f46b584f24836b857d80704f73 : SUCCESS in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5a16c7fdf7184d319fa5715c86d15eba : SUCCESS in 8m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/081c4f18ee8d4cbfb4afe7ebb97c415f : SUCCESS in 6m 47s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/a2c39a3a045945a4a01db4abc0f161dc : SUCCESS in 39m 49s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/c63c466fa3fa4ee4bd127ae3123e3319 : SUCCESS in 7m 13s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/746705dfd24d43cdab8b103b2201d9d6 : FAILURE in 1h 38m 54s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/8fe34bb1479d4881b3edb5ec20bf3a06 : SUCCESS in 35m 34s","accounts_in_message":[],"_revision_number":4},{"id":"513c0efc02b3e3e7e554a735072c71257c372b5b","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2022-10-10 14:47:45.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"822dd8353d039b92e2420c0c9a56b4d081410bc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-10 16:10:04.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5d80f07c0aaa4c8f9dcc2b3daf557c6f\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/d134c19709e945e6be02e93bbb56d06f : SUCCESS in 6m 17s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/6c8dee79c4874c95a59e836cb444f90f : SUCCESS in 7m 10s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/54abdbecd77b4ddba6097075336cf919 : SUCCESS in 3m 31s (non-voting)\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/dd24fce4b4224dec8f3b1694604e42f3 : SUCCESS in 4m 06s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/3aff5f481aa345ccb9f8f651b2e5f83a : SUCCESS in 7m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e99225c115946338f7c2c9e24a298ff : SUCCESS in 8m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd45f0cf6c584ebb8f92a838509a34c6 : SUCCESS in 5m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e4587b3137874278abdf8f4a7fe9ed1c : SUCCESS in 8m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c00dc0be46144a9c844ac48d82cd5fb1 : SUCCESS in 6m 14s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/218d3faf5cfe44ab986151f00ac6181c : SUCCESS in 6m 04s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/622151d0909c496e8a922cd5ce1f935b : FAILURE in 1h 20m 50s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/3a42abf162f546598a3bb47d9b676b14 : SUCCESS in 23m 33s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/329399f20539433aa59634a306365de4 : SUCCESS in 4m 07s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/69a3665a2eda4a1ab8d1452250c5e5c7 : SUCCESS in 21m 44s","accounts_in_message":[],"_revision_number":4},{"id":"60783497b6ea7d6c8b713231792257cb0e4e0b88","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2022-10-10 17:33:52.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5d1f32a5821d8bb21d225343499a505c7835d6f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-10 19:07:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19b4384b6d434be39c4e9534b3339839\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/d0f03391582446bcb5e3f36ed0da5e33 : SUCCESS in 5m 17s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/7cbea1befeb3418192feaafc61c1c65d : SUCCESS in 7m 12s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/ad220237c33a4f6fac63c0f307bfcf80 : SUCCESS in 3m 40s (non-voting)\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/8e5d7e9e93a54f6b99817bc0c35234d5 : SUCCESS in 4m 35s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/0b7e9548bd77431a8f2e6fbaecbabbeb : SUCCESS in 9m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/394644a5c5c4411b8be01a79e0143944 : SUCCESS in 10m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1705187fe1d94a76be98266aea1b0ad8 : SUCCESS in 5m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/061ee9ce62144f0ead18d917a330d7d1 : SUCCESS in 8m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec35cfbf35504e7d9e8b56ca7a641002 : SUCCESS in 5m 46s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/118b4a2eb38e45e5ade204b5c9946c6a : SUCCESS in 5m 48s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/5b714ebd57f049959989eb188ddb33bc : SUCCESS in 1h 32m 16s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/062a66a2cb8b4015a7161479963730a9 : SUCCESS in 29m 22s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c5f404e5d81d4b6b85862d5a6559d768 : SUCCESS in 4m 28s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/78ffbf4b28604a2cbd19cfe13d69d5b4 : SUCCESS in 21m 39s","accounts_in_message":[],"_revision_number":4},{"id":"8bfb93ea2df72015896ac9a25913d66918536bbd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-10 19:08:21.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a5c597b63af9ac3295309e2b497f8931cad6416a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-10 20:53:53.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6915698275cd4effaa3a2eb67e2609fe\n\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/ccbe78ff34ca4a9695b9fd6d237a9cae : SUCCESS in 4m 58s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/3c3ccbb304e74a98bccdfaca8b2fea56 : SUCCESS in 8m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad79b7b2cc4d4ca2aabae73cb31374bd : SUCCESS in 8m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb0bda08eaa14914ae1f02188735b145 : SUCCESS in 5m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f6c42f2577b6439f9a5f752b35f4b837 : SUCCESS in 9m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea7bb33134de4bbf9cc81fb996b373e9 : SUCCESS in 5m 17s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/9f1eb90dce5743f79772ac31b4d6f6ca : SUCCESS in 32m 53s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/b3ad5b2e7c3a42b38c1bb19736807ce4 : SUCCESS in 6m 02s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/737d81bc949c4513847bd50c73074496 : FAILURE in 1h 36m 07s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/5b355e61359b4ed4b8b49b1712346f1b : SUCCESS in 19m 25s","accounts_in_message":[],"_revision_number":4},{"id":"eb62fb9da9314fe220b7439073e945c83d744541","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2022-10-11 07:43:06.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"c05e54fa3d1f97d0950737448d1a1df95e316d27","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2022-10-11 07:44:13.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"670ddd3b40a625b591cc0993873e6d89424ce1e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 09:33:42.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a6c0f1fd8fdd4d29a299a85177f9b474\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/c367c2bce25f45f7a07822a38e52ad63 : SUCCESS in 6m 09s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/3c4e6f241f7a41598184a442d1c63551 : SUCCESS in 7m 42s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/89b9059aa4754a2da96626497ba8dc07 : SUCCESS in 4m 41s (non-voting)\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/d577c89bd38b49d5a31a4b6b9044ec8c : SUCCESS in 5m 02s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/c7560629cbc8489ca1e1233cd3cb01c0 : SUCCESS in 8m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a50cb2740fa47849e7fa4b49d6b7a84 : SUCCESS in 11m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/244189fd662a4e8eab0e69a8c016b4e3 : SUCCESS in 5m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/92dfed903bf0464eb616a6b2a8941071 : SUCCESS in 10m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cccb256be304fe096c18f2b80b2d203 : SUCCESS in 5m 39s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/12e3a269c7d44ac99e8b037aadbbef2e : SUCCESS in 6m 59s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/76ec6fbf74814e02b2dc1b3d385f0b4c : FAILURE in 1h 45m 47s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/b02b0c85478f4d679a2f593b582a4cbf : SUCCESS in 34m 10s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8d813f3f3a604aa98944b091aac24801 : SUCCESS in 3m 46s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/968d7a2e81784bd6aa96474e317ba627 : SUCCESS in 35m 17s","accounts_in_message":[],"_revision_number":5},{"id":"4eae455de68fb535a575e83898ebb56ac19f74ed","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2022-10-11 09:34:42.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9f3498b176ac88743ef28dcd96e5ab460f1ba7c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 11:07:40.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ad8f1a9a12734b40a088cf75288b8c42\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/f258e8758ba74911973aecf615fb1c5a : SUCCESS in 4m 54s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/c945e26981724c1aa1909190158785f8 : SUCCESS in 7m 08s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/6ca52191a3104b2399a7850797de60b8 : SUCCESS in 4m 27s (non-voting)\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/25f80b0292f2482ab885a7a3b5e06f21 : SUCCESS in 4m 34s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/2ff3de404f11483a8d504592a3853e55 : SUCCESS in 6m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d75aa98e7f034e1586fb2296d003e6b4 : SUCCESS in 8m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/389f6e036090427c9d3a909d6e6630d5 : SUCCESS in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/66865083c79647968e9d59fad4c4dc34 : SUCCESS in 9m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfc539b533894d2fb7b46f1813754aee : SUCCESS in 5m 36s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/d7480f9947ca47e9a188abb27b8cc3d9 : SUCCESS in 6m 44s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/1ad05492f45544f0a55a13f694b6b44d : FAILURE in 1h 18m 39s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/d6359d1f7e9f4b1eaa8e43f2500e4bf5 : SUCCESS in 23m 34s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5e8398f1451849cb893486882c1ed7b7 : SUCCESS in 4m 28s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/74a625d52b9146d9bd6ec127b988f421 : SUCCESS in 23m 16s","accounts_in_message":[],"_revision_number":5},{"id":"91ca2aec393cae34efcb62c57f36dba04bcd4639","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2022-10-11 12:51:43.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2694346b685b7be541ada9c391647f885ada1830","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 14:42:28.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bf0e11987e944679a13798ce774cefaf\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/35c2ba6c1a63483497e846e88776f60b : SUCCESS in 6m 03s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/cdf1e506908e4994b838dfe2e35eae38 : SUCCESS in 6m 43s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/15bdbc22ac584efc91edfd2d976dec42 : SUCCESS in 4m 10s (non-voting)\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/1be881a67cc94d8e9b2fbbca3633fb6a : SUCCESS in 4m 51s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/fdbe0b7f21954722ae64dec18766e4e0 : SUCCESS in 8m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e6d55f13828446c8913b65c481877edd : SUCCESS in 11m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7eb5965a8a024f7a8585b00d75b754b8 : SUCCESS in 5m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3ae881f81e20476f921664ebfc181281 : SUCCESS in 11m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3bf2f1628da43b0a2a5e0bf5f7077a5 : SUCCESS in 6m 06s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/f678f124159c4b61bf2c4d10f291b8dd : SUCCESS in 6m 51s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/93bf36a9973948a2a54c32d2a4aab021 : FAILURE in 1h 43m 55s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/710b87191a8642fb8300cd8f72078a15 : SUCCESS in 33m 58s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/08be2194f9174f9b8f35d894ac1650b4 : SUCCESS in 4m 12s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/e9fe42049cad4182bcbeffec23fa0f53 : SUCCESS in 35m 59s","accounts_in_message":[],"_revision_number":5},{"id":"126bd7601a83db79b07573d85938fd14ebd6e897","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2022-10-11 14:55:08.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"aae1ccaeabcec4aa754474f75857fef29d91a87b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 16:27:24.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cae12866d925485cbaf9ed86170f4c5a\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/b90f599d6cec4fdd890f480ef928aed5 : SUCCESS in 6m 47s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/3073b7c5a39c419a970fe498d8bcd5e5 : SUCCESS in 8m 03s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/5b9e78c45f1f49d79bec8bc365e6c3e0 : SUCCESS in 4m 11s (non-voting)\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/0b8486f7a1cd46ea86e3effaef2f904a : SUCCESS in 5m 13s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/f21b0ff281d14500b6ccb1f50299edfa : SUCCESS in 8m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f56c2d2a2755426ebcc304a3e305a484 : SUCCESS in 9m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d24574bcfbf043adac338ab656741fe8 : SUCCESS in 5m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5c191062d5ad41bcb2ffa79654421979 : SUCCESS in 10m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd050ec7aa5447c582122ee9f2f3dd18 : SUCCESS in 6m 57s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/a7735f3b0c81401e8f42edf4bdf303f7 : SUCCESS in 7m 15s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/3945bf48c0a04c4a8d003ee5f02158cd : FAILURE in 1h 30m 27s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/69e5187ed5c3424cb9cc6756fae3d869 : SUCCESS in 30m 43s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/300a10bf8f4044f2bc6d7929a65c245c : SUCCESS in 4m 46s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/89c0a59ac9f04b518e433c49bd04d814 : SUCCESS in 30m 40s","accounts_in_message":[],"_revision_number":5},{"id":"fa2601595063a982484c014a5b960183c6acc89a","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2022-10-11 17:38:41.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4ae02672ba42aa77f1e46825842d9384d3f35e1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 19:22:23.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/960a39f2d4174ff1952788a4757a34cd\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/dbb0acaa67774c0285339fa167eb380b : SUCCESS in 6m 17s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/540928754d0742239c750cbd18a4e795 : SUCCESS in 8m 43s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/b38afb97d7a0461b97939a934003b99e : SUCCESS in 4m 55s (non-voting)\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/b4eac57e5fd143d5a42a67f82d9b90d3 : SUCCESS in 5m 39s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/b3b17be1f48143c280da0425af9b92e5 : SUCCESS in 8m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/54cbe7f046b7433fb442ff61d55efe57 : SUCCESS in 9m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6c16e52594b4bb1af5e47f4bed4f1c2 : SUCCESS in 5m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/41e73b7a83274a698c5004d113b6f150 : SUCCESS in 9m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff6a2f1b34a74f44a63d998cb57d18de : SUCCESS in 6m 39s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/ca0e3e8a8bfe4342811d3b3aa13bbb9a : SUCCESS in 7m 50s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/569042a32d004e7d85488c454b3449a7 : SUCCESS in 1h 34m 51s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/ec1356408dea4c80a2c5d8c28a973e93 : SUCCESS in 25m 59s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/221b0511e66e44cca1053bb099ffc7de : SUCCESS in 4m 54s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/516084ee1c9e4f7ea9c6c97056746753 : SUCCESS in 26m 04s","accounts_in_message":[],"_revision_number":5},{"id":"6585d71be9970868a9ff204211446f25249bc921","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 19:22:56.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"2112f0bffb1b48a395ca5684dd344f80d871621c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 20:43:40.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d35cdb2f87894a99bb1678ef768371e8\n\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/017af03b2d824553aa6f118a18a3026c : SUCCESS in 4m 10s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/ad92c33864f243278236b264d49263c6 : SUCCESS in 7m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2720c3ba9c249f88503d6c7281bf94a : SUCCESS in 7m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/439ad6db5b10436da9925c65fdf35a3a : SUCCESS in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b68916284ddc4f07beb53d52dd5910b1 : SUCCESS in 8m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8a1c0687b704c1badb189cd8e2a9834 : SUCCESS in 5m 51s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/856c9a9bc03b4e3c90622e62f89609c3 : SUCCESS in 20m 59s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/843b0862df614f44bec50eb6082e3548 : SUCCESS in 5m 35s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/94c381d9604a4fc8a0ae806aaebfa9a5 : FAILURE in 1h 19m 58s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/d710bf83cb0345759813b1f66fd613e5 : SUCCESS in 18m 12s","accounts_in_message":[],"_revision_number":5},{"id":"648f525b3aef687cbdf727021a102770b6fa59bc","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2022-10-12 13:01:48.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0f46ffeeb0e60f1c11377ab3a9051c8f4570b380","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 14:45:15.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4118285f82594cad9b6fe6a32a36bf73\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/58222e9537794eda9457fa51ca2992cb : SUCCESS in 7m 04s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/77afca47aaf94b5fa6b6f1d7a09420d2 : SUCCESS in 6m 04s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/69549abca5504bed8a27da5bcd98d927 : SUCCESS in 4m 42s (non-voting)\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/8947a085a6114c40828a8366b2db58ea : SUCCESS in 4m 52s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/31d9d6cf0d6f49ffafe132a6f5fd8709 : SUCCESS in 8m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/30679dfa3a7645a0b500c803cc5899e9 : SUCCESS in 8m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/845dd814af7a40fc886bc7ef936c167f : SUCCESS in 5m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7205bd3f10154760bd882f91a6905344 : SUCCESS in 8m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b3db82cb3d346ea91863beab5b28dd6 : SUCCESS in 7m 15s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/7a04be0bef414efa95d9e8c48a91d65b : SUCCESS in 7m 16s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/cf5924ecc0e64d9dae6460f8c655045e : FAILURE in 1h 41m 56s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/3074a1a1d062496fb207ce263d85317d : SUCCESS in 23m 51s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/860c9669b22a42c28e6b8c3e980ce04c : SUCCESS in 4m 34s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/c6bcb4972f0d4e12a7cbff11ac7caa1f : SUCCESS in 27m 36s","accounts_in_message":[],"_revision_number":5},{"id":"fe32167c65b31a39fd663493656809da825d67da","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2022-10-12 20:50:55.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a6ea277f67c9f17f2f119a66654be0c5eb783866","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 22:22:56.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bad327b002c84489a52c3528d9388976\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/fe55a226e3f146d586909a61cee09330 : SUCCESS in 4m 43s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/d8a2f798085d4c84a5a13fee01a1e26a : SUCCESS in 6m 19s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/0a0603a3c6134eb190c926a30fd2469f : SUCCESS in 3m 52s (non-voting)\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/9bd19a0e44ee447cb990e950866a5508 : SUCCESS in 4m 04s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/e2e0e9b4f83046948c4ea125a3779a36 : SUCCESS in 9m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/58b3d03637ec403e8ab8e216eb98071c : SUCCESS in 9m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1ee03397a2548d08531f7307fd05b07 : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/754501f91e254a12b8f0b43df2675604 : SUCCESS in 8m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f70ffc8837894ffaafb5c55883cc6d35 : SUCCESS in 5m 40s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/181600289e27439391f7f6b01679a4ed : SUCCESS in 7m 09s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/accb9941d9504a74a16ab5d70d70ff65 : FAILURE in 1h 21m 32s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/469a31402e364a599594b1d382c66026 : SUCCESS in 29m 27s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1dca0501089645a2819bf4fbb940dae2 : SUCCESS in 5m 03s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/0a50f388c0794574822a47f730705acf : SUCCESS in 26m 00s","accounts_in_message":[],"_revision_number":5},{"id":"e4cf425361bc420a31c89d9b389eb451355ca386","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2022-10-13 02:15:55.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"bda3b72fee7e4798f28f8712cab5621bfc7abe9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 03:59:53.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22b82ef5ee6749bdbe3377c4df68d9f8\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/f217b16d5b3242b7af7159604fb79110 : SUCCESS in 6m 24s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/ac93bbf06f3e46e1a2b81da7521762e6 : SUCCESS in 8m 11s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/42cc84fd07db4476aad683d4499b4624 : SUCCESS in 5m 02s (non-voting)\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/fc2078ed6aed44c5b1c1b558b0d006ec : SUCCESS in 5m 02s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/061d8c9cb5594271a40d9eb89a868a82 : SUCCESS in 9m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fe52cf5e5c29493494eb47a95c370e05 : SUCCESS in 10m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1de385675f88432f874e65f7f2387db8 : SUCCESS in 5m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/60e49a7be08c43c189799abdc519d68e : SUCCESS in 11m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1051009a12834fe290b8ca57e8b3fceb : SUCCESS in 6m 54s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/ac29b78fc3024b63abfb40bc1a781ccc : SUCCESS in 8m 37s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/758ea0a87c4c4aefbe7005c6088cd3e0 : SUCCESS in 1h 34m 08s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/e0f3f2d3252a4e69b15bf30c48b81bf5 : SUCCESS in 28m 19s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cc3e182a7a1c480b8bfef42634056fd6 : SUCCESS in 5m 52s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/4a7e4bb345da4e34b06fb427d2c2a318 : SUCCESS in 31m 07s","accounts_in_message":[],"_revision_number":5},{"id":"9f9347e59a65a123715808479b85cfb93da3706f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 04:00:28.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"0c55be52ebcdef137af65e1d116e99bfd83c3369","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 05:35:07.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4933c98466194cd78d97924fe9619ab2\n\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/ad3cb5652feb4dac81064644417d5a49 : SUCCESS in 4m 26s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/4d12da6a69ae41658532788e4ac1b1a8 : SUCCESS in 7m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a7b7827c553b4d73bede5c8e46b0d688 : SUCCESS in 8m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffff6bea951f49ff8b42b1baa21a4992 : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/91d03bff23f34e99a26646e1a69e3991 : SUCCESS in 8m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f73d2b3979943a4a4b7d7b532de2fb7 : SUCCESS in 6m 28s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/210b631a83c34170985c1ac956a75a8c : SUCCESS in 29m 09s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/4109b183377344299aa771b276606c2f : SUCCESS in 7m 16s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/763e7c2544ba4801b4feec7991d29210 : SUCCESS in 1h 34m 03s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/d9f0c0156ea243319106bd80fef7d582 : SUCCESS in 22m 31s","accounts_in_message":[],"_revision_number":5},{"id":"e3e29edd641ccf89fd9d1883521876bee633d990","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 05:35:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"fe30b7909417d6037481ce8ed0443ff6253bbaef","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 05:36:12.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2b0476a94534a7a97ae4b354b4fd423\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f50fc9f22f2d4a55ac41f19bc68a677c : SUCCESS in 46s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"6eeaf9852478e25ff77c21117664ac126c5357a4","revisions":{"700784d841e464202eb5f1267773dd7effcc5744":{"kind":"REWORK","_number":1,"created":"2022-09-07 17:34:15.000000000","uploader":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"ref":"refs/changes/08/856308/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/08/856308/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/08/856308/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/08/856308/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/08/856308/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/08/856308/1"}}},"commit":{"parents":[{"commit":"58f83295377f566a764b23e954a3eb402650a416","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/58f83295377f566a764b23e954a3eb402650a416"}]}],"author":{"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","date":"2022-09-07 13:52:48.000000000","tz":-180},"committer":{"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","date":"2022-09-07 15:27:06.000000000","tz":-180},"subject":"Fix app cred create without project_id for domain admins","message":"Fix app cred create without project_id for domain admins\n\nUsers with domain admin role that are not cloud admins are\nnot able to get scoped context and create an application\ncredential with project_id, so this change removes the\ndomain context in that particular case.\n\nCloses-bug: #1827120\nChange-Id: I076a97a6f943ab74a2db8bc5179a7db194009db4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/700784d841e464202eb5f1267773dd7effcc5744"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/700784d841e464202eb5f1267773dd7effcc5744"}]},"branch":"refs/heads/master"},"421463f838bf831dd148825ae20e553d2c859c99":{"kind":"REWORK","_number":2,"created":"2022-09-07 20:41:46.000000000","uploader":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"ref":"refs/changes/08/856308/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/08/856308/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/08/856308/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/08/856308/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/08/856308/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/08/856308/2"}}},"commit":{"parents":[{"commit":"58f83295377f566a764b23e954a3eb402650a416","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/58f83295377f566a764b23e954a3eb402650a416"}]}],"author":{"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","date":"2022-09-07 13:52:48.000000000","tz":-180},"committer":{"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","date":"2022-09-07 20:41:23.000000000","tz":-180},"subject":"Fix app cred create without project_id for domain admins","message":"Fix app cred create without project_id for domain admins\n\nUsers with domain admin role that are not cloud admins are\nnot able to get scoped context and create an application\ncredential with project_id, so this change removes the\ndomain context in that particular case.\n\nCloses-bug: #1827120\nChange-Id: I076a97a6f943ab74a2db8bc5179a7db194009db4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/421463f838bf831dd148825ae20e553d2c859c99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/421463f838bf831dd148825ae20e553d2c859c99"}]},"branch":"refs/heads/master"},"a8f7b0c3d857cd67e9c3c3b45d8cd5d1f3b650e2":{"kind":"REWORK","_number":3,"created":"2022-09-28 21:18:38.000000000","uploader":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"ref":"refs/changes/08/856308/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/08/856308/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/08/856308/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/08/856308/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/08/856308/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/08/856308/3"}}},"commit":{"parents":[{"commit":"a0560b4bf5396988844520f76bc2fe457b6da8ce","subject":"Merge \"Fix \"Set admin password\" checkbox visibility\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/a0560b4bf5396988844520f76bc2fe457b6da8ce"}]}],"author":{"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","date":"2022-09-07 13:52:48.000000000","tz":-180},"committer":{"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","date":"2022-09-28 21:15:12.000000000","tz":-180},"subject":"Fix app cred create without project_id for domain admins","message":"Fix app cred create without project_id for domain admins\n\nUsers with domain admin role that are not cloud admins are\nnot able to get scoped context and create an application\ncredential with project_id, so this change forces the\nscoped context in that particular case.\n\nCloses-bug: #1827120\nChange-Id: I076a97a6f943ab74a2db8bc5179a7db194009db4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/a8f7b0c3d857cd67e9c3c3b45d8cd5d1f3b650e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/a8f7b0c3d857cd67e9c3c3b45d8cd5d1f3b650e2"}]},"branch":"refs/heads/master"},"9657ec7808ee770796f6ee79903c97a23f851bb0":{"kind":"REWORK","_number":4,"created":"2022-10-04 19:28:51.000000000","uploader":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"ref":"refs/changes/08/856308/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/08/856308/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/08/856308/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/08/856308/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/08/856308/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/08/856308/4"}}},"commit":{"parents":[{"commit":"a0560b4bf5396988844520f76bc2fe457b6da8ce","subject":"Merge \"Fix \"Set admin password\" checkbox visibility\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/a0560b4bf5396988844520f76bc2fe457b6da8ce"}]}],"author":{"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","date":"2022-09-07 13:52:48.000000000","tz":-180},"committer":{"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","date":"2022-10-04 19:28:28.000000000","tz":-180},"subject":"Fix app cred create without project_id for domain admins","message":"Fix app cred create without project_id for domain admins\n\nUsers with domain admin role that are not cloud admins are\nnot able to get scoped context and create an application\ncredential with project_id, so this change forces the\nscoped context in that particular case.\n\nCloses-bug: #1827120\nChange-Id: I076a97a6f943ab74a2db8bc5179a7db194009db4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/9657ec7808ee770796f6ee79903c97a23f851bb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/9657ec7808ee770796f6ee79903c97a23f851bb0"}]},"branch":"refs/heads/master"},"6eeaf9852478e25ff77c21117664ac126c5357a4":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-10-11 07:43:06.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/08/856308/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/08/856308/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/08/856308/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/08/856308/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/08/856308/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/08/856308/5"}}},"commit":{"parents":[{"commit":"a645545584bc04057600503ee0afd4623ceedbcd","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/a645545584bc04057600503ee0afd4623ceedbcd"}]}],"author":{"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","date":"2022-09-07 13:52:48.000000000","tz":-180},"committer":{"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","date":"2022-10-11 07:43:06.000000000","tz":0},"subject":"Fix app cred create without project_id for domain admins","message":"Fix app cred create without project_id for domain admins\n\nUsers with domain admin role that are not cloud admins are\nnot able to get scoped context and create an application\ncredential with project_id, so this change forces the\nscoped context in that particular case.\n\nCloses-bug: #1827120\nChange-Id: I076a97a6f943ab74a2db8bc5179a7db194009db4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/6eeaf9852478e25ff77c21117664ac126c5357a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/6eeaf9852478e25ff77c21117664ac126c5357a4"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
