)]}'
{"id":"openstack%2Fhorizon~858693","triplet_id":"openstack%2Fhorizon~master~I5ab71b2a53a7b18b719c4561cd0a48c3ef6ab5e4","project":"openstack/horizon","branch":"master","topic":"reno-zed","hashtags":[],"change_id":"I5ab71b2a53a7b18b719c4561cd0a48c3ef6ab5e4","subject":"Update master for stable/zed","status":"MERGED","created":"2022-09-21 12:08:30.000000000","updated":"2022-09-21 15:02:44.000000000","submitted":"2022-09-21 15:01:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"858693-reno-zed","meta_rev_id":"2471f2aae39a20e4a26ba74ac87eb0a2b2e521fe","_number":858693,"virtual_id_number":858693,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-21 15:01:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":2,"date":"2022-09-21 14:38:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":1,"date":"2022-09-21 14:38:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-21 12:26:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-21 14:38:54.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"}],"messages":[{"id":"f2912eeb09f7c1d338d1f6a2572bf4bb603ca794","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2022-09-21 12:08:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"35486f5812a36ac4faab76eb1f2a4ddfb6b4d80c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 12:26:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9d9cd19c00946b888f855cd14ba338e\n\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/2a34de47396c402ea76c2662ce0d85f3 : SUCCESS in 8m 43s (non-voting)\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/9075461363a24acab194dd71713f578c : SUCCESS in 6m 01s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/aeec44c9bf0d4a10b7f35a8d3867945d : SUCCESS in 9m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d34a176c19c49448bb9d929bd584225 : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fe193e410634e8fa7c37405a1c2d043 : SUCCESS in 7m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3821ae2dc0d84da3988e9a268f80cfd1 : SUCCESS in 10m 31s","accounts_in_message":[],"_revision_number":1},{"id":"06631778158438bba842bbb7ecd44e2fc40070c2","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2022-09-21 14:38:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ff446bb78877aa24e20bce08f4ccfe7bf2a7062e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 14:39:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3ba0c689512bcd19b7b68909b47804b3d9b5c523","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 15:01:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd58ae6e3c4f4f6985a1f1fa92ca8266\n\n- horizon-nodejs16-run-lint https://zuul.opendev.org/t/openstack/build/5816cbd5a4b34052a5ecff75902d06a5 : SUCCESS in 5m 48s\n- horizon-nodejs16-run-test https://zuul.opendev.org/t/openstack/build/829e7f6005464fd2980b6e77d939a725 : SUCCESS in 10m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a22b243e7bd4a4baaf095a2f46386e4 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df5a057fed934a3cbb65d529d3450cc9 : SUCCESS in 8m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c38e142808634addaba5f183121a6474 : SUCCESS in 12m 23s","accounts_in_message":[],"_revision_number":1},{"id":"51e515edd7119c7193c5b5e27116bfe0c7fa7d1d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 15:01:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2471f2aae39a20e4a26ba74ac87eb0a2b2e521fe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 15:02:44.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/784150d8b1b3462ebdc6efbb07bcbc87\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/691012bf3b0a4e38828cd67502e6792d : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d066ea6c2a04cbf8fb640583077824d : FAILURE in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2da8878d6a9169645308ab82e73ad024731a48f9","revisions":{"2da8878d6a9169645308ab82e73ad024731a48f9":{"kind":"REWORK","_number":1,"created":"2022-09-21 12:08:30.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/93/858693/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/93/858693/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/93/858693/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/93/858693/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/93/858693/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/93/858693/1"}}},"commit":{"parents":[{"commit":"443134b54ada8a483521c62066abca25acdfffd8","subject":"Merge \"Don\u0027t try and update port security if its not changing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/443134b54ada8a483521c62066abca25acdfffd8"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-09-21 12:08:28.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-09-21 12:08:28.000000000","tz":0},"subject":"Update master for stable/zed","message":"Update master for stable/zed\n\nAdd file to the reno documentation build to show release notes for\nstable/zed.\n\nUse pbr instruction to increment the minor version number\nautomatically so that master versions are higher than the versions on\nstable/zed.\n\nSem-Ver: feature\nChange-Id: I5ab71b2a53a7b18b719c4561cd0a48c3ef6ab5e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/2da8878d6a9169645308ab82e73ad024731a48f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/2da8878d6a9169645308ab82e73ad024731a48f9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
