)]}'
{"id":"openstack%2Fhorizon~860465","triplet_id":"openstack%2Fhorizon~stable%2Fxena~I1fe45063c9d1cdd8682998329d81f843d30f80b3","project":"openstack/horizon","branch":"stable/xena","topic":"flavour-info-fix-stable/xena","attention_set":{},"removed_from_attention_set":{"6914":{"account":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"last_update":"2022-10-28 05:23:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1fe45063c9d1cdd8682998329d81f843d30f80b3","subject":"Fix flavor specification at instance overview page","status":"MERGED","created":"2022-10-06 16:23:23.000000000","updated":"2022-10-28 05:24:17.000000000","submitted":"2022-10-28 05:23:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"860465-flavour-info-fix-stable/xena","cherry_pick_of_change":848905,"cherry_pick_of_patch_set":6,"meta_rev_id":"e4c59dc631c9beb1b909983ea27fdc561c5ba058","_number":860465,"virtual_id_number":860465,"owner":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-28 05:23:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":2,"date":"2022-10-28 03:44:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":2,"date":"2022-10-21 14:49:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":1,"date":"2022-10-28 03:44:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-19 06:54:03.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"CC"},{"updated":"2022-10-19 08:14:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-21 14:49:08.000000000","updated_by":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"reviewer":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"state":"REVIEWER"},{"updated":"2022-10-28 03:44:26.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"}],"messages":[{"id":"27b3faa3b71e5ef98685e11478c27d8d25c813de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2022-10-06 16:23:23.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b0c6d7c07f68d8d9da529ee58a292d879bd590c4","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2022-10-19 06:54:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3fc04d9d9473bdfacb18918b3208fe129ce8a256","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 08:14:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f4892ee0d4034b2c9eb0895bbce02689\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/2ce1a95f0bbb4448b001e46bdabeab0d : SUCCESS in 5m 14s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/5235cbfa778f48c3ad62b93d819d709a : SUCCESS in 3m 30s (non-voting)\n- horizon-nodejs14-run-lint https://zuul.opendev.org/t/openstack/build/5a52d2d19ef34e79bca5f4a225eaff88 : SUCCESS in 4m 13s\n- horizon-nodejs14-run-test https://zuul.opendev.org/t/openstack/build/a374954bde2d4087b26cd84cda3df454 : SUCCESS in 8m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4a9dbcbdd6b4e328e354cb006c2a2f2 : SUCCESS in 9m 44s\n- horizon-tox-python3-django32 https://zuul.opendev.org/t/openstack/build/4a0d15511811450c9fb0bb519566845c : SUCCESS in 11m 13s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ee0f0ca6498c4412844496c1255e073c : SUCCESS in 9m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1081c9d3edba4616bbccdbbd0feef8c0 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1261997e7128458cb5b7f4cb13c53cb1 : SUCCESS in 10m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/68689425ff6a401e819b3b6a6b0a720c : SUCCESS in 9m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ed38495f75f46aba5306176b7043657 : SUCCESS in 6m 30s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/fe48bcacbe0440539d801eca061b9028 : SUCCESS in 5m 49s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/afe339a278094ff09e591c74660e2ee5 : FAILURE in 1h 18m 48s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/ad956522287b4539a8172dc16f65e1ca : SUCCESS in 25m 19s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ecbd3af713df40c4a6a833ae24537304 : SUCCESS in 3m 36s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/2b0a680225724e40be685d36eda26dc1 : SUCCESS in 28m 17s","accounts_in_message":[],"_revision_number":1},{"id":"26ca32b2b560ab176a58c53fb2863e412b1e1564","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2022-10-20 20:47:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f4aca7ec51118ca41d20a88ae53afef5ca714723","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 22:24:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48f27c5464d147af91d6d7c958fb5360\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/5ae22ed8009d451481c5af2af647ce2a : SUCCESS in 5m 48s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/356c4042c0584c7388795a259821e5fa : SUCCESS in 4m 47s (non-voting)\n- horizon-nodejs14-run-lint https://zuul.opendev.org/t/openstack/build/d004eccaf32e40c29630c5b97d3b97be : SUCCESS in 4m 36s\n- horizon-nodejs14-run-test https://zuul.opendev.org/t/openstack/build/9379e6349dcc4ef3a35cef5090406749 : SUCCESS in 12m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f05e360bdf6c4c9a99aa909b06ad7fff : SUCCESS in 10m 59s\n- horizon-tox-python3-django32 https://zuul.opendev.org/t/openstack/build/3ea4d3346cbc499599bc51b412226f2c : SUCCESS in 12m 15s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/78b63d57ad28411499561adea6a008ed : SUCCESS in 10m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1cf74870a684d07889cf3de508c7deb : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a044dce9dc5443f6a83c898b479eff99 : SUCCESS in 11m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2fdf3a55aa7e42e4a3850e17a6a3c334 : SUCCESS in 10m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b02b96abe124a0599b054fd6e7d389f : SUCCESS in 5m 58s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/5c33376e61094ab79701bd47404d07de : SUCCESS in 6m 10s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/97c78b16054644b6b7bfd4c10702e6ba : SUCCESS in 1h 29m 23s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/165382026ced40f0a10c0d58eae17e58 : SUCCESS in 36m 07s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6582e88e98c042cc9f185613ed9741d5 : SUCCESS in 4m 05s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/efd57f608cc44e8c8d84551115e80b39 : SUCCESS in 35m 28s","accounts_in_message":[],"_revision_number":1},{"id":"1e9708b5a87146de88d3e27ddf832aba217cf83c","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2022-10-21 14:49:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"faad74822599342137a4748eb9039c8103392617","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2022-10-28 03:44:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"26598bab1eed94a25a942036afc38d03e99b057a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 03:45:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3cb7827d4fc3a075a02f8f5a39665fcfe4d6bb5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 05:23:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c18e4185b5b74648a1842b47b2fbbbec\n\n- horizon-nodejs14-run-lint https://zuul.opendev.org/t/openstack/build/b20d5f14380046c4a7f1cf0b801d2fa5 : SUCCESS in 4m 38s\n- horizon-nodejs14-run-test https://zuul.opendev.org/t/openstack/build/e46236a0f02e4ea083b01af5075e045c : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c479274773c74c098dbb0be330a8cb5d : SUCCESS in 9m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/709e04b4c8264bc59fd65322064bc9b6 : SUCCESS in 7m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea08600483b74f32b72d9879670597b4 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cffcbdb5a7804e959646c8e304f73173 : SUCCESS in 10m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc536191158242db8d712f5c911280d4 : SUCCESS in 6m 18s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/2f660a003db9486dbf19170e67e4dc50 : SUCCESS in 34m 03s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/ebe72174c5d9417c879daa2f4d517bf8 : SUCCESS in 6m 07s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/a02f8b6e40304c819e56503ec9210ba1 : SUCCESS in 1h 31m 38s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/76c224d9a50745838e9598ef9d2369d6 : SUCCESS in 33m 33s","accounts_in_message":[],"_revision_number":1},{"id":"3be0229040bcabf35c928052cf443bcc1b5db9a4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 05:23:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e4c59dc631c9beb1b909983ea27fdc561c5ba058","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 05:24:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7549875bc825430a8226b6ca812621ff\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d6d885264d54920aae3ca27f8d7772b : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6a6f49e51528e516754405ff194dcfa521e54c7d","revisions":{"6a6f49e51528e516754405ff194dcfa521e54c7d":{"kind":"REWORK","_number":1,"created":"2022-10-06 16:23:23.000000000","uploader":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"ref":"refs/changes/65/860465/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/65/860465/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/65/860465/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/65/860465/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/65/860465/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/65/860465/1"}}},"commit":{"parents":[{"commit":"48b702adf75b7f19420df578d2bf9bf813099af5","subject":"Make existing empty metadata properties optional","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/48b702adf75b7f19420df578d2bf9bf813099af5"}]}],"author":{"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","date":"2022-06-27 14:52:48.000000000","tz":-300},"committer":{"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","date":"2022-10-06 16:23:23.000000000","tz":0},"subject":"Fix flavor specification at instance overview page","message":"Fix flavor specification at instance overview page\n\nThe details page uses flavor id to resolve instance flavor, relying\non legacy Nova API v2.46. This patch adds the current Nova API\noption, resolving instance flavor by flavor name, so flavor\nspecification at instance overview page is displayed properly.\n\nChange-Id: I1fe45063c9d1cdd8682998329d81f843d30f80b3\n(cherry picked from commit d0e9e976f2769b289196d481b0928893b702b011)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/6a6f49e51528e516754405ff194dcfa521e54c7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/6a6f49e51528e516754405ff194dcfa521e54c7d"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
