)]}'
{"id":"openstack%2Fhorizon~861613","triplet_id":"openstack%2Fhorizon~stable%2Fvictoria~Ie63dd3b161cb8fd9be89002027d699ce2c4a67a5","project":"openstack/horizon","branch":"stable/victoria","topic":"fix-queue-config-stable/victoria","attention_set":{},"removed_from_attention_set":{"29313":{"account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"last_update":"2022-10-18 12:20:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie63dd3b161cb8fd9be89002027d699ce2c4a67a5","subject":"zuul: Declare queue at top level","status":"MERGED","created":"2022-10-17 09:54:35.000000000","updated":"2022-10-18 12:23:50.000000000","submitted":"2022-10-18 12:20:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861613-fix-queue-config-stable/victoria","cherry_pick_of_change":845516,"cherry_pick_of_patch_set":1,"meta_rev_id":"585233d52211531b0bb8fd503f16da4f464b10f1","_number":861613,"virtual_id_number":861613,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-18 12:20:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"all":[{"value":2,"date":"2022-10-18 10:07:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":2,"date":"2022-10-18 10:31:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":1,"date":"2022-10-18 10:55:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":1,"date":"2022-10-18 10:31:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-17 09:54:35.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2022-10-17 11:34:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-18 10:07:09.000000000","updated_by":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"reviewer":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"state":"REVIEWER"},{"updated":"2022-10-18 10:31:11.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2022-10-18 10:55:25.000000000","updated_by":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"reviewer":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"state":"REVIEWER"}],"messages":[{"id":"53882474ffaff1aeaa197fe55d22397555a19729","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2022-10-17 09:54:35.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e6c22b72d86135f1a3ce4afe95d808d17e1cd7fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 11:34:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30589a7be553448f80a43f335048eb81\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/be1e1646ae7a4513965bea4ad3f76054 : SUCCESS in 2m 22s\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/d99b2dd7ca4c4b8c8435e9117464b53a : SUCCESS in 5m 11s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/fb52bd72f6e44043ba0e982847774b38 : SUCCESS in 7m 19s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/dcdd5f7565fa4582ad9d037fcee4f9b5 : SUCCESS in 4m 17s (non-voting)\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/a77df2a15b534881a82d4f1ee4ba7531 : SUCCESS in 5m 16s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/151bd90183bb494b92feb7e3c70f42eb : SUCCESS in 8m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ce93027a0534c038f5c39d5f66983e9 : SUCCESS in 10m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b1eac6eec8a498ba58d3e1a6855b903 : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7bdafd684a4f4014bbfa26fb7e1e33fa : SUCCESS in 9m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e49ea729ffa487b89bc3995a0497f8a : SUCCESS in 6m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb98b08227144728b12d1d9814c0bd2b : SUCCESS in 13m 55s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/9006d6bcdda74ab8aad2f8f654696465 : SUCCESS in 5m 56s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/59616268fff24e72af1fd0b9e959dc80 : SUCCESS in 1h 33m 24s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/0174b746ef434f6c99387846d8711130 : SUCCESS in 38m 32s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/799183ee8c194afaac76832b570a4296 : SUCCESS in 3m 45s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/d1b8aa70eae24a759521417eae423af6 : SUCCESS in 39m 29s","accounts_in_message":[],"_revision_number":1},{"id":"ea163d1b7b7bbe656ade7bfcea9cfc9d1dc20bc0","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2022-10-18 10:07:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"50de3333d97fd833d2f5b089fae58a6b8ad11cc4","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2022-10-18 10:31:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5932a52e2eeecd4ac7ee47bd4fd7032fae53a718","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 10:32:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c9bd7ede075d82f4357be49d5ec7d86752b318fe","author":{"_account_id":30396,"name":"qian wang","email":"wang.qian@99cloud.net","username":"wang.qian"},"date":"2022-10-18 10:55:25.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f4138079e3f62fd2bb09992c46aa5cf3fb2d740f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 12:20:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e39f943d4914885acf597bb047330aa\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/4125717b6b0046f480d125f51c62c439 : SUCCESS in 2m 40s\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/5e44d3a3c69b467883c60b7f9ed8e850 : SUCCESS in 5m 11s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/68ba9e9c677145e3b1f2b9c7874f9985 : SUCCESS in 10m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7bff95a0b6c24e2da99b3f1e22ad27ae : SUCCESS in 10m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b60f5015f084ca090efbec3a23570d7 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6db07987ac4943138c88d34f48924f73 : SUCCESS in 12m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bac939ddcfcb4a0c8f71c83be3461a9f : SUCCESS in 5m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/190d700f598242859748bbca00dc9737 : SUCCESS in 12m 49s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/7506202d3f4a4edd8a3721b8c5c5fdae : SUCCESS in 38m 35s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/68be8608205144c8961b7ae82e280fad : SUCCESS in 6m 24s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/d56a8cb7dfac45a8b15b6c9db6da04ac : SUCCESS in 1h 41m 46s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/55e1016da37a49b5bd29161fd1c476b1 : SUCCESS in 39m 03s","accounts_in_message":[],"_revision_number":1},{"id":"7953a390435cbccc1b1743443a94e6c080b92678","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 12:20:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"585233d52211531b0bb8fd503f16da4f464b10f1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 12:23:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ade1220553e94676ab1fcfd85f09f420\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75e0ce9c3a2f424ba07f8b2f6f1621a3 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1da7f7f21607faae1fa727788046736d209144f7","revisions":{"1da7f7f21607faae1fa727788046736d209144f7":{"kind":"REWORK","_number":1,"created":"2022-10-17 09:54:35.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/13/861613/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/13/861613/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/861613/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/861613/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/861613/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/13/861613/1"}}},"commit":{"parents":[{"commit":"d329996a929b6a03d7429a0ea7f2b05041f1714a","subject":"Make existing empty metadata properties optional","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/d329996a929b6a03d7429a0ea7f2b05041f1714a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-13 00:53:35.000000000","tz":540},"committer":{"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","date":"2022-10-17 09:54:35.000000000","tz":0},"subject":"zuul: Declare queue at top level","message":"zuul: Declare queue at top level\n\nZuul deprecated declaring shared queues at a pipeline level with\nrelease 4.1.0[1]. This updates the job definition to use the top level\ndeclaration instead.\n\nSee [2] for details.\n\n[1] https://zuul-ci.org/docs/zuul/latest/releasenotes.html#relnotes-4-1-0-deprecation-notes\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2022-May/028603.html\n\nChange-Id: Ie63dd3b161cb8fd9be89002027d699ce2c4a67a5\n(cherry picked from commit 4212950a4a6b036f563a236cb005604e1c096f04)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/1da7f7f21607faae1fa727788046736d209144f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/1da7f7f21607faae1fa727788046736d209144f7"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
