)]}'
{"id":"openstack%2Fhorizon~868083","triplet_id":"openstack%2Fhorizon~master~Id921b69df13af3cc209236f0446d82dd30e4d8a2","project":"openstack/horizon","branch":"master","topic":"fix_default_user_data","attention_set":{},"removed_from_attention_set":{"29313":{"account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"last_update":"2023-01-27 15:38:33.000000000","reason":"Change was submitted"},"8648":{"account":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"last_update":"2023-01-27 15:38:33.000000000","reason":"Change was submitted"},"9545":{"account":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"last_update":"2023-01-27 15:38:33.000000000","reason":"Change was submitted"},"841":{"account":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"last_update":"2023-01-27 04:43:24.000000000","reason":"\u003cGERRIT_ACCOUNT_841\u003e replied on the change","reason_account":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"}}},"hashtags":[],"change_id":"Id921b69df13af3cc209236f0446d82dd30e4d8a2","subject":"Fix applying OPENSTACK_SERVER_DEFAULT_USER_DATA value","status":"MERGED","created":"2022-12-19 12:54:35.000000000","updated":"2023-01-27 15:40:08.000000000","submitted":"2023-01-27 15:38:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":28,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"868083-fix_default_user_data","meta_rev_id":"2e597bf817c752f73923ab222f41f200c661d24a","_number":868083,"virtual_id_number":868083,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-27 15:38:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":2,"date":"2023-01-27 13:55:09.000000000","_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":2,"date":"2023-01-27 04:43:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":1,"date":"2023-01-27 13:55:09.000000000","_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"}],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-19 14:19:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-21 18:10:02.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2022-12-21 18:27:06.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2022-12-21 18:27:06.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"state":"REVIEWER"},{"updated":"2022-12-28 01:48:56.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2022-12-28 09:32:40.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2023-01-06 04:57:54.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"}],"messages":[{"id":"bc083d886e8ca817fa7fafae7be2433b4992db4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2022-12-19 12:54:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9174ebcb17861f7aaf4cc61ec580b9a4031cdec6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 14:19:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e91de207c8c48cd961aa46b2ee2c043\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/3f6eca9e6c7f4f018b8cd35da33533e0 : SUCCESS in 6m 59s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/55062e4ade4d4df9a6e4e74a6493ca9f : SUCCESS in 5m 25s (non-voting)\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/5429f7b19e6b4fd0ac9d068f94e04186 : SUCCESS in 4m 28s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/deba57324f284be7bc53f9f1c5d1400a : SUCCESS in 9m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4dcb2744591a4a4180cc8dc40bf863d1 : SUCCESS in 9m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ef363b55108438f8ec4b78d4363c833 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5dfc5bf3587d42e5a43aa5e183bc06f8 : SUCCESS in 8m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c5339224f1784079a210de823e5ed0a1 : SUCCESS in 9m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a624313d2347495784e283ff9c482ed3 : SUCCESS in 8m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f90d62027104725bf110a93d56cc13f : SUCCESS in 10m 04s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/0ea82b3f22ba438d84f79cc1366355bb : SUCCESS in 6m 55s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/49b52c8e08c54cae8b69ded6871e9fd4 : SUCCESS in 1h 18m 09s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/bb47584c07b34a309d4beabdc64a2323 : SUCCESS in 32m 01s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5845fabbb6784b008fedaa34fa8a8db7 : SUCCESS in 5m 18s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/c46c1954a41c4fb19f6cba00f538af15 : SUCCESS in 21m 00s","accounts_in_message":[],"_revision_number":1},{"id":"77557a74efdc2246c2bf48d1d9fc930e6093c17a","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2022-12-21 18:10:02.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4a41ac346ada4dd765d2b7a9535615a7faf5a737","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2022-12-21 18:10:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cfad81cd2c48d3da2881e853d3000f7398da8fc9","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-12-28 01:48:56.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3c34f699928c84b3e1fb5b5887a1a070cfa16dd4","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2022-12-28 09:32:40.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"94a12b26de0484e9ac5c85ee725a936fa5013532","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2023-01-06 04:57:54.000000000","message":"Patch Set 1: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"08bd66a8bf09967774c833dd38a466423188a978","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-01-20 14:54:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bf372ef1f1b232c694019b450c431887aab63184","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-01-20 14:55:07.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c5251518a6c54c7d6908e55cd5d23dda084a013d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 17:04:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c680a2d36a9b4b9ab0e7d9952a03fd91\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/a6c72e9594b3404e94710e798a9dbb71 : SUCCESS in 5m 31s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/f88ce370f2834eceaf3e1ed5f498a7be : SUCCESS in 3m 51s (non-voting)\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/dacc5ac168a441628137a555b4f65aaa : SUCCESS in 5m 16s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/122b4ae8924541eda0fe65a151274843 : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f9a66f81aa24f49b8f232f0a3f31c50 : SUCCESS in 8m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f101f1877414e779793e3ee8f2fd1f5 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49f67133645b49edad0c211c3e2d02c4 : SUCCESS in 10m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/20b227da6cfa4e0ba7089e77b717bfb1 : SUCCESS in 8m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/673fcd20571a47d1a7d1c056def0b163 : SUCCESS in 6m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a549bcc0d644790838959f97dcd01da : SUCCESS in 11m 13s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/ed57250b12eb4d749e43a843d68a7028 : SUCCESS in 6m 00s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/aba9d05958144c339a1dfe7a1b9e5966 : SUCCESS in 1h 48m 13s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/d0b7b0e248b04a5ba211e33957f9efec : SUCCESS in 33m 08s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2c1ddcf788204155b04129db209abf57 : SUCCESS in 4m 30s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/c8ba8ccd5f1b45c5a71203f9740bfb3f : SUCCESS in 30m 43s","accounts_in_message":[],"_revision_number":2},{"id":"33a71206e99939f30dd8df62096d8850dad74906","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-01-24 10:09:22.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"086631d877a54ca631cb174cd0d982b0ee96b7db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-01-24 14:50:31.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"146cc95e53df4247b2ef2edc35b174efcc956bb3","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-01-24 14:50:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bc79e686a98b85849f072950a365c8da6a46f560","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 16:29:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a8103b76df34077a1ce7b28e290e1fb\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/32779444e8db46e886e8cd4bf8b65686 : SUCCESS in 5m 54s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/d6438789eca342a884ebe9308649db84 : SUCCESS in 3m 16s (non-voting)\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/786592ce5a6d4543bbf2ced413c89473 : SUCCESS in 5m 16s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/4fa0ad73b3724377824036556eba746c : SUCCESS in 6m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f58c123edd32444ba5538e1a08a823b2 : SUCCESS in 11m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/516e2667c4a049799046e4f64bda98f0 : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/19f6fec23a504496b1cca7129dbe0868 : SUCCESS in 9m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d5507e61d3c7407a8ec207dd145065f7 : SUCCESS in 10m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dfc17453b3144839310b52ea4fef877 : SUCCESS in 7m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e940e4384cb4e3e8ede1affb78c40cd : SUCCESS in 10m 10s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/68f428a8270b4ac2a4a3703347777208 : SUCCESS in 5m 52s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/1f76dcd24cda4db5a6530d340a299e95 : SUCCESS in 1h 34m 01s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/e0000dee46e7434c8d5fa274470f3545 : SUCCESS in 31m 31s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b408ed682b594d1293e3ca0abbcf705c : SUCCESS in 4m 02s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/67a38e95f65f419f8f8c2783641784c5 : SUCCESS in 37m 32s","accounts_in_message":[],"_revision_number":3},{"id":"bb7c886fe5324201c1e1c75e626171683433be71","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-01-24 17:51:03.000000000","message":"Patch Set 3: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"9c60dd51debb51beaeeced714c793d81dd6b43d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-01-25 11:26:36.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ba57e9d8871514da1ab8bd5b6d3e6b782195e558","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-01-25 11:27:09.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f75cc5e8c0c9df3c0c9ca58f29c3bf7af58ce838","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-01-25 11:28:47.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"823907dde66c76b355ec6fff5c4afeb17d4c2aa8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 13:38:31.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d435d4e2b4d8411a9620cf92aec6cb9f\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/f76c324b44ce4bb4a007c48a01dfdad9 : SUCCESS in 6m 00s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/a8fc29dae6b04a71b8a804c1830681d6 : SUCCESS in 5m 34s (non-voting)\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/ea97d6b5bed74505967633d1f436e73b : SUCCESS in 5m 29s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/49ba8e688ada4cd8bbe8beb607c4f120 : SUCCESS in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/205fb1adde7f43619cc39f7174365104 : SUCCESS in 12m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e09cb39061e1467ea595dd26004f9831 : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc31caaac3e342feaf7c19e8cf3270df : SUCCESS in 10m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/16c0b3b3b9554a02855bcae29c8978fc : SUCCESS in 10m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12993c77f1154e508aed23e8446b9035 : SUCCESS in 8m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f4044bc552a49a2996a23639643a592 : SUCCESS in 10m 10s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/fb1900423dca480c89f0059a64de95b7 : SUCCESS in 5m 40s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/1f74ae582fdf49f8a61eab3e9054403a : TIMED_OUT in 2h 02m 55s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/8e11a3814ffe4a28be8fc1f38c02aa63 : SUCCESS in 36m 23s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9b017850e1054e5197015e95aa2a7e8e : SUCCESS in 4m 26s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/252a43f7df3a4b669add1c1083973384 : SUCCESS in 22m 56s","accounts_in_message":[],"_revision_number":4},{"id":"e70056b1ab57aab06b3eb85130523ebdddb24bd9","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2023-01-25 14:20:09.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"661f6f389ea0282ba7756be1e810687fbc9d7ca1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-01-25 14:58:02.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"cde8b2bf81145aed0e0bc5312c6a4fedf786ee31","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-01-25 14:59:03.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"1cf3a67da487459daa5da1e3dfa7630bd2e263d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 16:44:59.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a64dbe3680f461a83cecfff00048093\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/1d0c1607ca1b435ea0e6b0c8c25590e3 : SUCCESS in 7m 05s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/2960356d5c7f471cbcb459ee454ca4fa : SUCCESS in 4m 09s (non-voting)\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/d14248333100459b973a8c1c65f92464 : SUCCESS in 5m 05s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/e20632501f7f44cb84d22c13cd521087 : SUCCESS in 7m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/582a56e7159e44c08c591b83991b1ca8 : SUCCESS in 10m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6710a4c2486d43a28507c7cda234a009 : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ddff62c632840a2bce8e37241290407 : SUCCESS in 10m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ea8b31c5015847459c6e13a5164dc179 : SUCCESS in 10m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32b9476c1133457c9710a16d621f7cb2 : SUCCESS in 7m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f03fd47271f4e969a5852c0fff44dbc : SUCCESS in 10m 37s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/92922b55b9d94dbda55a4a6ed1369690 : SUCCESS in 9m 22s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/ea52636962e04968b95c835ae589d303 : SUCCESS in 1h 42m 18s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/2c3d592ea8014ccc906780eacb0e02af : SUCCESS in 39m 36s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d92bc58fcb8d4490825cd13837a066f7 : SUCCESS in 4m 54s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/ef3e8aac103549f58e2f5b80090dcb16 : SUCCESS in 43m 48s","accounts_in_message":[],"_revision_number":5},{"id":"ac47a43cf882fcf2e7d3cf4d2a2e52b17a7abcb3","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2023-01-25 20:06:45.000000000","message":"Patch Set 5: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"6cea4a8d2a094298ad69d78602bcc9d873a78ad6","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-01-26 13:52:39.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ab8c71cbac5dd2e4704b3710d5dc97ad81b4c349","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2023-01-26 13:52:49.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"80d8d75059f999dd2e7333c89dce243bb30d1edb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 14:22:52.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb3d755e0bbc40a6a868bf0c69712c59\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/881d07d24fbf4ee5a76ac52a93234932 : SUCCESS in 6m 07s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/02f81fd484d24ee2806d3df586343dbc : SUCCESS in 3m 43s (non-voting)\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/c76ccc15d4c64ee5857c1760b9903c32 : SUCCESS in 4m 32s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/fc3e67490e344b37a91edc4ab2e2e8ac : SUCCESS in 6m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b58f74d1ab8e432a9ef9bc191a7a89a6 : SUCCESS in 8m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9833920960944c4a07577dc9db8522b : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af59bd2ab84943b0a9572a6e16899acc : SUCCESS in 8m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e32f233b8fa8433d942fd38578e20817 : SUCCESS in 8m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15aa32c82d064dd8a66122bde56a2882 : SUCCESS in 7m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d82452c81ee9437290d504bbf09bbc6c : SUCCESS in 7m 46s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/ef91bf7d09ed4ef5aebf681399d3efa2 : SUCCESS in 5m 38s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/87b965e7b70449e5af81e4a14bc72e78 : FAILURE in 24m 39s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/bcbcea8b23aa4fd8b298268b477812bf : SUCCESS in 27m 22s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d8c900285fb24a659d9ae5b8ac341d96 : SUCCESS in 4m 27s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/2de6b848ef614d8e9daec93435b9031d : SUCCESS in 25m 17s","accounts_in_message":[],"_revision_number":6},{"id":"5820d1e69c21f3afdc12c3f9d525ef65df040287","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2023-01-27 04:43:24.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"35499a55f57376b40e269d72f351cce6323d79d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 06:06:01.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5db3b88870894d2f9da8566c90dc7524\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/01724271421c4c5cb2d1330b5e7117e4 : SUCCESS in 6m 28s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/89f708ecb95847ed8f5e744aef09ab28 : SUCCESS in 4m 16s (non-voting)\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/66c6edd8c1f94bf08dde0e48738a7bdc : SUCCESS in 5m 08s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/7a116421840c4e2184045bd387a7a6bc : SUCCESS in 6m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98797b9dee5c4579a4b620e0589e2d6c : SUCCESS in 7m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8685060d4f7a4debab6baca2e9247cf0 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d9d29202f964ef9bc09896b6621871d : SUCCESS in 7m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b53ba6f67c99419fbda44469dfeb642d : SUCCESS in 10m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8646f0e6941b429886b1b1a147f4c661 : SUCCESS in 6m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/088852fbd10240248c45df734dfb00e5 : SUCCESS in 13m 30s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/c322bb33c8134920a7392d938472f7f7 : SUCCESS in 6m 02s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/b44b59747bc0428ca8b34a0a493084e2 : SUCCESS in 1h 12m 38s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/e656917c28eb473fadd9c3537230fcae : SUCCESS in 34m 29s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/690e4076a9bd45a6bc2a24c2b7d8a396 : SUCCESS in 4m 14s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/0ae7d3c81bff44088cf5c4266ba04dc2 : SUCCESS in 39m 22s","accounts_in_message":[],"_revision_number":6},{"id":"397fcf347aebb21b1e0973b9621ddccbd0786b11","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2023-01-27 13:55:09.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"851b6394fd64f8499444bdc4d0696abc87afbedf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 13:55:42.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"46478e890fdd222a6ab6cfc0cfce6e5ab5fa17a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 15:38:33.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbe10b08b79a44df84c590301c684529\n\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/29e69a9fa535433c83e211c2bf660a4b : SUCCESS in 7m 30s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/10ba507383bf4c499909bc67d5687184 : SUCCESS in 8m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/678933ddfac84e3887b29d6f21c8d947 : SUCCESS in 8m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f469f504569c4528a38650ce71c4d96a : SUCCESS in 6m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85bea92176bf4a819180a130ef77d758 : SUCCESS in 7m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/01f0b07f2adb4d8899528552ef5f4f60 : SUCCESS in 8m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27c656601b4a430ba67dc0281f910ac3 : SUCCESS in 7m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f56baea6f5be4cf4b9a86eb3462dbc5e : SUCCESS in 11m 24s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/0e77aed199f24daebe2c11b93982bf1c : SUCCESS in 36m 41s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/f03462291eb848faa1952fee1030a3e6 : SUCCESS in 7m 15s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/f2e4efb08423447380dc876bbd489887 : SUCCESS in 1h 38m 43s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/58b5b95e70e44228bef221c396323499 : SUCCESS in 35m 28s","accounts_in_message":[],"_revision_number":6},{"id":"db274be6eb08e6d327297bc7acd7ab019b7a1ad2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 15:38:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"2e597bf817c752f73923ab222f41f200c661d24a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 15:40:08.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4662c6ebd144352a3b6ef8b3283f5e6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7257b91a586c48afaa894c9999da2303 : SUCCESS in 56s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd562f02b4d74478874f5bd2298df533 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"c8ee8c1965b280e24fb6266cbb2523230fea605f","revisions":{"229ae162faf43ed51457ef448641eccc6c8b837d":{"kind":"REWORK","_number":1,"created":"2022-12-19 12:54:35.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/83/868083/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/83/868083/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/83/868083/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/83/868083/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/83/868083/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/83/868083/1"}}},"commit":{"parents":[{"commit":"0add65eddcd6e70d4fbe76951d76c66ecc1a34b3","subject":"Merge \"Add OPENSTACK_KEYSTONE_ENDPOINT_TYPE config opt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/0add65eddcd6e70d4fbe76951d76c66ecc1a34b3"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2022-12-19 12:54:25.000000000","tz":60},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2022-12-19 12:54:25.000000000","tz":60},"subject":"Fix applying OPENSTACK_SERVER_DEFAULT_USER_DATA value","message":"Fix applying OPENSTACK_SERVER_DEFAULT_USER_DATA value\n\nThe final revision of I86ac21bf82c1667135abd4f20fb4514da0899450 is buggy\nand doesn\u0027t work.\nThis commit fixes it and adds a proper test.\n\nChange-Id: Id921b69df13af3cc209236f0446d82dd30e4d8a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/229ae162faf43ed51457ef448641eccc6c8b837d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/229ae162faf43ed51457ef448641eccc6c8b837d"}]},"branch":"refs/heads/master"},"6e4d351ad55e958b5b0a99e61957b273a3da9db1":{"kind":"REWORK","_number":2,"created":"2023-01-20 14:54:30.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/83/868083/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/83/868083/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/83/868083/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/83/868083/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/83/868083/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/83/868083/2"}}},"commit":{"parents":[{"commit":"0add65eddcd6e70d4fbe76951d76c66ecc1a34b3","subject":"Merge \"Add OPENSTACK_KEYSTONE_ENDPOINT_TYPE config opt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/0add65eddcd6e70d4fbe76951d76c66ecc1a34b3"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2022-12-19 12:54:25.000000000","tz":60},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-01-20 14:54:17.000000000","tz":60},"subject":"Fix applying OPENSTACK_SERVER_DEFAULT_USER_DATA value","message":"Fix applying OPENSTACK_SERVER_DEFAULT_USER_DATA value\n\nThe final revision of I86ac21bf82c1667135abd4f20fb4514da0899450 is buggy\nand doesn\u0027t work.\nThis commit fixes it and adds a proper test.\n\nChange-Id: Id921b69df13af3cc209236f0446d82dd30e4d8a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/6e4d351ad55e958b5b0a99e61957b273a3da9db1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/6e4d351ad55e958b5b0a99e61957b273a3da9db1"}]},"branch":"refs/heads/master"},"489b10b94139bf554610a35a388cc4f10137b0a2":{"kind":"REWORK","_number":3,"created":"2023-01-24 14:50:31.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/83/868083/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/83/868083/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/83/868083/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/83/868083/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/83/868083/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/83/868083/3"}}},"commit":{"parents":[{"commit":"0add65eddcd6e70d4fbe76951d76c66ecc1a34b3","subject":"Merge \"Add OPENSTACK_KEYSTONE_ENDPOINT_TYPE config opt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/0add65eddcd6e70d4fbe76951d76c66ecc1a34b3"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2022-12-19 12:54:25.000000000","tz":60},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-01-24 14:47:50.000000000","tz":60},"subject":"Fix applying OPENSTACK_SERVER_DEFAULT_USER_DATA value","message":"Fix applying OPENSTACK_SERVER_DEFAULT_USER_DATA value\n\nThe final revision of I86ac21bf82c1667135abd4f20fb4514da0899450 is buggy\nand doesn\u0027t work.\nThis commit fixes it and adds a proper test.\n\nAlso, this commit updates the docs with a proper horizon version that\ncorresponds to Zed release (the feature was introduced before final Zed\ncut so the author tried to guess the final release number).\n\nChange-Id: Id921b69df13af3cc209236f0446d82dd30e4d8a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/489b10b94139bf554610a35a388cc4f10137b0a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/489b10b94139bf554610a35a388cc4f10137b0a2"}]},"branch":"refs/heads/master"},"8bd373d2deed34c900650a246dd0c20234905d1a":{"kind":"REWORK","_number":4,"created":"2023-01-25 11:26:36.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/83/868083/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/83/868083/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/83/868083/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/83/868083/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/83/868083/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/83/868083/4"}}},"commit":{"parents":[{"commit":"0add65eddcd6e70d4fbe76951d76c66ecc1a34b3","subject":"Merge \"Add OPENSTACK_KEYSTONE_ENDPOINT_TYPE config opt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/0add65eddcd6e70d4fbe76951d76c66ecc1a34b3"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2022-12-19 12:54:25.000000000","tz":60},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-01-25 11:26:15.000000000","tz":60},"subject":"Fix applying OPENSTACK_SERVER_DEFAULT_USER_DATA value","message":"Fix applying OPENSTACK_SERVER_DEFAULT_USER_DATA value\n\nThe final revision of I86ac21bf82c1667135abd4f20fb4514da0899450 is buggy\nand doesn\u0027t work.\nThis commit fixes it and adds a proper test.\n\nAlso, this commit updates the docs with a proper horizon version that\ncorresponds to Zed release (the feature was introduced before final Zed\ncut so the author tried to guess the final release number).\n\nChange-Id: Id921b69df13af3cc209236f0446d82dd30e4d8a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/8bd373d2deed34c900650a246dd0c20234905d1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/8bd373d2deed34c900650a246dd0c20234905d1a"}]},"branch":"refs/heads/master"},"bb69655154e1a584cca8ac5a75b00f43e57ea7f6":{"kind":"REWORK","_number":5,"created":"2023-01-25 14:58:02.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/83/868083/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/83/868083/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/83/868083/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/83/868083/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/83/868083/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/83/868083/5"}}},"commit":{"parents":[{"commit":"0add65eddcd6e70d4fbe76951d76c66ecc1a34b3","subject":"Merge \"Add OPENSTACK_KEYSTONE_ENDPOINT_TYPE config opt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/0add65eddcd6e70d4fbe76951d76c66ecc1a34b3"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2022-12-19 12:54:25.000000000","tz":60},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-01-25 14:57:52.000000000","tz":60},"subject":"Fix applying OPENSTACK_SERVER_DEFAULT_USER_DATA value","message":"Fix applying OPENSTACK_SERVER_DEFAULT_USER_DATA value\n\nThe final revision of I86ac21bf82c1667135abd4f20fb4514da0899450 is buggy\nand doesn\u0027t work.\nThis commit fixes it and adds a proper test.\n\nAlso, this commit updates the docs with a proper horizon version that\ncorresponds to Zed release (the feature was introduced before final Zed\ncut so the author tried to guess the final release number).\n\nChange-Id: Id921b69df13af3cc209236f0446d82dd30e4d8a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/bb69655154e1a584cca8ac5a75b00f43e57ea7f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/bb69655154e1a584cca8ac5a75b00f43e57ea7f6"}]},"branch":"refs/heads/master"},"c8ee8c1965b280e24fb6266cbb2523230fea605f":{"kind":"REWORK","_number":6,"created":"2023-01-26 13:52:49.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/83/868083/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/83/868083/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/83/868083/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/83/868083/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/83/868083/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/83/868083/6"}}},"commit":{"parents":[{"commit":"5307672367d90324d7b313eb432374de0047d319","subject":"Merge \"Remove console_type parameter for server_mks_console function\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/5307672367d90324d7b313eb432374de0047d319"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2022-12-19 12:54:25.000000000","tz":60},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2023-01-26 13:42:15.000000000","tz":60},"subject":"Fix applying OPENSTACK_SERVER_DEFAULT_USER_DATA value","message":"Fix applying OPENSTACK_SERVER_DEFAULT_USER_DATA value\n\nThe final revision of I86ac21bf82c1667135abd4f20fb4514da0899450 is buggy\nand doesn\u0027t work.\nThis commit fixes it and adds a proper test.\n\nAlso, this commit updates the docs with a proper horizon version that\ncorresponds to Zed release (the feature was introduced before final Zed\ncut so the author tried to guess the final release number).\n\nChange-Id: Id921b69df13af3cc209236f0446d82dd30e4d8a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/c8ee8c1965b280e24fb6266cbb2523230fea605f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/c8ee8c1965b280e24fb6266cbb2523230fea605f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
