)]}'
{"id":"openstack%2Fhorizon~878751","triplet_id":"openstack%2Fhorizon~stable%2Fussuri~If43781468b4c68efef723c1a4a9749e3d1f8350b","project":"openstack/horizon","branch":"stable/ussuri","topic":"fix-zuul","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-03-29 11:49:23.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"hashtags":[],"change_id":"If43781468b4c68efef723c1a4a9749e3d1f8350b","subject":"Drop retired openstack/xstatic-font-awesome repo","status":"ABANDONED","created":"2023-03-28 08:59:13.000000000","updated":"2023-03-29 11:49:26.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6e0d5505fab761d4271ddc70d0224ce97cae0eff","_number":878751,"virtual_id_number":878751,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-03-29 07:31:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":-1,"date":"2023-03-29 10:43:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-28 08:59:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-29 10:43:45.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"}],"messages":[{"id":"36068f3a9b27e15cc8b6b0ca1261bfecff57d81d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-03-28 08:59:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"68c9a95041d868563a4050ac4318f9729296bb30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 08:59:39.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/horizon on branch master.  The error was:\n\n  Job horizon-nodejs18-run-test-xstatic-master not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      queue: horizon\n      templates:\n        - check-requirements\n        - horizon-cross-jobs\n        - horizon-nodejs-jobs\n        - horizon-non-primary-django-jobs\n        - openstack-python3-jobs\n        - periodic-stable-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      check:\n        jobs:\n          - horizon-selenium-headless\n          - horizon-integration-tests\n          - horizon-dsvm-tempest-plugin\n          - horizon-tox-bandit-baseline\n          - horizon-tempest-plugin-ipv6\n      gate:\n        jobs:\n          - horizon-dsvm-tempest-plugin\n          - horizon-selenium-headless\n          - horizon-integration-tests\n          - horizon-tempest-plugin-ipv6\n      experimental:\n        jobs:\n          - horizon-integration-tests-xstatic-master\n          - horizon-tox-py36-xstatic-master\n          - horizon-nodejs18-run-test-xstatic-master\n      periodic:\n        jobs:\n          - horizon-nodejs18-run-test\n          - horizon-integration-tests\n\n  in \"openstack/horizon/.zuul.d/project.yaml@master\", line 1, column 3","accounts_in_message":[],"_revision_number":1},{"id":"fc683fbb1188f589e03e53d651eff864ef509bea","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-03-28 14:55:01.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3df31591bcab126303a0c8f484b4375e70a5fd8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-03-28 14:57:53.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"8040a25f60bfae0647619fd662981c0ad298708a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 14:58:28.000000000","message":"Patch Set 2: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/horizon on branch master.  The error was:\n\n  Job horizon-nodejs18-run-test-xstatic-master not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      queue: horizon\n      templates:\n        - check-requirements\n        - horizon-cross-jobs\n        - horizon-nodejs-jobs\n        - horizon-non-primary-django-jobs\n        - openstack-python3-jobs\n        - periodic-stable-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      check:\n        jobs:\n          - horizon-selenium-headless\n          - horizon-integration-tests\n          - horizon-dsvm-tempest-plugin\n          - horizon-tox-bandit-baseline\n          - horizon-tempest-plugin-ipv6\n      gate:\n        jobs:\n          - horizon-dsvm-tempest-plugin\n          - horizon-selenium-headless\n          - horizon-integration-tests\n          - horizon-tempest-plugin-ipv6\n      experimental:\n        jobs:\n          - horizon-integration-tests-xstatic-master\n          - horizon-tox-py36-xstatic-master\n          - horizon-nodejs18-run-test-xstatic-master\n      periodic:\n        jobs:\n          - horizon-nodejs18-run-test\n          - horizon-integration-tests\n\n  in \"openstack/horizon/.zuul.d/project.yaml@master\", line 1, column 3","accounts_in_message":[],"_revision_number":2},{"id":"2488352aa327e6e6e8920e6255351eb7915b3fda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-03-29 06:03:37.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"72bf1a008b8c6cef21e60949ac62e16c6d8ea585","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 07:31:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e327bc6e793346bb98d6429368887c00\n\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/df7d62a7ab0d4e5ab1d09f7544f66546 : SUCCESS in 5m 53s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/f0de5999e53c43cea631d8639931009c : SUCCESS in 8m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2b5ef9b7e048451fa85a60c6b31916bc : SUCCESS in 9m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7f3091e35444b06ba05e54d6116fc83 : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9381cc8660ea42a685dd69bcb0b68c33 : SUCCESS in 8m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f066b5cc3134249865ec7027a8a0e8e : SUCCESS in 7m 00s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/1791d06a981342c2ba96af0d49d5a847 : SUCCESS in 7m 33s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/ec03d5fe28684e02ae949641e0d9aa17 : SUCCESS in 1h 27m 24s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/ba45ae5b6ac048f081c3889edf88dca5 : SUCCESS in 27m 04s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b8386d2fb27a49219e40e576f79728b7 : SUCCESS in 5m 19s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/51c52977e74b4561a1d00283ea26d364 : SUCCESS in 26m 29s","accounts_in_message":[],"_revision_number":3},{"id":"8fa4f38913d6275b928be1e93528c3f0545abe95","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-03-29 10:43:45.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e9ef001187e5e3c270da6bbfc1a701303427a0ce","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-03-29 11:49:23.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6e0d5505fab761d4271ddc70d0224ce97cae0eff","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-03-29 11:49:26.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"da7c0299311f229699c70f34e9a1380f7503326b","revisions":{"278454bc82e6d9555b3af8018dfcd4dfd6be909d":{"kind":"REWORK","_number":1,"created":"2023-03-28 08:59:13.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/51/878751/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/51/878751/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/51/878751/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/51/878751/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/51/878751/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/51/878751/1"}}},"commit":{"parents":[{"commit":"e6f3952b878d6b04fde9742987e0f37a1cfad3e5","subject":"Merge \"Modified the image page to use CSS class\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/e6f3952b878d6b04fde9742987e0f37a1cfad3e5"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-03-28 08:58:31.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-03-28 08:59:11.000000000","tz":120},"subject":"Drop retired openstack/xstatic-font-awesome repo","message":"Drop retired openstack/xstatic-font-awesome repo\n\nAlso drop job that no longer exists.\n\nChange-Id: If43781468b4c68efef723c1a4a9749e3d1f8350b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/278454bc82e6d9555b3af8018dfcd4dfd6be909d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/278454bc82e6d9555b3af8018dfcd4dfd6be909d"}]},"branch":"refs/heads/stable/ussuri"},"115c4f45e1112e769eb72111a389095932c93503":{"kind":"REWORK","_number":2,"created":"2023-03-28 14:57:53.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/51/878751/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/51/878751/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/51/878751/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/51/878751/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/51/878751/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/51/878751/2"}}},"commit":{"parents":[{"commit":"e6f3952b878d6b04fde9742987e0f37a1cfad3e5","subject":"Merge \"Modified the image page to use CSS class\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/e6f3952b878d6b04fde9742987e0f37a1cfad3e5"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-03-28 08:58:31.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2023-03-28 14:57:53.000000000","tz":0},"subject":"Drop retired openstack/xstatic-font-awesome repo","message":"Drop retired openstack/xstatic-font-awesome repo\n\nAlso drop job that no longer exists.\n\nChange-Id: If43781468b4c68efef723c1a4a9749e3d1f8350b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/115c4f45e1112e769eb72111a389095932c93503"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/115c4f45e1112e769eb72111a389095932c93503"}]},"branch":"refs/heads/stable/ussuri"},"da7c0299311f229699c70f34e9a1380f7503326b":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-03-29 06:03:37.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/51/878751/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/51/878751/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/51/878751/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/51/878751/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/51/878751/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/51/878751/3"}}},"commit":{"parents":[{"commit":"e6f3952b878d6b04fde9742987e0f37a1cfad3e5","subject":"Merge \"Modified the image page to use CSS class\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/e6f3952b878d6b04fde9742987e0f37a1cfad3e5"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-03-28 08:58:31.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2023-03-29 06:03:36.000000000","tz":0},"subject":"Drop retired openstack/xstatic-font-awesome repo","message":"Drop retired openstack/xstatic-font-awesome repo\n\nChange-Id: If43781468b4c68efef723c1a4a9749e3d1f8350b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/da7c0299311f229699c70f34e9a1380f7503326b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/da7c0299311f229699c70f34e9a1380f7503326b"}]},"branch":"refs/heads/stable/ussuri","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
