)]}'
{"id":"openstack%2Fhorizon~878813","triplet_id":"openstack%2Fhorizon~master~I228881b7a9153fbe3c623346cb130930fa2028d3","project":"openstack/horizon","branch":"master","topic":"ru_lang_fix","attention_set":{},"removed_from_attention_set":{"32358":{"account":{"_account_id":32358,"name":"Mikhail Stolyarov","display_name":"toggetit","email":"x3om6ak@gmail.com","username":"toggetit"},"last_update":"2023-03-29 09:15:56.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I228881b7a9153fbe3c623346cb130930fa2028d3","subject":"[locale][ru] Just small annoying fix RABC-\u003eRBAC","status":"ABANDONED","created":"2023-03-29 07:27:08.000000000","updated":"2023-03-29 09:15:56.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8eb2cd988fbccb9324a2a2e89435b88b7e00d32b","_number":878813,"virtual_id_number":878813,"owner":{"_account_id":32358,"name":"Mikhail Stolyarov","display_name":"toggetit","email":"x3om6ak@gmail.com","username":"toggetit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-03-29 07:43:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":-2,"date":"2023-03-29 08:44:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-29 07:43:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-29 08:44:40.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"}],"messages":[{"id":"b570a9003ba5965202dd1388c0de46fe326ae3ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32358,"name":"Mikhail Stolyarov","display_name":"toggetit","email":"x3om6ak@gmail.com","username":"toggetit"},"date":"2023-03-29 07:27:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"089cc8d784d7c712dcb4911782b951bdc51c4218","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 07:43:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9c1a0b546594c64a48d5bc512839f38\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/e1514e1c356d42658114abfeddd85664 : SUCCESS in 6m 14s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/0c029a3084294184b40d32e36c6133ca : SUCCESS in 6m 43s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/76c5858af4504948a69fe193e71795f3 : SUCCESS in 3m 35s (non-voting)\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/3fc2b1f0337b44caab0868f822d6df07 : SUCCESS in 4m 46s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/7ad10c814d2847b080026913bd233cb3 : SUCCESS in 7m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/29bd265b0ff146028e4bd072390b0c76 : SUCCESS in 11m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3efe9579c82d41408fbf675f06f2b151 : SUCCESS in 5m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/940a0359eff7462aa517db3435b6b906 : SUCCESS in 10m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1069b352753249bda8a07d1fb18723bb : SUCCESS in 8m 00s","accounts_in_message":[],"_revision_number":1},{"id":"924e115a9097524a57d6970a8aa41f272ab49f5c","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-03-29 08:44:40.000000000","message":"Patch Set 1: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8eb2cd988fbccb9324a2a2e89435b88b7e00d32b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32358,"name":"Mikhail Stolyarov","display_name":"toggetit","email":"x3om6ak@gmail.com","username":"toggetit"},"date":"2023-03-29 09:15:56.000000000","message":"Abandoned\n\nNot needed","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"330014a0b643f55dfb543093b95b904200a787c6","revisions":{"330014a0b643f55dfb543093b95b904200a787c6":{"kind":"REWORK","_number":1,"created":"2023-03-29 07:27:08.000000000","uploader":{"_account_id":32358,"name":"Mikhail Stolyarov","display_name":"toggetit","email":"x3om6ak@gmail.com","username":"toggetit"},"ref":"refs/changes/13/878813/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/13/878813/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/878813/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/878813/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/878813/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/13/878813/1"}}},"commit":{"parents":[{"commit":"ac24c85b629b04d1bc170c4b32f752763f3b62b8","subject":"Merge \"Remove usage of openstack/xstatic-font-awesome\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/ac24c85b629b04d1bc170c4b32f752763f3b62b8"}]}],"author":{"name":"Mikhail Stolyarov","email":"x3om6ak@gmail.com","date":"2023-03-29 07:25:45.000000000","tz":180},"committer":{"name":"Mikhail Stolyarov","email":"x3om6ak@gmail.com","date":"2023-03-29 07:25:45.000000000","tz":180},"subject":"[locale][ru] Just small annoying fix RABC-\u003eRBAC","message":"[locale][ru] Just small annoying fix RABC-\u003eRBAC\n\nChange-Id: I228881b7a9153fbe3c623346cb130930fa2028d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/330014a0b643f55dfb543093b95b904200a787c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/330014a0b643f55dfb543093b95b904200a787c6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
