)]}'
{"id":"openstack%2Fhorizon~879997","triplet_id":"openstack%2Fhorizon~master~I1ddd89bd99510a981f2f3b6e198786890be22280","project":"openstack/horizon","branch":"master","topic":"update-jquery","attention_set":{},"removed_from_attention_set":{"29313":{"account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"last_update":"2023-04-24 21:46:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1ddd89bd99510a981f2f3b6e198786890be22280","subject":"Fix \"jQuery.fn.submit() event shorthand is deprecated\" warning","status":"MERGED","created":"2023-04-10 19:20:57.000000000","updated":"2023-04-24 21:47:39.000000000","submitted":"2023-04-24 21:46:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"879997-update-jquery","meta_rev_id":"904c60e21b910f7a26371a6fa9f1bbb634dd53d4","_number":879997,"virtual_id_number":879997,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-24 21:46:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"all":[{"value":2,"date":"2023-04-24 20:00:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":2,"date":"2023-04-19 12:15:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":2,"date":"2023-04-19 15:32:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"all":[{"value":1,"date":"2023-04-24 20:00:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-10 20:39:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-19 12:15:46.000000000","updated_by":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"reviewer":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"state":"REVIEWER"},{"updated":"2023-04-19 15:32:22.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2023-04-24 20:00:59.000000000","updated_by":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"reviewer":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"state":"REVIEWER"}],"messages":[{"id":"6da23db3591a248ea8f560fbb9882d20470133a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-04-10 19:20:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a02f4a4445022631c4ec5dfb9d7a00f83286113","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-04-10 19:21:38.000000000","message":"Topic set to update-jquery","accounts_in_message":[],"_revision_number":1},{"id":"e7a9f0beccaa02e24d47d5d23d73a2ceec0bd77a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-10 20:39:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7debd06189bb45d19c34cec27c69ec2d\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/7bebeb7a33854a10b11f4c582495cf8f : SUCCESS in 5m 49s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/d1c877b6a9af44088233964be7809332 : SUCCESS in 5m 13s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/9a2f66fc36df43d888024200b10e2aa6 : SUCCESS in 3m 53s (non-voting)\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/d7bcffd9a729422399c3b57e9639e298 : SUCCESS in 3m 57s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/d65222ee964c4de3a770fdd31ed56753 : SUCCESS in 6m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f1756da88d84bd68a73ecc4a3e9392c : SUCCESS in 9m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea83fcfad76d4dc788bbd8a9995a0bc6 : SUCCESS in 4m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0155e9dcfde649598fd8fa83e753e627 : SUCCESS in 7m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfd5e5fb2bae49778a63080729b5323b : SUCCESS in 7m 42s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/7a2ac2056e2a4e819bc87017c129f9af : SUCCESS in 5m 43s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/0abccb418cef4f2e9d199085a0e7b575 : SUCCESS in 1h 16m 58s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/28cb0a9713cb44f29e9046f5b2686d29 : SUCCESS in 17m 33s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/007a2fb619da4391842057a002a2e01d : SUCCESS in 4m 23s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/4a26f40f211f45d9a397f1e922f855a9 : SUCCESS in 20m 10s","accounts_in_message":[],"_revision_number":1},{"id":"d8d8c233098e5bacb1cec948a4f810d3a35ea931","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2023-04-19 12:15:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"06de75e8a03822ea995397512e707b6783017941","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2023-04-19 15:32:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"10200869a669cd9a234ec1e4367056b1bdad9fd0","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2023-04-24 20:00:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2ee6066ffdc7ba8f3deee4580f6ea1c5c910cbbd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 20:01:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"13f3cbe08ae6807d9004adf8038503a3f6c556b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 21:46:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62b284a283924bff91d9f3902436341e\n\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/fbc170cd92354a098bab890396b2aa1c : SUCCESS in 4m 54s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/abdbaa80239c4b38a78e4cb0768565be : SUCCESS in 7m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d5f6416009d48b6bb37c2e81a1a2f0d : SUCCESS in 8m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b83ef0893fb641d59f7d6dfe2634b94c : SUCCESS in 5m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c600b7b6dabc4a61936335b895033531 : SUCCESS in 7m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6095e073c0b043779333337e4eb3a9b0 : SUCCESS in 6m 31s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/6395cb2e1af44ef68b1b9a4b8d5bf824 : SUCCESS in 21m 03s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/63995c0a39fe4045a8b3eeafd40360a8 : SUCCESS in 5m 40s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/dce9859e5b1a4e3eb51da53e667abad0 : SUCCESS in 1h 41m 16s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/f44021b5a2794ce19be3e452c3f720b9 : SUCCESS in 36m 33s","accounts_in_message":[],"_revision_number":1},{"id":"7f7235472e3a8fa32f88d95113495b1357cbe118","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 21:46:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"904c60e21b910f7a26371a6fa9f1bbb634dd53d4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 21:47:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/947cbc1fc0914fb1b85b185f7f83e000\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b917833df1c4284afa448d784ee8798 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cf158a213a03473d3df25123a075bdeca847db38","revisions":{"cf158a213a03473d3df25123a075bdeca847db38":{"kind":"REWORK","_number":1,"created":"2023-04-10 19:20:57.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/97/879997/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/97/879997/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/97/879997/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/97/879997/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/97/879997/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/97/879997/1"}}},"commit":{"parents":[{"commit":"7d44796eca10fa61d4d98ab98358c5932746734e","subject":"Merge \"Unskipping test_floatingip_associate_disassociate test case\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/7d44796eca10fa61d4d98ab98358c5932746734e"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-04-10 18:42:52.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-04-10 18:42:52.000000000","tz":330},"subject":"Fix \"jQuery.fn.submit() event shorthand is deprecated\" warning","message":"Fix \"jQuery.fn.submit() event shorthand is deprecated\" warning\n\nIn horizon.forms.handle_submit(), replaced the shorthand .submit()\neventbinding with .on(\u0027submit\u0027, function ()) to avoid the deprecated\nshorthand usage as suggested by the jQuery documentation[1].\n\n[1] https://api.jquery.com/submit/\n\nChange-Id: I1ddd89bd99510a981f2f3b6e198786890be22280\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/cf158a213a03473d3df25123a075bdeca847db38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/cf158a213a03473d3df25123a075bdeca847db38"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
