)]}'
{"id":"openstack%2Fhorizon~886113","triplet_id":"openstack%2Fhorizon~master~I6a06c787192b99c562d692531714d5f2a0d017e2","project":"openstack/horizon","branch":"master","topic":"883953","attention_set":{},"removed_from_attention_set":{"29313":{"account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"last_update":"2023-06-15 18:02:58.000000000","reason":"\u003cGERRIT_ACCOUNT_29313\u003e replied on the change","reason_account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}}},"hashtags":[],"change_id":"I6a06c787192b99c562d692531714d5f2a0d017e2","subject":"[DNM] Revert Update jQuery Changes","status":"ABANDONED","created":"2023-06-14 17:19:53.000000000","updated":"2023-06-16 06:32:30.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8313ede2a5db14f9c4d6a483f8e52cce4c261e7f","_number":886113,"virtual_id_number":886113,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-14 18:33:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8b4833683257529ab1dac6f5c2ee817d07257f5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-06-14 17:19:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"356cf97fe84d074091ddd6dbcbbdf0267f449162","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 18:33:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e3dbef52664144f49adab91d7dffe873\n\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/e91623e4fde84d98af4c80b19d1bdd0c : SUCCESS in 4m 48s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/bf61c8797abe4a8080c4da78b6e6a356 : SUCCESS in 7m 15s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/50eef8774afc42f89347b18be1e8d060 : FAILURE in 1h 13m 11s","accounts_in_message":[],"_revision_number":1},{"id":"3a9b5aa4ee7419ef43d407eb9b0001a303edf766","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-06-15 17:32:13.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d64eba8d25cad2612232cab6498b4a3aa277f3bc","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-06-15 18:02:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cdfd48772c7754459546924b9b2cff913ea4664d","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-06-16 06:32:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8313ede2a5db14f9c4d6a483f8e52cce4c261e7f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-06-16 06:32:30.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"12ae30290e2828ce9cf2582893223da6a65a60a5","revisions":{"35e2a13f64bcca97b76aeb87365b911dafe41c8f":{"kind":"REWORK","_number":1,"created":"2023-06-14 17:19:53.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/13/886113/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/13/886113/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/886113/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/886113/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/886113/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/13/886113/1"}}},"commit":{"parents":[{"commit":"29eefb9d6d5ad131ed98eb6766a9492c0efbe312","subject":"Update cirros image to cirros-0.6.1-x86_64 to fix integration job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/29eefb9d6d5ad131ed98eb6766a9492c0efbe312"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-06-14 17:15:19.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-06-14 17:17:08.000000000","tz":330},"subject":"[DNM] Revert Update jQuery Changes","message":"[DNM] Revert Update jQuery Changes\n\nChange-Id: I6a06c787192b99c562d692531714d5f2a0d017e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/35e2a13f64bcca97b76aeb87365b911dafe41c8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/35e2a13f64bcca97b76aeb87365b911dafe41c8f"}]},"branch":"refs/heads/master"},"12ae30290e2828ce9cf2582893223da6a65a60a5":{"kind":"REWORK","_number":2,"created":"2023-06-15 17:32:13.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/13/886113/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/13/886113/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/886113/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/886113/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/13/886113/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/13/886113/2"}}},"commit":{"parents":[{"commit":"29eefb9d6d5ad131ed98eb6766a9492c0efbe312","subject":"Update cirros image to cirros-0.6.1-x86_64 to fix integration job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/29eefb9d6d5ad131ed98eb6766a9492c0efbe312"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-06-14 17:15:19.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-06-15 17:31:08.000000000","tz":330},"subject":"[DNM] Revert Update jQuery Changes","message":"[DNM] Revert Update jQuery Changes\n\nChange-Id: I6a06c787192b99c562d692531714d5f2a0d017e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/12ae30290e2828ce9cf2582893223da6a65a60a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/12ae30290e2828ce9cf2582893223da6a65a60a5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
