)]}'
{"id":"openstack%2Fhorizon~890217","triplet_id":"openstack%2Fhorizon~master~I8fdcdddac6869af59a330d1181aed83c2c5770b3","project":"openstack/horizon","branch":"master","topic":"update-jquery","attention_set":{},"removed_from_attention_set":{"29313":{"account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"last_update":"2023-08-04 04:55:06.000000000","reason":"\u003cGERRIT_ACCOUNT_29313\u003e replied on the change","reason_account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}},"6914":{"account":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"last_update":"2023-08-02 15:10:07.000000000","reason":"\u003cGERRIT_ACCOUNT_6914\u003e replied on the change","reason_account":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"}},"8648":{"account":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"last_update":"2023-08-02 09:35:31.000000000","reason":"\u003cGERRIT_ACCOUNT_8648\u003e replied on the change","reason_account":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"}}},"hashtags":[],"change_id":"I8fdcdddac6869af59a330d1181aed83c2c5770b3","subject":"Bump minor version of XStatic-jQuery","status":"MERGED","created":"2023-08-01 15:41:46.000000000","updated":"2023-08-04 08:21:52.000000000","submitted":"2023-08-04 08:20:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890217-update-jquery","meta_rev_id":"daab6a11da8a071c60affbba158f0e56b1f7d399","_number":890217,"virtual_id_number":890217,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-04 08:20:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"all":[{"value":2,"date":"2023-08-02 15:10:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":2,"date":"2023-08-02 09:35:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"all":[{"value":1,"date":"2023-08-02 15:10:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-01 17:46:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-02 07:54:36.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"state":"REVIEWER"},{"updated":"2023-08-02 07:54:36.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"state":"REVIEWER"}],"messages":[{"id":"5d2162925fe1dc5b15d92d1bbd29e98a25d862d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-08-01 15:41:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"faf2a4f78a7a5b2b4d173d997637c70868bc8a78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 17:46:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3ca4972e0a34fc29506ef94a062d932\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/036d88d25654439f8fe3bc356e64808e : SUCCESS in 2m 02s\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/f70b9667868043089b2a3cc32c035cda : SUCCESS in 5m 27s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/a9096647ee814476905d5c1f8ea5cc49 : SUCCESS in 7m 04s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/a69a244dc5094190a53f611fa0c48f9a : SUCCESS in 4m 30s (non-voting)\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/f0ca31a59f86433d96b3d8b21b6fceae : SUCCESS in 4m 39s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/77c2596696fc4870ac1d6eb32a9346da : SUCCESS in 6m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bba0c694018946e589d34345d9baa3ce : SUCCESS in 11m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cb9056f1aa443c18799b3d5357890ae : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/585cb6e2cfb04ee782b667cfb092e015 : SUCCESS in 11m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/df5c1ca282c1440a89f9ab5443d0bf3a : SUCCESS in 11m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65b1c1802644452bbe1c7253765ac8f3 : SUCCESS in 6m 26s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/af1ef5e96b9d4314a40acefd6b4a273f : SUCCESS in 7m 34s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/cbdfa8eb6283466794cdd04bf5e2efd4 : SUCCESS in 1h 57m 58s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/56c57bad716e44c98944c9c0ff2e7c08 : SUCCESS in 18m 47s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/89f32f61ae6f4f3f9f6d83d17449bff7 : SUCCESS in 4m 15s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/9baa70daad0049c99a275f58343a1a6c : SUCCESS in 37m 34s","accounts_in_message":[],"_revision_number":1},{"id":"5d96c3bc9a75fb6881c249f431c0379f3508119d","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2023-08-02 09:35:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8a7291fb4660bb02d156645ad1f079c2fe62153e","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2023-08-02 15:10:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ed869871632fe7a3813a4cf9eefe03bba467b54a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 15:10:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f2f7d2a038b75c03fb7013fc172411a96b604322","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 16:13:17.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d7d0f34c92fb49a38952874b431656bf\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c41118372b8246baa695a6e3a1a34740 : SUCCESS in 2m 46s\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/94336d7fc4a3444cbe28d90d1eb28868 : SUCCESS in 6m 34s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/2a143f9fabf049ce9feec193d1a852aa : SUCCESS in 6m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9c26cc0be0864b6abbffd79cade5b88e : SUCCESS in 8m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d73970b87d444dc587677c7955641d4a : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b74500e243c64e2b81bb69d64d397bf1 : SUCCESS in 11m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5c3eb8547ee0412cb255af23df911853 : SUCCESS in 11m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab9a1795a16f44528913b3a376abf9f2 : SUCCESS in 7m 13s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/85347da0f2c348459a8516a6f929437a : SUCCESS in 41m 14s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/9a68a99988ed4ab2b00bd823d7911276 : SUCCESS in 6m 21s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/0c2cf3abd82c4ff192aa173082576694 : FAILURE in 51m 32s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/34a7de8c0e914d9a83827a5cac3a5d2f : SUCCESS in 39m 20s","accounts_in_message":[],"_revision_number":1},{"id":"e5434f7f5493e8266b55243b14e5bc9a7fc9a34e","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-08-03 01:26:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f8ceb5d392423ef3539a61e6a96573eaf4f4db29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 02:59:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c9dbbf4e2f64da69dc54b8922c7bf0b\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/164a164f04b64310b17efaf2879fa217 : SUCCESS in 2m 01s\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/f60bac45ac234e81a5de7a6017060b3d : SUCCESS in 6m 19s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/7a30a1cdcafe4e70b4c61e6f60994dd0 : SUCCESS in 9m 15s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/b72da2660f134a7d8efdc137fae89882 : SUCCESS in 3m 44s (non-voting)\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/fd288ec2f85147b192c1b06562e0c8cc : SUCCESS in 4m 37s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/6e1032d112924b669d17b6e64c61e8c9 : SUCCESS in 6m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef7ffc309e224e8f977969b88058aede : SUCCESS in 8m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88691560119d4c6a8ba85c865140f757 : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31c18255b5904292b19b5a87e4c32a01 : SUCCESS in 11m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/72c551888dcb4af29845b14f70fdbff0 : SUCCESS in 8m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a0378cd35ed47d19b2fda3f8c626d73 : SUCCESS in 7m 56s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/7403c58a7bea419db66727917eada323 : SUCCESS in 6m 47s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/bf1868f0b281421fb3f103c528680387 : SUCCESS in 1h 24m 31s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/51a19fc00f034a98a0bc8d28fcc61475 : SUCCESS in 24m 12s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d7bf5524c44746b2b43e58f78490c84a : SUCCESS in 4m 47s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/d088d4cac4f0452f9c357f01e9b04b4c : SUCCESS in 24m 07s","accounts_in_message":[],"_revision_number":1},{"id":"9016d5d547078a94f51f95ec44bfb324558169f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 02:59:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"227eebd76b546a66932d72d87f110fa3483553ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 03:52:48.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f1990bd13a8a47c197109239ebdd5b1e\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d610031db19a435ebe525782b7a73093 : SUCCESS in 2m 14s\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/3cf638e14dfd47dbaec96b070429e2d0 : SUCCESS in 4m 57s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/dc399960d98a4c23a57ebc289ad4ca05 : SUCCESS in 10m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ebbdb0e34fe94e8287c0d396f7ff6a9e : TIMED_OUT in 41m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b06e12d522c4031843299e4fa8199e3 : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8378cbc6edc4f6480981f402543d880 : SUCCESS in 10m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/84d71c3bef834534860a3256f939b61f : SUCCESS in 10m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad157d80d3d14760b5ad7212c3c8f3c8 : SUCCESS in 8m 58s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/8fa6fc95ddc5439387ba24ee1da4c817 : SUCCESS in 46m 33s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/31f2f3df6ac646f2905b457e97bba255 : SUCCESS in 8m 08s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/519e2cba024f47ae810a1ac7a75585ea : FAILURE in 50m 05s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/7243d91b605448bda3c60db79cdf501b : SUCCESS in 36m 02s","accounts_in_message":[],"_revision_number":1},{"id":"01f5b6e741663bee22b2353fe841685bb57bd1fa","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-08-03 08:20:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"85e30aaa4e8f98356b88bb2f977085eb4d6de90e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-08-03 08:22:53.000000000","message":"Topic bug/1955556 removed","accounts_in_message":[],"_revision_number":1},{"id":"96aebf73180f5f1e66db4f564dfd0cdf60dfbc12","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-08-03 08:23:02.000000000","message":"Topic set to update-jquery","accounts_in_message":[],"_revision_number":1},{"id":"83eb5bfcbba63300f17486d7d10914ca246db145","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 10:31:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/48c22cb7e6d242cf98b69c77188f5261\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/1807906ac9794ba181fc4bd3fe57f13c : SUCCESS in 2m 36s\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/74436a1a9da648fa852da43d0e13a673 : SUCCESS in 6m 20s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/5f9ce67554cd4db7b6bdfad8d08bb4ed : SUCCESS in 5m 51s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/89989cf1e81245ad97ff8021968e8d0a : SUCCESS in 3m 38s (non-voting)\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/60cf900c3dfc4cf1a989477e132b1a14 : SUCCESS in 4m 33s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/628a7f44402442e5821c6393cb5a4fb3 : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/354edbc5556146448c4249c5c9767366 : SUCCESS in 10m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73160a6bbf1e4441940ab6723dece807 : SUCCESS in 6m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08da3366771447e7aafa5e8cda1c7459 : SUCCESS in 10m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/23c333d273004386806ac81f3eba0cde : SUCCESS in 10m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d631d966138449c58cf6667af4da6d89 : SUCCESS in 6m 47s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/0188629ed51b4bc7a123a26064180d54 : SUCCESS in 6m 22s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/dca9cc0e133240cf802c8833374047fa : TIMED_OUT in 2h 03m 14s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/f349bd6f976d43ddbb8ed78d67546083 : SUCCESS in 24m 29s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3e747bf1ac7b44d7b4b456d6edb55e01 : SUCCESS in 3m 59s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/304ab457f07b4b589c291917ea1bf0a6 : SUCCESS in 23m 37s","accounts_in_message":[],"_revision_number":1},{"id":"ff7ef1263d00571638b4d63d201480151271fc3a","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-08-03 14:26:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"73717aa44d8ff2dae3dc15f686be463ae167aa52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 15:15:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fa0460c4d1764167a7b7e1fcca68f321\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/267eece9521d4c4fbd825e026ceb0d19 : SUCCESS in 2m 40s\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/e063355a4d7d42c89c7933d7730e7be8 : SUCCESS in 5m 21s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/baf72dd3d417456184d638d58f385a72 : SUCCESS in 7m 01s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/02e52811af8b410a9a14309e7458d5a7 : SUCCESS in 3m 36s (non-voting)\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/3c6af30904fc4122861f28ba1a2df7ce : SUCCESS in 5m 15s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/15950d9306274e938e39039e9b82d112 : SUCCESS in 7m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/84cf3db0ff6b432dabd8416dd60e058d : SUCCESS in 10m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aef8f4a19dda496db736e5e82c85ce29 : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2ef701e6cd945b0a8f60a6ee517fb85 : SUCCESS in 10m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2764bac1fd9143cfb22aeafc16791b00 : SUCCESS in 11m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5645e667505c4580b90ac764c1696857 : SUCCESS in 8m 41s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/83a45e1ad89d473d8450ea399808b11c : SUCCESS in 8m 28s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/3cade46a591840269025714f19866fd4 : FAILURE in 36m 48s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/95952691f8b94f0795df6b9d3a50cf81 : SUCCESS in 39m 34s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ee0cfb1574824336ae2608fe7b5e7eb6 : SUCCESS in 4m 50s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/31438e2b70354da482455da6eae64702 : SUCCESS in 40m 34s","accounts_in_message":[],"_revision_number":1},{"id":"a9f88d3e6322c6dc8af44e9452bde99610632fc7","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-08-04 04:55:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1c291ce98f915bb80db39c35b6283aa004a60398","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 06:16:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4730c313fc24adebb6720f4b868c708\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d6c93611592c456cbc2ade3e22e76b71 : SUCCESS in 2m 17s\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/39ced662c1914e2f930b826c5e31fa8f : SUCCESS in 5m 42s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/a586133630d0484388e376d159606594 : SUCCESS in 7m 35s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/98fb6f212ee9416b96f9a4cb5396bbcc : SUCCESS in 3m 48s (non-voting)\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/e87ff78123194db7ae39e607908e0ef8 : SUCCESS in 4m 56s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/563f712a167440af957da70e7b9ddbe0 : SUCCESS in 7m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/214447da63a9422ba47f59d1341e7524 : SUCCESS in 11m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3cba92c659b4dd5a70b504c0f2cc868 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d3be220b3604cd79d202f6990c2ede1 : SUCCESS in 10m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/54b0111dce7541a5af7dfa5b892665df : SUCCESS in 8m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e7c984391164df99fe25a002ec54a0c : SUCCESS in 7m 45s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/b27bc8e7061349dc8beefc72b867319c : SUCCESS in 12m 36s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/1fae5c1fcde64013aaa8ccb1fbb51d0c : SUCCESS in 1h 20m 17s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/d4521bb24e3c428ba5059164b3fd86c9 : SUCCESS in 38m 26s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/35a5c5a38f3942ffb6ee96bcb7f66946 : SUCCESS in 3m 32s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/cf25d52d2338402095ccf4d9f86c7f3c : SUCCESS in 24m 06s","accounts_in_message":[],"_revision_number":1},{"id":"b4def7d72f3137de3838ee9de9bde411719a4e35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 06:16:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"deaaff3060c3c4c6d251e3d498370cb9cec86098","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 08:20:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf647bf70d3349c9ac76bec09be38a2e\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/708aa9dfb2974659ba0f2ff6ef3bc824 : SUCCESS in 2m 06s\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/e8e31bb32e9c4319b3223606ef2d1442 : SUCCESS in 4m 33s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/52dec572aae1480e82cfe348046d6b62 : SUCCESS in 7m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/efbe730499374e0385b88d0554ff8935 : SUCCESS in 10m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/786e315631af415b874cf2de15c4af75 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c6b80b7389b4b95929eeebd47860c47 : SUCCESS in 11m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9d8438a18bf048bda8b2ef6e2cef3000 : SUCCESS in 11m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7681014f653842f9891659add62c7687 : SUCCESS in 7m 29s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/48211e0e050f4aecac5b831d5900c955 : SUCCESS in 37m 01s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/51662e65680c4dfab0b72c06bd61d0a2 : SUCCESS in 6m 15s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/8ef9be0d6a2d4e53a94aecd065373fc8 : SUCCESS in 1h 57m 21s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/8eca11c984df4b618cd98ad05e071cfe : SUCCESS in 35m 04s","accounts_in_message":[],"_revision_number":1},{"id":"e68d22100899a56eccbe79c12043454106b3c14f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 08:20:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"daab6a11da8a071c60affbba158f0e56b1f7d399","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 08:21:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf6b165a9f8f47e6980d5b0d6de225ec\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ca7c750abef4383994ec8dfc4909556 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"20bdaa386791cc23809b298762217dd1a2b0f071","revisions":{"20bdaa386791cc23809b298762217dd1a2b0f071":{"kind":"REWORK","_number":1,"created":"2023-08-01 15:41:46.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/17/890217/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/17/890217/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/17/890217/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/17/890217/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/17/890217/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/17/890217/1"}}},"commit":{"parents":[{"commit":"b4554a195151ae6343e7b204394c11412078c036","subject":"Merge \"Fix ports tab at Launch Instance dialog\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/b4554a195151ae6343e7b204394c11412078c036"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-08-01 15:36:07.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2023-08-01 15:36:07.000000000","tz":330},"subject":"Bump minor version of XStatic-jQuery","message":"Bump minor version of XStatic-jQuery\n\nThis patch update minor version of XStatic-jQuery\nto 3.5.1.1 in the ``requirements.txt`` file.\n\nCloses-bug: #1955556\nChange-Id: I8fdcdddac6869af59a330d1181aed83c2c5770b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/20bdaa386791cc23809b298762217dd1a2b0f071"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/20bdaa386791cc23809b298762217dd1a2b0f071"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
