)]}'
{"id":"openstack%2Fhorizon~890907","triplet_id":"openstack%2Fhorizon~master~I0fac006d5d2a95807c5a14d60b387b6ccaa749e8","project":"openstack/horizon","branch":"master","topic":"fix_duplicate_import_middleware","attention_set":{},"removed_from_attention_set":{"29313":{"account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"last_update":"2023-10-16 03:59:38.000000000","reason":"\u003cGERRIT_ACCOUNT_29313\u003e replied on the change","reason_account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}},"36274":{"account":{"_account_id":36274,"name":"Elyas Ebrahimpour","display_name":"Elyas Ebrahimpour","email":"eliasebrahimpour@gmail.com","username":"eliasebrahimpour"},"last_update":"2025-05-21 17:25:47.000000000","reason":"Change was abandoned"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-05-21 17:25:47.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I0fac006d5d2a95807c5a14d60b387b6ccaa749e8","subject":"removed duplicate lines","status":"ABANDONED","created":"2023-08-09 11:32:05.000000000","updated":"2025-05-21 17:25:47.000000000","total_comment_count":8,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"a23b138a50d1026f92390feda52f348090e73a73","_number":890907,"virtual_id_number":890907,"owner":{"_account_id":36274,"name":"Elyas Ebrahimpour","display_name":"Elyas Ebrahimpour","email":"eliasebrahimpour@gmail.com","username":"eliasebrahimpour"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-08-09 12:13:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":0,"date":"2023-10-16 03:59:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":-1,"date":"2023-10-16 01:36:59.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":-1,"date":"2023-10-17 07:06:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}],"CC":[{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-09 12:13:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-29 18:40:29.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2023-10-12 14:29:02.000000000","updated_by":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"reviewer":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"state":"CC"},{"updated":"2023-10-16 01:36:49.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2023-10-16 01:36:59.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2023-10-17 07:06:45.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2025-05-21 17:25:02.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2025-05-21 17:25:02.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"state":"CC"}],"messages":[{"id":"ad595db48b5853e95ba415df89612400b937eb16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36274,"name":"Elyas Ebrahimpour","display_name":"Elyas Ebrahimpour","email":"eliasebrahimpour@gmail.com","username":"eliasebrahimpour"},"date":"2023-08-09 11:32:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b8f604f343b4a9692596a6ebe8b467d5acc3f01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36274,"name":"Elyas Ebrahimpour","display_name":"Elyas Ebrahimpour","email":"eliasebrahimpour@gmail.com","username":"eliasebrahimpour"},"date":"2023-08-09 11:32:38.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b10c2a56d0e59b8d1454b0c7b103125fa20f2835","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 12:13:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/17a12d171a6c434a88862f7a4b097435\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/544ba7181cde4d69a16f1496245a7fac : SUCCESS in 6m 03s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/95df6142fe1c45629a87a085309b29ed : SUCCESS in 6m 53s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/49f1857c7f054000a4600064d7911b13 : SUCCESS in 3m 33s (non-voting)\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/6f4c52eb790f4b9a92494feb281dc27d : SUCCESS in 4m 40s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/95bbbd9015df4f10b7a8d0657ca5948e : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3ac8ed07f9c4ce3a64014242b0dcdf3 : SUCCESS in 8m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a70ee880c7b442cf8239ace0a6027b91 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6de38a496cf4422aa54032d1ad042f0d : SUCCESS in 10m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/faf02b96746d47fcba745f9391d35601 : SUCCESS in 9m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/660c660209ef4518827eaa0e4dc647d6 : SUCCESS in 6m 44s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/80cee84f557149a6916fab83d55f5ca0 : SUCCESS in 6m 28s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/b02468ac0fcd4704906079f88c7b418c : FAILURE in 34m 23s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/c37d28b09e2648c795599c7ad3ab8dad : FAILURE in 26m 59s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4e8b3f74657345db869ddd023b6e623e : SUCCESS in 4m 44s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/637f42e88205431d93838b8391b204cf : FAILURE in 27m 05s","accounts_in_message":[],"_revision_number":2},{"id":"a6993ca769208a57f2593c0f6dd9fe087537e650","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-08-29 18:40:29.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c31f6cb6b7b2482c4faf4172f582667615040fef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 19:08:31.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c42ae43ef5264bcc9ee6f3dabdf027f9\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/eeee1cf5f9a64489a91959bf82c0d9a8 : FAILURE in 4m 22s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/f5af5836d3824fd6ad811bfbb54763a5 : FAILURE in 7m 04s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/1c9638f9e88f4146a9c2250cab03b91c : SUCCESS in 3m 37s (non-voting)\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/814916b66b224775bfd7b3d2e371dcb2 : FAILURE in 4m 00s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/8d13a22a2bc849deae636af30ef43479 : FAILURE in 6m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0f2d078a5a834f03bbc06177a20e0d7d : FAILURE in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78abb1231fa845078d8e79b50e6358e4 : FAILURE in 2m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e38c8e3f746043f39021708046c67a0e : FAILURE in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9c00fc418b4940eaabeade0c905f97ca : FAILURE in 3m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f779fc90d1df4abe919d5ff3389de870 : FAILURE in 4m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/186e1058ab8f4bb8b9145fbd476c3852 : FAILURE in 4m 13s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/e14d3d039a804727aa2c763ed7ed4227 : FAILURE in 4m 05s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/42d752e4bbea4381acfd7c3ef659a32d : FAILURE in 22m 55s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/1994f041259c4ff1b8a78bb4305b32f3 : FAILURE in 21m 06s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2f8799120f6b4aaaab334742299c7c8f : FAILURE in 3m 58s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/938cbeb4f2af4e34956638f2f53820d3 : FAILURE in 26m 11s","accounts_in_message":[],"_revision_number":2},{"id":"fb1c0b4a039351af50fe11878dda58bdba831de9","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2023-10-12 14:29:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"86627dbfdb9e96b14ac68e5671c5320aadc7edd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 15:08:09.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/66b726f0dadf435c86f6d14a197783cc\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/cbe5d04d463e4f50be40a17f2c23dc94 : SUCCESS in 5m 35s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/6422a0801b68421383e7892868f76ba8 : SUCCESS in 6m 41s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/7e03072fe41149de9452b3f06d897006 : SUCCESS in 5m 11s (non-voting)\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/617850be6fa74b76bcf3b9b0eba6c7fb : SUCCESS in 4m 39s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/588f834a60b14e189c18f99ba946627d : SUCCESS in 7m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b6b06a184f4476295b4431c76d0bca4 : SUCCESS in 10m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fb0c49c8c714dedbdeed9591494366a : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c58967e5b2ff46b59a93b19d049afd61 : SUCCESS in 8m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5b416f7854dd4070a7113c2ed336064a : SUCCESS in 11m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/57a2570bfb744f16b272ea35aad8e3f0 : SUCCESS in 11m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61118c04c10f44d4b6cf367e96f7d4af : SUCCESS in 7m 41s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/abae003c42264368be1c42a29253595e : SUCCESS in 6m 27s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/9b5bd114c6d74f6fb84912da2669659f : FAILURE in 32m 55s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/4d497f09b333486db69cf6428b387aaf : FAILURE in 33m 44s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/109068641fcb427d8657cd3d17790d85 : SUCCESS in 4m 43s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/8822d17efda74e68a2124c2814bfc075 : FAILURE in 24m 40s","accounts_in_message":[],"_revision_number":2},{"id":"c44ed6185df1ce859f631ff12e5278116124ebe2","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-16 01:36:49.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6019ba6459d58e3ccc166df14a8947e385c3a1b7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-16 01:36:59.000000000","message":"Patch Set 2: Code-Review-1","accounts_in_message":[],"_revision_number":2},{"id":"316bad637854423ef4e1de7428e48a6a3bbf8c24","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-10-16 03:59:38.000000000","message":"Patch Set 2: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"faa7edf87fc43ead5c18dae4889591bd0df76897","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2023-10-17 07:06:45.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"46ef7cbbb106e6bdece48df211b1268ce1c4c2c4","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2023-10-17 07:07:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"18442f41d28e3c1204de9b904e43e81ce7fb3125","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-21 17:25:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a23b138a50d1026f92390feda52f348090e73a73","tag":"autogenerated:gerrit:abandon","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2025-05-21 17:25:47.000000000","message":"Abandoned\n\nPatch without activity for multiple years.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c426c2c738a14e5cf386872670eb1534f24a735c","revisions":{"30aaa3d8a011412e39d247ba1a9d08d85264f73e":{"kind":"REWORK","_number":1,"created":"2023-08-09 11:32:05.000000000","uploader":{"_account_id":36274,"name":"Elyas Ebrahimpour","display_name":"Elyas Ebrahimpour","email":"eliasebrahimpour@gmail.com","username":"eliasebrahimpour"},"ref":"refs/changes/07/890907/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/07/890907/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/07/890907/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/07/890907/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/07/890907/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/07/890907/1"}}},"commit":{"parents":[{"commit":"20bdaa386791cc23809b298762217dd1a2b0f071","subject":"Bump minor version of XStatic-jQuery","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/20bdaa386791cc23809b298762217dd1a2b0f071"}]}],"author":{"name":"Elyas Ebrahimpour","email":"eliasebrahimpour@gmail.com","date":"2023-08-09 11:31:18.000000000","tz":210},"committer":{"name":"Elyas Ebrahimpour","email":"eliasebrahimpour@gmail.com","date":"2023-08-09 11:31:18.000000000","tz":210},"subject":"removed duplicate lines","message":"removed duplicate lines\n\n removed duplicate lines in djang middleware and fix mistake\n\nChange-Id: I0fac006d5d2a95807c5a14d60b387b6ccaa749e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/30aaa3d8a011412e39d247ba1a9d08d85264f73e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/30aaa3d8a011412e39d247ba1a9d08d85264f73e"}]},"branch":"refs/heads/master"},"c426c2c738a14e5cf386872670eb1534f24a735c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-08-09 11:32:38.000000000","uploader":{"_account_id":36274,"name":"Elyas Ebrahimpour","display_name":"Elyas Ebrahimpour","email":"eliasebrahimpour@gmail.com","username":"eliasebrahimpour"},"ref":"refs/changes/07/890907/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/07/890907/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/07/890907/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/07/890907/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/07/890907/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/07/890907/2"}}},"commit":{"parents":[{"commit":"20bdaa386791cc23809b298762217dd1a2b0f071","subject":"Bump minor version of XStatic-jQuery","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/20bdaa386791cc23809b298762217dd1a2b0f071"}]}],"author":{"name":"Elyas Ebrahimpour","email":"eliasebrahimpour@gmail.com","date":"2023-08-09 11:31:18.000000000","tz":210},"committer":{"name":"Elyas Ebrahimpour","email":"eliasebrahimpour@gmail.com","date":"2023-08-09 11:32:38.000000000","tz":0},"subject":"removed duplicate lines","message":"removed duplicate lines\n\nremoved duplicate lines in django middleware and fix mistake\n\nChange-Id: I0fac006d5d2a95807c5a14d60b387b6ccaa749e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/c426c2c738a14e5cf386872670eb1534f24a735c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/c426c2c738a14e5cf386872670eb1534f24a735c"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
