)]}'
{"id":"openstack%2Fhorizon~892133","triplet_id":"openstack%2Fhorizon~master~I8255b6082ffdfd4d90a26cc7d06b9049cae27594","project":"openstack/horizon","branch":"master","attention_set":{},"removed_from_attention_set":{"35153":{"account":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"last_update":"2023-09-01 14:02:45.000000000","reason":"Change was submitted"},"29313":{"account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"last_update":"2023-09-01 10:04:28.000000000","reason":"\u003cGERRIT_ACCOUNT_29313\u003e replied on the change","reason_account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}},"6914":{"account":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"last_update":"2023-09-01 14:02:45.000000000","reason":"Change was submitted"},"8648":{"account":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"last_update":"2023-08-28 07:05:35.000000000","reason":"\u003cGERRIT_ACCOUNT_8648\u003e replied on the change","reason_account":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"}},"35133":{"account":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"last_update":"2023-09-01 10:47:43.000000000","reason":"\u003cGERRIT_ACCOUNT_35133\u003e replied on the change","reason_account":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}}},"hashtags":[],"change_id":"I8255b6082ffdfd4d90a26cc7d06b9049cae27594","subject":"pytest-based selenium tests add first image test","status":"MERGED","created":"2023-08-21 11:03:15.000000000","updated":"2023-09-01 14:03:46.000000000","submitted":"2023-09-01 14:02:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":37,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892133","meta_rev_id":"bcb42b1a79a4c77a5ddf85be8f017d9b5929fe27","_number":892133,"virtual_id_number":892133,"owner":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-01 14:02:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":2,"date":"2023-09-01 10:04:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":2,"date":"2023-08-31 11:17:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":1,"date":"2023-09-01 10:04:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"}],"reviewers":{"REVIEWER":[{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-21 11:07:37.000000000","updated_by":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"reviewer":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"state":"REVIEWER"},{"updated":"2023-08-21 11:07:37.000000000","updated_by":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"reviewer":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"state":"REVIEWER"},{"updated":"2023-08-21 11:07:37.000000000","updated_by":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"reviewer":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"state":"REVIEWER"},{"updated":"2023-08-21 12:55:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-28 07:08:50.000000000","updated_by":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"}],"messages":[{"id":"337161c9fe1f292747c54076535e2e93023411c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2023-08-21 11:03:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c689192e22c8dedb97fdc03a7ed4f2779e7dce7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 12:55:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/157c443730374ba4a4dd299054bd38fc\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/2f67ba2caf2f4f3da37a423c1369671f : SUCCESS in 5m 36s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/eb308db194564be1b806f1c06a2f4524 : SUCCESS in 6m 13s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/e37c2b8543d6490eaf51d2abd02526dd : SUCCESS in 3m 42s (non-voting)\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/7e5d7a406aec498190932013786ff91b : SUCCESS in 5m 14s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/23f924ef816c414189c1adbefea94fa1 : SUCCESS in 6m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/00b1960646494dffabd1a7c27b635e2d : SUCCESS in 10m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5328b9fa519248f69d5941f958bdf5ff : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de3c656d3a37434cb1b14acce816ed6b : SUCCESS in 10m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e4c7ced92e754f5cadb865f2bd3b122e : SUCCESS in 11m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e80b735ed9964eee8ccc0d2e40de45dc : FAILURE in 10m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bea06a4e7c24adb92c90cefed08e29d : SUCCESS in 6m 03s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/bd961f26535a4f7ab8478c9580843107 : SUCCESS in 8m 22s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/9da1552ec9f549fdace3d81d88fff374 : SUCCESS in 1h 47m 15s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/94a5cd1a417244ebb965b8106eb77f42 : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":1},{"id":"0a0e937bffee4dd1ac27d422573e9327436f670a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2023-08-22 11:49:21.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"ec183a575abd8b0e89521e5281d8862ec61673b1","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2023-08-22 12:44:10.000000000","message":"Patch Set 2: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":2},{"id":"7c4e47f62757f9ee2f0de7bfcf6bdde8eeab42a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 13:39:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2b1c73cb17741d39715234173b9e54c\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/63957dc21c6145ddb1a7faed9c273e98 : SUCCESS in 5m 40s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/95afbc06d0784572a02a76b191b22b79 : SUCCESS in 6m 27s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/cc66b60ae69f413ba8d4f6107854e7d7 : SUCCESS in 3m 28s (non-voting)\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/93b70c403902445f9f871b43821cf127 : SUCCESS in 4m 27s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/00075aed3b174b38a1ea9a23e10e7751 : SUCCESS in 8m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d187d8df7ed9439c80d97b3ea2226839 : SUCCESS in 8m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f886bf0f61441559188cf1a1a3e2c8b : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b81d6078d57e4b0ebd63e7076290923d : SUCCESS in 10m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7d0ba212361f4dcead1b3d684e3da2a4 : SUCCESS in 10m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/59dacc779e894d509215258f453dd038 : FAILURE in 8m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17d663d33573477395036bbd05cf5485 : SUCCESS in 7m 12s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/7a66b368cfe14b439dcef9e2eaff8caf : SUCCESS in 8m 06s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/ac80ef9fc3dc444ba6b8bea6eaa28710 : SUCCESS in 1h 45m 28s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9d3ee0e701ed452c84f724bd7a6fca1f : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":2},{"id":"cc72f8cab5e094d5e7c6961cd62662266ec68185","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2023-08-22 20:28:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"34d6999c705ad2dd8afc6d38cec5888d7f8e8380","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2023-08-23 08:44:18.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"0bd514e8ab7ec04888c07d956b5eecfd606e56f9","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2023-08-23 08:48:22.000000000","message":"Patch Set 3:\n\n(7 comments)","accounts_in_message":[],"_revision_number":3},{"id":"b442cee2fb4ffc3fdb665ac82dc86233520a08de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 09:42:39.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d2f4456470ad4464a60a05543fe969e5\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/98ec4546bba54995a42f3c8576329c5e : SUCCESS in 5m 55s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/c6eda3ca457f4b1aa8a817e9b482a5ed : SUCCESS in 5m 28s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/8fabe325f72f49fc84eadecbc69a1465 : SUCCESS in 4m 30s (non-voting)\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/6dd73c47d4bf4dd498934202eb072a45 : SUCCESS in 4m 37s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/08c94cc6782d479685a5f537d4b993d9 : SUCCESS in 8m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e26a0175d8174c0085b424157f06aa9a : SUCCESS in 11m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/361212823fbc41579d1e44ba89625920 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc1e8624748a4e75a30963b401b156c1 : SUCCESS in 8m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3ed4992e3f2f4619bdd666a1e6ea8444 : SUCCESS in 9m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d51e2500cdf345f59986a604c0c67e5f : FAILURE in 9m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b2f842156ad46ff93eff85740b3880b : SUCCESS in 8m 01s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/18861619ce4e4c819683db981a11c4a9 : SUCCESS in 5m 59s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/cc9233adc7794572810de7dd5f61490b : FAILURE in 56m 26s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/96394c6f82404f48a3b1f50606e29390 : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":3},{"id":"59fa2dc0940e0f44cb834540320a6552eb2d5d02","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2023-08-23 09:46:32.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c70c7dc6b9853b88789f973a8e6b5ff0342b2859","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2023-08-23 11:12:58.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"c3c96105995ad7410e155d9afb3a50d271fc700a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 11:37:29.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4898a5a18f0842e885ca0a46f09eb45a\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/c62081d625564140a671f5be7b2eaeda : SUCCESS in 4m 45s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/a622347b4b934802bf59d5aed2a74866 : SUCCESS in 5m 50s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/b017a037a1904d4aa9e2b30b862f327e : SUCCESS in 3m 50s (non-voting)\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/04d38457cab24899aa5d08932f1bc9ad : SUCCESS in 3m 55s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/2df742beab7343668e3be5e88a3d588d : SUCCESS in 7m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b9911d629d384d72975035542d18a41c : SUCCESS in 10m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/def8c8d3e3db4d0d8829172b1343002a : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/87897bb14b044782b4288d155a03783c : POST_FAILURE in 11m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d0e86a1233a0483a928db5c850438cb6 : SUCCESS in 10m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/767c8203610449ec97228d6b5715f1f0 : FAILURE in 10m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9247242b65347d4b7aa387a75e5ee4b : SUCCESS in 7m 12s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/045a39da2c8b4b539de6f794fe896a4b : SUCCESS in 6m 27s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/1c0e871cef5745e1be0661e379342c3c : SUCCESS in 1h 45m 42s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e17a847822304e599dc5e4abe5aa3b29 : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":3},{"id":"763760cdabb20135109d673ae71fc3086945c7af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2023-08-23 11:54:40.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":4},{"id":"562354a95cadbe79f89e488e825ae1e19f6a5d6a","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2023-08-23 12:05:55.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"77808c222b42c7a7d89db0049d77b044f2d20cc4","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2023-08-23 12:07:57.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"46fe4b61b0f8f6fd2c3e6a26c3bd3d9f843bfc2a","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2023-08-23 12:45:06.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"5dd733e213894d7d528bc61d59c60cfe0ff5e904","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 12:53:03.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/96373896a56d4e0eab51014eb6496290\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/cd7d83b456aa412ea2454ebf93deef0b : SUCCESS in 5m 55s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/d37ec2294973467e93d7a1b2fb2e2615 : SUCCESS in 6m 31s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/d90077decc784423ae42f5158456a465 : SUCCESS in 4m 06s (non-voting)\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/bdb4d762a5734586a1d5238adfd24fdc : SUCCESS in 4m 59s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/508e455708d845db85517cd2ea8ec436 : SUCCESS in 6m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7bba4baed6f418798fdef99b12f1e5c : SUCCESS in 10m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/860c463e343c444cbf65f8885154a38e : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eff9e083d52c494b9d0045e1c62447d0 : SUCCESS in 9m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4ac0974d4f2549759b3eaa4e9111bb6d : SUCCESS in 9m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/05818346981e4b409f929fffd9435954 : FAILURE in 8m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef08658af4224a54a057bb26855803cd : SUCCESS in 7m 29s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/1751501c73494911804b6bd7d03166d7 : SUCCESS in 6m 45s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/365f03510ad64e2c9cdd74c66ba7bac4 : FAILURE in 57m 16s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4460522eda3b488a979dde519f5c8c93 : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":4},{"id":"a91d4d75ca136029704dd287d88fb66ad968c5ea","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2023-08-23 13:51:46.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"fc7fb4cbae9cfae67c04e98e1e04c1a168e45960","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2023-08-24 09:40:20.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"54323c3c1fc821928ed35f482fa6b0f93e34c4f0","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2023-08-24 09:46:19.000000000","message":"Patch Set 5:\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"5f8e9bf3238936d8af27bf79e933c0d900e9476b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 11:37:56.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/888fcc6481564ae5b241d32bca2e0ce6\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/a147063adecd4b6d8b6049f955086498 : SUCCESS in 5m 47s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/35876a9a2a884240a67500557e73a641 : SUCCESS in 5m 28s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/406631b4b1454190bd2f90ecd25485c2 : SUCCESS in 4m 26s (non-voting)\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/bb311af3fa264f65877395fc8932ed2e : SUCCESS in 7m 22s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/7d044ef0694549158156ee199fc874be : SUCCESS in 6m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/750761a5a37644af92e0520109b87f24 : SUCCESS in 10m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b44d8cb8cf141dfbfed852d7a5edbb9 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f84a17e4dc9e4b03a2922506b992c2c8 : SUCCESS in 9m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ea61b766e957447a8a4e4a1f76f8302a : SUCCESS in 11m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a2de2780d0fc4b6dbd0ec157088d8a09 : FAILURE in 10m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4ea1cf0d8824f888041f3aaa1568082 : SUCCESS in 8m 09s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/6e0790aa12ec415096b20a4c2ee620a9 : SUCCESS in 6m 10s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/ff0cd22fb0b745e092effdd555e86ee2 : SUCCESS in 1h 52m 31s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/23d2a25780474ee0ac94d59d20a76005 : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":5},{"id":"6560162bcb82a6170a7e684aad9da7760c7ed5c1","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2023-08-28 07:05:35.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"284e4d8f7c28a3852bf62a77b9fb9413b8fdb2c2","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-08-29 16:42:11.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"33a0f51dfb48455272f8483a0ef49c66bc2ee231","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 16:42:40.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"0313ceb0e1972c0da721f2d6f95da0bbf782e499","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 18:34:49.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f4902539e8fc48bfb1ccd6c8a9d7a5ca\n\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/d838c2ea87344d09b90e6420a94f434e : SUCCESS in 9m 53s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/8d25dc7f560149afa39101a211bf569d : SUCCESS in 12m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/96d358516f4b4a589fd1649028828b80 : SUCCESS in 10m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e6a0b2ba362447582f3c137931631ee : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba3bb92da9a0462295074465e7a3b832 : SUCCESS in 10m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3649e112058c4bdfbd41add1807ccde9 : SUCCESS in 9m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51cf71ff5814489f87523abcc06cb832 : FAILURE in 5m 30s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/a34c1866311d4d94a437873764bed797 : SUCCESS in 6m 43s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/2259ad3ab0444348bf28b13e71f6b018 : SUCCESS in 1h 27m 09s","accounts_in_message":[],"_revision_number":5},{"id":"31806b276ea3ec8d9001056fa3a0bd74cc11597a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2023-08-31 11:17:11.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":6},{"id":"b4b60892f0d41c55c154a0d60e3e2449235d380f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 11:55:00.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/75a468734c114b05913c00e8d4e4bf99\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/3aa112bb19414fbba79a413d8f19790c : SUCCESS in 5m 15s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/63376b1a9b2e4518a77efed25a6c8003 : SUCCESS in 7m 39s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/46822cfc17184d61aae50ea27f27b393 : SUCCESS in 3m 23s (non-voting)\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/9f45cfe23be0414687ed85d1fd94689c : SUCCESS in 4m 27s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/abb8e069d9f54e7aad2de51979613e71 : SUCCESS in 7m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/73c810398b3b4aa1823d18819bb7af8f : SUCCESS in 10m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/123300688e6a4430870bb9125aad363a : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e2f26d8b99244c5a85d2de04ef869f91 : SUCCESS in 10m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c19c47202a11496fbb98cdbc3dfde12a : SUCCESS in 11m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5d6356a7de00410ebb29825a1364fe6d : FAILURE in 10m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7e3c2f08ab94c77bff90a79238a7312 : SUCCESS in 6m 40s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/ffac54104a414c2d9b25578a884963ca : SUCCESS in 5m 56s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/8381f3460cd14f768ad5e9c552e1c15f : FAILURE in 32m 37s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/70b4623af466426b99bfc961ea941efa : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":6},{"id":"335e4ca00a72aff3f95e9accedd460d444bad487","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2023-08-31 14:11:43.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"fc52b5b58e6ce650f014bdb481f4dca9fafb131b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 14:39:02.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1df0006dfeef42a7894f65f25e06ea29\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/095c1f7e68894ec8b13774179a8e362d : SUCCESS in 5m 38s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/6607cfe12c844d8da618a10a4c92206c : SUCCESS in 5m 48s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/932940e4694e413c80e781de2ba1991b : SUCCESS in 3m 34s (non-voting)\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/be020feecdc04c99959333a65a11f7fe : SUCCESS in 5m 07s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/8193b6d2d2de4e72bdba1981ef7a03e7 : SUCCESS in 8m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aaa6fd390c0a4b76b47fff294d27ab70 : SUCCESS in 11m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0de08e120e76419583467647c9752b04 : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/681fdd230a55461a86ae7c042c77e118 : SUCCESS in 11m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/90462e56eb8a4ff5a10c84b17e59f589 : SUCCESS in 10m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d4c4c6c0f83548d5a1afa5d321321ecd : FAILURE in 8m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f68f774d3d348978b1615ac2fdc8680 : SUCCESS in 7m 29s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/10f859ead9b14878959eb96645954c52 : SUCCESS in 5m 48s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/eb733172ba1546ccbf43b3ddfca76dc1 : FAILURE in 26m 34s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/91ed312f65e8414e8f9f5623e28af85d : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":6},{"id":"1954e41dd45eb49a82b44494c96a3d6c3252ebe5","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2023-08-31 20:41:21.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"8bf9b15148dc6b6170611d614f7b33b236563575","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 21:52:23.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f58e7d056dc745d99ffbdce50a8637d5\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/8cb6e05fd3ae49cd9f77f99a70408fd6 : SUCCESS in 4m 44s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/e1071f873a74422f85fe29c23b779b33 : SUCCESS in 7m 44s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/0a7caabd8e304ae5938f4a1f0b77081b : SUCCESS in 4m 51s (non-voting)\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/ea84c6463bbd4f08b6c0c2d9fca0c2a0 : SUCCESS in 7m 59s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/b10d07a2a00a490cb79589abb16af690 : SUCCESS in 12m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c1c3dad9bf224448952047cc22f0c3fc : SUCCESS in 10m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/babb5d503ec04866a5e253fb86fbd62b : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11d3abb546a643d0b8d699faef08d3f8 : SUCCESS in 10m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1412d2ec140e43a9818f95210552c66b : SUCCESS in 10m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/09c20c846fa0458d9b867f257ea88825 : FAILURE in 11m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76ab2803336142ec9a7b6435c877d5b5 : SUCCESS in 7m 50s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/45e5946504da4350a561318545d7cc61 : SUCCESS in 7m 07s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/98785ab2704a4b1ab295e6fec56c6c71 : FAILURE in 54m 57s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/af6423d063f04d3594887a8bfd342607 : SUCCESS in 6m 04s","accounts_in_message":[],"_revision_number":6},{"id":"670cf39ceef2de4b0cc3579153434d2467e2a512","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2023-09-01 06:56:12.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"41fcf1f338ee6a89a2fef9bed4c5c2bccb609ad3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 07:48:55.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cfe0204396444a7a93b0a0fa4f4dfd81\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/b73cc7965ebe40b8afa2416e9306775f : SUCCESS in 5m 44s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/2a397d73a7c94bceaec0d1cd41c86c2e : SUCCESS in 7m 37s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/cda9e9c9053a4512adee4fbf7eed2d2d : SUCCESS in 3m 44s (non-voting)\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/856a701275b64b7aaa07d048287431ff : SUCCESS in 5m 43s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/85856336cc57487da5551c7682f50529 : SUCCESS in 8m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74ac0fb24f884b619a7a0d0b462bd7cd : SUCCESS in 10m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4b088c3410344b5b1fd212e57d45f9a : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/afc57570571a4518bdfb3b43dde141fb : SUCCESS in 10m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7ea59aad50324ad582e28a59f50f1532 : SUCCESS in 8m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/06b6494d8a0a4923b6da62b6e34f0838 : FAILURE in 10m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b74bbcef2b0d47b491cb4f74ae7e2d80 : SUCCESS in 6m 34s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/eda55c8fb4d742f38a15207c09d3e1f6 : SUCCESS in 5m 58s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/99b20577cc7946618c60e9ff79d5e114 : FAILURE in 46m 51s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/584a546cf9c3408ea54f610ab417131f : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":6},{"id":"8dc6c5c7418cc4220d4a04c854c540680b499bef","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2023-09-01 08:04:06.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"93b3a23d09598576d9c6591d8813907fb9b0bc30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 09:46:53.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/574afdd7e0ac471c86e0c6a8b8ae0199\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/e93ee68b00cf484c997f16c31af7c4dc : SUCCESS in 5m 35s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/b2e2b069330d4d009042ec7842857e2e : SUCCESS in 5m 35s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/40a5f3c6d94944fba07624b8c8cebb72 : SUCCESS in 3m 42s (non-voting)\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/020995e0145c4c7fbcd685c9bb8e8bb2 : SUCCESS in 4m 33s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/c80199db843c49c59888f5945159ac38 : SUCCESS in 8m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f646f01f39384544ada4ce0737543f22 : SUCCESS in 10m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58aa9cb2e4de4fcdb4dc73c00189f411 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb35414814aa43d79b88369c14351d5e : SUCCESS in 10m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b806a04e1ff244f59de601d31d98f672 : SUCCESS in 9m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c39e34ca50ee40be91c2c4d3e6d5d37f : FAILURE in 10m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a46e09a58fb429396aef0726db9cc75 : SUCCESS in 7m 30s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/a96c1064a0a245d7b8f7f798dd57e4ae : SUCCESS in 6m 45s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/466367d47cda4da5ad778a01fbf9b236 : SUCCESS in 1h 42m 01s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/da247f9d78c34466b6f270e32351d7e4 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":6},{"id":"b9ba738514f2e34775bacbc0450d1ac5633655ac","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-09-01 10:04:28.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"29b961609ee779cb519319572195f4b6bd797dce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 10:04:56.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"bcc4debdbf25e60d4c74d34196d54ff8d92a0a58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 10:44:45.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/de993d2f9cc44ad6b5c514675b546121\n\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/445abbd1716b4001bbaa154cd1e1dbd6 : SUCCESS in 4m 18s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/ce85b445fee74b5692d15a84d0c8f8dd : SUCCESS in 6m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c2ef21ca88f4de2a7906840b9091be8 : SUCCESS in 7m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecedcdf6d2164f22bfcb27736231675a : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1562a035f6804a73b2a7b06628cc382f : SUCCESS in 7m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8a741726d20544e7aad99968ac96cbda : SUCCESS in 11m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efdb832b45a249218d9782144aad94d4 : SUCCESS in 6m 15s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/af1c3e77cef14b28820bf1515a62a99c : SUCCESS in 6m 08s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/f980e3bc3cc84774a2a19082fe2be438 : FAILURE in 39m 25s","accounts_in_message":[],"_revision_number":6},{"id":"97642d080d8cf6f621be3dda8a4d93137a6cbf48","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2023-09-01 10:47:43.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"11fbdc3d1c2857d257a0c407bb4d120306957ea1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 12:16:55.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0833b3a687cc41548226a5d23fdd109d\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/152a790e6fb94e06b84a647eb5fbcaf1 : SUCCESS in 6m 19s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/eb4ec3d7a0624e34bd910b06e4d6a5f4 : SUCCESS in 6m 46s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/0f509845e41648c5afb141312fcb55da : SUCCESS in 3m 41s (non-voting)\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/dba556647a644309aa7bf4df97b51b58 : SUCCESS in 4m 44s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/4d02debf2f344d24bb304b870a222c03 : SUCCESS in 6m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/047a3becb4a9474faed73778bd3d71a0 : SUCCESS in 10m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86a59a9741c54c05b00f46c753edd4f2 : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5cce4d1653944c908592e8e5db218dad : SUCCESS in 10m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d079d5379d164dcdb4c974d4397eb79b : SUCCESS in 10m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/79071f3b43714ca5a78623514d7ba0c9 : FAILURE in 10m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39932fb11dbc401599ddb2550ec8ed91 : SUCCESS in 7m 21s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/8c17d9b33b67427090250782611d45be : SUCCESS in 5m 54s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/dee787a6d73848009311c8d103519f8b : SUCCESS in 1h 28m 10s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/03e9ce4c10f942b8b3075e580c84e04e : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":6},{"id":"ebe682fea527e1ef8ed6c0992544d523b74fd6be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 12:17:27.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"d6a74e21c3ef990841903fe35d70c8d103ef33b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 14:02:44.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/205788f2f8a740d4bf18903863bf2901\n\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/b9feef72f24548c487416a8927f89c31 : SUCCESS in 5m 10s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/41813c55367146eaa51c8218b1d9298c : SUCCESS in 6m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a46a114405143ce904db400fd6f03ac : SUCCESS in 12m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7939c8b7127646da97de20b3e9a9df8e : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eeb91579eb6c4c50af601d14df4593e8 : SUCCESS in 11m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0dd7cf80a92f435680730fa1bb3ec6cf : SUCCESS in 10m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d345961e41fe46b98f3fceec2a31a945 : SUCCESS in 7m 51s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/860d8702a7414564893c5096a00e31ee : SUCCESS in 9m 08s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/4e97b010366743bba119f502d84e4a36 : SUCCESS in 1h 26m 26s","accounts_in_message":[],"_revision_number":6},{"id":"109b5a043db8697b3d76c4aad0ffd8b16c582f0c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 14:02:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"bcb42b1a79a4c77a5ddf85be8f017d9b5929fe27","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 14:03:46.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da14ec4a8c704a3c9b6bede2260d30b7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ac21af514854e69950b9af77c13d79f : SUCCESS in 39s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"9c59a3dd9879dce540e61b575f306756ff994bd3","revisions":{"dbb029688eddc484bcdb9d0da5f4d8b9c465b2b3":{"kind":"REWORK","_number":1,"created":"2023-08-21 11:03:15.000000000","uploader":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"ref":"refs/changes/33/892133/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/33/892133/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/33/892133/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/33/892133/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/33/892133/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/33/892133/1"}}},"commit":{"parents":[{"commit":"6981184a15b6f230f2e01c7503e5b4fce6923ce2","subject":"pytest-based selenium tests fix conf. variables and import fixtures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/6981184a15b6f230f2e01c7503e5b4fce6923ce2"}]}],"author":{"name":"Jan Jasek","email":"jasekjan96@gmail.com","date":"2023-08-21 10:59:58.000000000","tz":120},"committer":{"name":"Jan Jasek","email":"jasekjan96@gmail.com","date":"2023-08-21 11:03:11.000000000","tz":120},"subject":"pytest-based selenium tests add first image test","message":"pytest-based selenium tests add first image test\n\ntest_image_create_from_local_file_demo\n\nChange-Id: I8255b6082ffdfd4d90a26cc7d06b9049cae27594\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/dbb029688eddc484bcdb9d0da5f4d8b9c465b2b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/dbb029688eddc484bcdb9d0da5f4d8b9c465b2b3"}]},"branch":"refs/heads/master"},"8abbe7bff6f2bc79cd5e753fc2ed8b2db6905985":{"kind":"REWORK","_number":2,"created":"2023-08-22 11:49:21.000000000","uploader":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"ref":"refs/changes/33/892133/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/33/892133/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/33/892133/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/33/892133/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/33/892133/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/33/892133/2"}}},"commit":{"parents":[{"commit":"6981184a15b6f230f2e01c7503e5b4fce6923ce2","subject":"pytest-based selenium tests fix conf. variables and import fixtures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/6981184a15b6f230f2e01c7503e5b4fce6923ce2"}]}],"author":{"name":"Jan Jasek","email":"jasekjan96@gmail.com","date":"2023-08-21 10:59:58.000000000","tz":120},"committer":{"name":"Jan Jasek","email":"jasekjan96@gmail.com","date":"2023-08-22 11:48:16.000000000","tz":120},"subject":"pytest-based selenium tests add first image test","message":"pytest-based selenium tests add first image test\n\ntest_image_create_from_local_file_demo\n\nChange-Id: I8255b6082ffdfd4d90a26cc7d06b9049cae27594\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/8abbe7bff6f2bc79cd5e753fc2ed8b2db6905985"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/8abbe7bff6f2bc79cd5e753fc2ed8b2db6905985"}]},"branch":"refs/heads/master"},"2a360f60716c37ed9e3168302ca7ce7c66e1538c":{"kind":"REWORK","_number":3,"created":"2023-08-23 08:44:18.000000000","uploader":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"ref":"refs/changes/33/892133/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/33/892133/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/33/892133/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/33/892133/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/33/892133/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/33/892133/3"}}},"commit":{"parents":[{"commit":"6981184a15b6f230f2e01c7503e5b4fce6923ce2","subject":"pytest-based selenium tests fix conf. variables and import fixtures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/6981184a15b6f230f2e01c7503e5b4fce6923ce2"}]}],"author":{"name":"Jan Jasek","email":"jasekjan96@gmail.com","date":"2023-08-21 10:59:58.000000000","tz":120},"committer":{"name":"Jan Jasek","email":"jasekjan96@gmail.com","date":"2023-08-23 08:42:05.000000000","tz":120},"subject":"pytest-based selenium tests add first image test","message":"pytest-based selenium tests add first image test\n\ntest_image_create_from_local_file_demo\n\nChange-Id: I8255b6082ffdfd4d90a26cc7d06b9049cae27594\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/2a360f60716c37ed9e3168302ca7ce7c66e1538c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/2a360f60716c37ed9e3168302ca7ce7c66e1538c"}]},"branch":"refs/heads/master"},"587c8c1ac8e33ffd8ad61041a032b173535aaedc":{"kind":"REWORK","_number":4,"created":"2023-08-23 11:54:40.000000000","uploader":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"ref":"refs/changes/33/892133/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/33/892133/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/33/892133/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/33/892133/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/33/892133/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/33/892133/4"}}},"commit":{"parents":[{"commit":"6981184a15b6f230f2e01c7503e5b4fce6923ce2","subject":"pytest-based selenium tests fix conf. variables and import fixtures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/6981184a15b6f230f2e01c7503e5b4fce6923ce2"}]}],"author":{"name":"Jan Jasek","email":"jasekjan96@gmail.com","date":"2023-08-21 10:59:58.000000000","tz":120},"committer":{"name":"Jan Jasek","email":"jasekjan96@gmail.com","date":"2023-08-23 11:52:49.000000000","tz":120},"subject":"pytest-based selenium tests add first image test","message":"pytest-based selenium tests add first image test\n\ntest_image_create_from_local_file_demo\n\nChange-Id: I8255b6082ffdfd4d90a26cc7d06b9049cae27594\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/587c8c1ac8e33ffd8ad61041a032b173535aaedc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/587c8c1ac8e33ffd8ad61041a032b173535aaedc"}]},"branch":"refs/heads/master"},"1a666ec621f29effcb318a02f5dae7561322ec50":{"kind":"REWORK","_number":5,"created":"2023-08-24 09:40:20.000000000","uploader":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"ref":"refs/changes/33/892133/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/33/892133/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/33/892133/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/33/892133/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/33/892133/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/33/892133/5"}}},"commit":{"parents":[{"commit":"6981184a15b6f230f2e01c7503e5b4fce6923ce2","subject":"pytest-based selenium tests fix conf. variables and import fixtures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/6981184a15b6f230f2e01c7503e5b4fce6923ce2"}]}],"author":{"name":"Jan Jasek","email":"jasekjan96@gmail.com","date":"2023-08-21 10:59:58.000000000","tz":120},"committer":{"name":"Jan Jasek","email":"jasekjan96@gmail.com","date":"2023-08-24 09:38:30.000000000","tz":120},"subject":"pytest-based selenium tests add first image test","message":"pytest-based selenium tests add first image test\n\ntest_image_create_from_local_file_demo\n\nChange-Id: I8255b6082ffdfd4d90a26cc7d06b9049cae27594\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/1a666ec621f29effcb318a02f5dae7561322ec50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/1a666ec621f29effcb318a02f5dae7561322ec50"}]},"branch":"refs/heads/master"},"9c59a3dd9879dce540e61b575f306756ff994bd3":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-08-31 11:17:11.000000000","uploader":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"ref":"refs/changes/33/892133/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/33/892133/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/33/892133/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/33/892133/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/33/892133/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/33/892133/6"}}},"commit":{"parents":[{"commit":"80f1e309028d266b60b33324e7e74675279a7131","subject":"Fix sphinx extlinks captions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/80f1e309028d266b60b33324e7e74675279a7131"}]}],"author":{"name":"Jan Jasek","email":"jasekjan96@gmail.com","date":"2023-08-21 10:59:58.000000000","tz":120},"committer":{"name":"Jan Jasek","email":"jjasek@redhat.com","date":"2023-08-31 11:17:11.000000000","tz":0},"subject":"pytest-based selenium tests add first image test","message":"pytest-based selenium tests add first image test\n\ntest_image_create_from_local_file_demo\n\nChange-Id: I8255b6082ffdfd4d90a26cc7d06b9049cae27594\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/9c59a3dd9879dce540e61b575f306756ff994bd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/9c59a3dd9879dce540e61b575f306756ff994bd3"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
