)]}'
{"id":"openstack%2Fhorizon~897477","triplet_id":"openstack%2Fhorizon~master~Ia7800712ce8d5a5c560fabfaa59222e5c9804e82","project":"openstack/horizon","branch":"master","topic":"fix-translate-ptbr","attention_set":{},"removed_from_attention_set":{"36419":{"account":{"_account_id":36419,"name":"Kássio Augusto de Moura Silva","display_name":"Kássio Silva","email":"kassio.silva@lsd.ufcg.edu.br","username":"kassiosilva"},"last_update":"2023-10-10 14:24:50.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ia7800712ce8d5a5c560fabfaa59222e5c9804e82","subject":"\"Updating Brazilian Portuguese translation, some corrections to the text\"","status":"ABANDONED","created":"2023-10-05 16:32:55.000000000","updated":"2023-10-10 14:24:50.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0e1423cdf0dd3f8f94abdf82e5a5fe6f47721eea","_number":897477,"virtual_id_number":897477,"owner":{"_account_id":36419,"name":"Kássio Augusto de Moura Silva","display_name":"Kássio Silva","email":"kassio.silva@lsd.ufcg.edu.br","username":"kassiosilva"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-10-05 16:48:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":-1,"date":"2023-10-06 02:58:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-05 16:48:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-06 02:58:59.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"}],"messages":[{"id":"efd23c239bb6df15c9235c9bbd230e35e42de178","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36419,"name":"Kássio Augusto de Moura Silva","display_name":"Kássio Silva","email":"kassio.silva@lsd.ufcg.edu.br","username":"kassiosilva"},"date":"2023-10-05 16:32:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"99d9592d9db949a657c5dafd4d7848e647cad361","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 16:48:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7507344bfca54d6cb8ab5475170c1323\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/3f1fcec01e454d22a9138536ddc0f181 : SUCCESS in 5m 51s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/a6c76c8740d74c3d988e01795581c3e2 : SUCCESS in 6m 10s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/f465d180e2734f1daa051a50ffa434e0 : SUCCESS in 3m 35s (non-voting)\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/a0342b99c35441eda903db4ae51aac9d : SUCCESS in 5m 03s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/8c5c7107320c4d10b95f13293fc92401 : SUCCESS in 6m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1409a88c778b4542a6b47395d22d869e : SUCCESS in 10m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d6fc12f02e34e5fb31b3f346b00c412 : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/522c2ce2bb6345efb3ca0293ab009adc : SUCCESS in 10m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d471664a05794754aba0e6cfd7b060fa : SUCCESS in 10m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0c04f2461cda467db2fec0d1f306ac23 : SUCCESS in 10m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/635aa1d4fc3e4a7b947df314a16bfe9d : SUCCESS in 8m 32s","accounts_in_message":[],"_revision_number":1},{"id":"e6bc47977c0d8401ca9dd653e5be50b02c580715","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2023-10-06 02:58:59.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0e1423cdf0dd3f8f94abdf82e5a5fe6f47721eea","tag":"autogenerated:gerrit:abandon","author":{"_account_id":36419,"name":"Kássio Augusto de Moura Silva","display_name":"Kássio Silva","email":"kassio.silva@lsd.ufcg.edu.br","username":"kassiosilva"},"date":"2023-10-10 14:24:50.000000000","message":"Abandoned\n\nThis patch is no longer necessary","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"647ce511f7978579048c3ba72d0030b73a8c66d9","revisions":{"647ce511f7978579048c3ba72d0030b73a8c66d9":{"kind":"REWORK","_number":1,"created":"2023-10-05 16:32:55.000000000","uploader":{"_account_id":36419,"name":"Kássio Augusto de Moura Silva","display_name":"Kássio Silva","email":"kassio.silva@lsd.ufcg.edu.br","username":"kassiosilva"},"ref":"refs/changes/77/897477/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/77/897477/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/77/897477/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/77/897477/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/77/897477/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/77/897477/1"}}},"commit":{"parents":[{"commit":"12dbdf93ac8b53b4c8f08228d218c3946b113683","subject":"Merge \"pytest-based selenium tests add routers create, delete tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/12dbdf93ac8b53b4c8f08228d218c3946b113683"}]}],"author":{"name":"Kássio","email":"kassio.silva@lsd.ufcg.edu.br","date":"2023-10-05 16:30:36.000000000","tz":-180},"committer":{"name":"Kássio","email":"kassio.silva@lsd.ufcg.edu.br","date":"2023-10-05 16:30:36.000000000","tz":-180},"subject":"\"Updating Brazilian Portuguese translation, some corrections to the text\"","message":"\"Updating Brazilian Portuguese translation, some corrections to the text\"\n\nChange-Id: Ia7800712ce8d5a5c560fabfaa59222e5c9804e82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/647ce511f7978579048c3ba72d0030b73a8c66d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/647ce511f7978579048c3ba72d0030b73a8c66d9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
