)]}'
{"id":"openstack%2Fhorizon~903370","triplet_id":"openstack%2Fhorizon~master~Ic577a94514881e56d32087a4fc4d6a06414ff568","project":"openstack/horizon","branch":"master","topic":"test_volumes","attention_set":{},"removed_from_attention_set":{"29313":{"account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"last_update":"2024-01-18 15:43:24.000000000","reason":"Change was submitted"},"35153":{"account":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"last_update":"2024-01-18 15:43:24.000000000","reason":"Change was submitted"},"6914":{"account":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"last_update":"2024-01-17 16:49:32.000000000","reason":"\u003cGERRIT_ACCOUNT_6914\u003e replied on the change","reason_account":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"}},"8648":{"account":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"last_update":"2024-01-08 13:58:43.000000000","reason":"\u003cGERRIT_ACCOUNT_8648\u003e replied on the change","reason_account":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"}},"35133":{"account":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"last_update":"2024-01-18 11:54:09.000000000","reason":"\u003cGERRIT_ACCOUNT_35133\u003e replied on the change","reason_account":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}}},"hashtags":[],"change_id":"Ic577a94514881e56d32087a4fc4d6a06414ff568","subject":"pytest-based selenium tests add volume actions tests","status":"MERGED","created":"2023-12-11 18:01:16.000000000","updated":"2024-01-18 15:44:33.000000000","submitted":"2024-01-18 15:43:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"903370-test_volumes","meta_rev_id":"7e9b618b64728f53458e3d9ad90cf36a7fb6086e","_number":903370,"virtual_id_number":903370,"owner":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-18 15:43:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"all":[{"value":2,"date":"2024-01-17 16:49:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":2,"date":"2024-01-08 13:58:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"all":[{"value":1,"date":"2024-01-17 16:49:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"}],"reviewers":{"REVIEWER":[{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-11 18:44:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-12 08:03:10.000000000","updated_by":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"reviewer":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"state":"REVIEWER"},{"updated":"2023-12-12 08:03:10.000000000","updated_by":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"reviewer":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"state":"REVIEWER"},{"updated":"2023-12-12 08:03:10.000000000","updated_by":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2023-12-12 08:03:10.000000000","updated_by":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"reviewer":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"state":"REVIEWER"}],"messages":[{"id":"46ef670266c5b7f26b7eb8a17f560449f3b67e5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2023-12-11 18:01:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dcf4fc1da87663649cc58f7d9a9228d207c980e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-11 18:44:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/addffeaf5c024689a77ce1ac2d7e8246\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/6f2dd3a8676a40479bf3c05349412748 : SUCCESS in 5m 20s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/7b06b0e4fca64c57ae892c965a8d46a6 : SUCCESS in 5m 34s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/67fb963e9aad4014aa254fbe0a8d840c : SUCCESS in 3m 26s (non-voting)\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/e03184bbda3f44969d9e77b31a76d737 : SUCCESS in 4m 16s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/48705ff97b784e77bd5bb66913e1351a : SUCCESS in 5m 49s\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/6a37afc734c24adb9c40b76cf83c1c85 : SUCCESS in 4m 47s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/73c7e523518a4b438c98ad15d4a2d16e : SUCCESS in 5m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/31ccc3a57fcb4f20b58f0fbc4b778ab8 : SUCCESS in 12m 57s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/1ef06e9ee3ae473e8c28c791273e96eb : SUCCESS in 9m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ab7fca48a74445b845ba0491a316467 : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c50b0f871f0140f6a56f6a305908a76c : SUCCESS in 8m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4eee45e05a5c4acf863f2358a62a3f1a : SUCCESS in 7m 06s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/87b608a2419c4871b1a11b5462d2fcf3 : SUCCESS in 8m 07s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/c16d9b7a30a64c919c3913f337f07ba4 : FAILURE in 42m 20s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/91ff98a0b87a4e169eaba69246ae5958 : SUCCESS in 34m 31s (non-voting)\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a22c5c362a9c4e73a646801ece68c0cd : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":1},{"id":"883e0a6293d6e460f87a86f8f3ac0f1f398d6229","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2023-12-11 23:06:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ae4df213a70d4645851b9420da0e71010bb7a0ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 00:13:54.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d32e1bc0f2b348a3a47ef97898cb3d34\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/760a2f690e8b4dd18b493e0b7cc36e0d : SUCCESS in 7m 19s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/94c08ed2210e4de7b144140fa39e010f : SUCCESS in 8m 52s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/8b12646d87e3486989eb76657113c48c : SUCCESS in 5m 55s (non-voting)\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/b70a94b0f07243d49455934a0535d386 : SUCCESS in 7m 01s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/913cbce50ade4a70b64c3d83aee52293 : SUCCESS in 7m 49s\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/208a918b3c54412298a2d94a3fbfc5f3 : SUCCESS in 5m 55s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/a664aafdd688464fb6ddb2bbd2c1fb37 : SUCCESS in 8m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/25f1428f1b3b413db327be00c3b35a8d : SUCCESS in 11m 10s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/6e4b27df66a047d4be9c441f99d35423 : SUCCESS in 13m 11s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b58e40ad81fa40408266dcfa184b6dde : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/deb7d0a8c3fb4e6db8177b33e906f9df : SUCCESS in 7m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2a32b822c364ee1a3fcbad349a65094 : SUCCESS in 9m 07s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/5267b816abb146f58663b9d4daea47ca : SUCCESS in 9m 06s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/9a3d0bcaa9584681a34d82c356253a79 : FAILURE in 1h 06m 29s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/88fe2da33068468eadc9256e27d9de72 : SUCCESS in 50m 56s (non-voting)\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/3e82baa55157439798b92fad6decc64a : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":1},{"id":"3fc481e2a6f58fff3dfd8531484b24f4832522b1","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2023-12-12 05:56:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"48d349a7f4c951ccca0a53c7c56c9251c5d05f63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 07:24:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d23e624683b40fda57ed5c3d511e752\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/7e014c6a6beb4033b02e53bd68ebe841 : SUCCESS in 6m 57s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/9bffa97cde5b48778632b5e4cdeed109 : SUCCESS in 8m 54s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/12ae468147854946a96ae6edf0c3a181 : SUCCESS in 4m 32s (non-voting)\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/d150651cda1843d79a20f4ed126d6e57 : SUCCESS in 6m 14s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/c34b246b688c4dde9412b5ccb394f052 : SUCCESS in 6m 17s\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/fb8b43887ee141eaa573603c9f13d981 : SUCCESS in 4m 48s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/7d7a7cc2751849d0a4df1862443b611a : SUCCESS in 7m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0a1e561365e44374a126ee757cb5827a : SUCCESS in 9m 37s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/99900404381c432ba838663d27bbea01 : SUCCESS in 11m 03s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fd732c04dfb42ada473376473f62d61 : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac11fb09a65d4f6697eee57a4de97173 : SUCCESS in 8m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dd8186f738d4694840271a205fb0c6f : SUCCESS in 8m 27s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/f0efc8cf64a44f6a8f59c7241507ef44 : SUCCESS in 6m 54s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/cc652adc99ae4c81868610420ac8bf17 : SUCCESS in 1h 22m 37s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/cb43e1800f06452a9de188521d13da58 : SUCCESS in 44m 45s (non-voting)\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bd9eeb8fb4394ced8d14e822c38ed9b7 : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":1},{"id":"2da012e11228c02d8d2f59aace869b115ebd21f7","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2024-01-04 11:40:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"48cac8e95e92f5e290093c22075768270e4983b0","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2024-01-04 18:55:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9634124beb77136fa9c7410cdb1dc5ae8597c0f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-04 18:55:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"72fdbbe98996762c113757670afa3b2ddbc50034","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-04 20:41:55.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/82d3d15b3f654825be1fd9339d9ebf51\n\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/69d691d040804415a02974ca8cd6b9fa : SUCCESS in 6m 12s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/69bddb3760414f42b145d4acb02c74b4 : SUCCESS in 8m 08s\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/048e942da08744608c02994a01feabaa : SUCCESS in 6m 15s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/c8f114090970411ab560e8b41533f58c : SUCCESS in 6m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bdd942a6dd454d71898fce2ac07be468 : SUCCESS in 8m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6652d484d42e463bb352b2b0322b0a77 : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59c32289a4bd48c6a538206329d9fca7 : SUCCESS in 9m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f52f31ea7b7b41db8f04523fbc3825e8 : SUCCESS in 6m 38s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/8fb49712da1747babe4bd98a180b2687 : SUCCESS in 6m 38s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/aaea7eda6c4c44018d6a5afdb6d5b640 : FAILURE in 1h 03m 05s","accounts_in_message":[],"_revision_number":1},{"id":"5c5ba55f4633f91114db33726701c4482068d81e","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2024-01-05 07:59:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"afd38384461b1ac50b0c2a697983a241738cb3a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 09:01:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/22efe13a5717472b92b2ff8775703004\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/d277ad2ac1874987bc98bf531a3dca1e : SUCCESS in 5m 37s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/4dbdb9eebe53459fb910ad9aa573c38b : SUCCESS in 5m 55s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/c85533f107ad4e198f1c3d91b5be678f : SUCCESS in 3m 43s (non-voting)\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/bc1b029e382d4088bcc599edc54d1986 : SUCCESS in 4m 36s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/9141cca175b943c795b177bc7c9d789f : SUCCESS in 8m 14s\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/b67a12626a0c4fd2947eb5b28c402e83 : SUCCESS in 5m 49s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/860e23b02064485a811f1d26d1c9c4f8 : SUCCESS in 6m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8d37a949b69b4049ab2f9dd85097ba60 : SUCCESS in 9m 47s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/e52b99dc7573433380995ddd2abab21d : SUCCESS in 13m 25s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebad634e9c5b4e6388a2061c78ef323c : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/00334f71be58476c94c709584816514a : SUCCESS in 9m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9a2dc29ec9d43ee88d79cb624552fd2 : SUCCESS in 7m 40s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/d0f4bc0bd3b2416f9e7e6d530358dcaa : SUCCESS in 6m 27s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/900a0e54edb441d9b2d39ffaf628cc69 : FAILURE in 37m 29s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/86f2c3580df34d13928a06a9aa05a08a : SUCCESS in 55m 19s (non-voting)\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/84b963649df54d868d5960db43cafcbb : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":1},{"id":"57f5534067f1f6d0e6ff34d6f0c9ef509132e978","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2024-01-05 09:33:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"267f495fb9e388a4131e641971a267bbaa21b31e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 10:37:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fad4d5f440cb4a22b291d07c022316fb\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/5d942585ed90458fa316105569b1332c : SUCCESS in 6m 24s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/43be56fd56aa4667b388747589ce370a : SUCCESS in 5m 48s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/aead410e536e43698e2a9fed7e7a9816 : SUCCESS in 3m 40s (non-voting)\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/6bbee1997a2a4adeb80052a8d57fa71b : SUCCESS in 5m 58s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/fcbe6b0a843948aab62748f82aa2ad5c : SUCCESS in 6m 46s\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/d8ef3b05b0df4b33bb3673c734473c36 : SUCCESS in 4m 56s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/fe1dcda4c40646fba4846f70da40a86c : SUCCESS in 7m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6e0b93b62ac7426ab2298f843c617115 : SUCCESS in 10m 14s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/e5426e862de14cab89c2c437525abb4a : SUCCESS in 13m 39s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/158af8fb34fb4223915efc2e92366546 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f055f794a56f4a038810e5b9a43d8a92 : SUCCESS in 9m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c11444cadf14c82b56bea9237c4a5e2 : SUCCESS in 7m 22s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/d4211149c97a420caea8d8c524b395cc : SUCCESS in 5m 51s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/4218d5a9f705444ca4dbedabea84322a : FAILURE in 59m 26s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/1ad307c20c84498d8641a50384666bf4 : SUCCESS in 51m 35s (non-voting)\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/257049a7ce1b482c8e21ab956e06c9be : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":1},{"id":"464f6aea46c8ca13f1512624eef378c6585dac19","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2024-01-05 10:44:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1e69843f82bd5b229d83c857695bf83f9d59b623","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 12:07:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/88045c2c5f0b4ef8ae28d98535329c68\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/681183267f4d414983ebe4b7cbad8d7f : SUCCESS in 5m 46s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/dfc6e3b93ee24ca3b40e37acdb57142e : SUCCESS in 6m 06s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/6f42d05687ff4eda8a3d83cea33af8a8 : SUCCESS in 3m 20s (non-voting)\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/0485bd9c5a6d4f0f9ae39372ddb59dc9 : SUCCESS in 8m 05s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/89af9847adfa4ba8bd1858c1ceef4e04 : SUCCESS in 8m 52s\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/08c657e057b14683aeb40c17809082da : SUCCESS in 4m 27s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/d2a8cc9bd8e947999f9603050c5d7647 : SUCCESS in 9m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f5e83e0cab584d34b02aa95770e31071 : SUCCESS in 11m 16s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/d00ece79a776419b860c63a6e4017174 : SUCCESS in 15m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec6069225f0746a99dfad37b46efce85 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eec3aca5289640ac97c1c97a1c71dc5f : SUCCESS in 8m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74f8e6dc6ec8496098a36b4b189301c6 : SUCCESS in 8m 37s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/fd50ac004c7f4906a76c2b7e230570a1 : SUCCESS in 6m 33s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/8a68a31173d342ab8da5a64f461d2424 : FAILURE in 49m 14s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/f6b4c0833f9f4740839cc3ae13536896 : SUCCESS in 1h 17m 13s (non-voting)\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/29ed8d90773249acb5431331fab8fe06 : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":1},{"id":"5b8e520c89eb69f74d601cd9de3b882415701511","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2024-01-05 12:10:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b181ca01849a9b613bb3b4e1ecec421146e96217","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 13:19:14.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9733befee63c47ada916a8ddecb29760\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/13ff918666ac4d94a6a74eec9c41c4c8 : SUCCESS in 5m 20s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/ac92efdc95ec409487e10802701ea54f : SUCCESS in 6m 46s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/ec5890bec2704366b53bf1379fd19475 : SUCCESS in 3m 53s (non-voting)\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/c56787377f4245ca965a77ca653ef01d : SUCCESS in 4m 00s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/53e628526ae14957a14b7979ce84d052 : SUCCESS in 7m 00s\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/d2d5511e75e04375affef3952ce9fa11 : SUCCESS in 7m 10s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/428c9c9be1754752b5875402ab8a9aa9 : SUCCESS in 6m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5b851e122a674ca5875e82ba956668d5 : SUCCESS in 11m 53s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/5122f1981be849df87050800e17d03db : SUCCESS in 11m 00s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d9827b82779492ca870312d1bfe0a96 : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/685873d43cab48c880678ad5bd23971a : SUCCESS in 11m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a53ad64e0924b1d9f5998e29edbbf99 : SUCCESS in 6m 53s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/3ba926a7a2e04d4d8b5774996693ead3 : SUCCESS in 6m 23s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/e9484fab1ca54598b16be6a8bf6584f1 : FAILURE in 1h 00m 51s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/73dc02acd2aa4217bc28c5d4915db043 : SUCCESS in 1h 03m 25s (non-voting)\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f76ecbcf9faa4eb09cd8a73de6a7d893 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":1},{"id":"aaee4984fbc0db2e61fb1e83a71b5d6b407afb17","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2024-01-05 13:20:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"224f5864a62f93347b384582177a73b0fb4fb548","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 14:27:29.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9c116e7f771d4b809bc2d52b91fce950\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/0b2e98bd906847b6b01ba78441e94011 : SUCCESS in 5m 50s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/74fccbbcec6d46e6b1a983b151497c29 : SUCCESS in 6m 11s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/bdec2b56d13440aa830b8a26cf69cbca : SUCCESS in 3m 44s (non-voting)\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/11d9f289966046b6870e285d15079cbf : SUCCESS in 5m 02s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/bc25fb8044bb4f93b47161100e81e261 : SUCCESS in 8m 56s\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/e82d8ce944294d8c9bc2aca31f79cc2a : SUCCESS in 6m 17s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/b30c3e86f5d84a8cb6740e9d5dc6432a : SUCCESS in 8m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7c85205fad1f471fbe78cc2dc5293981 : SUCCESS in 10m 43s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/75d08d91ffff4a8a91187f5149da310b : SUCCESS in 13m 45s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e032c9a5e698451a979265a17679eaff : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7098422cb67d4da19974a48eaa8548d0 : SUCCESS in 10m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd26e11c72a24db79d0d2a5a9ce71136 : SUCCESS in 6m 46s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/87660b71e2bc4db1b0b8904e457cb855 : SUCCESS in 6m 30s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/b6ca6febeab14ff8820b2d0684d265e8 : FAILURE in 1h 01m 50s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/3eee9b44b08749dbbbcb380f680270e7 : SUCCESS in 38m 40s (non-voting)\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6546ccccae8840f8b9945cf35dc23b31 : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":1},{"id":"cf85d624097616e3bd266bd4ce3af70616f4c6c4","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2024-01-05 15:00:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"005eba9f03ad6760c4ed94a0571eea87cb275df0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 16:58:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61948b44cc4c448f86cb01ce2e978631\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/1b797231f04842eab8d59c71944bba0b : SUCCESS in 5m 32s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/5826d7499d0a4dfe96cb5f31e1b95f72 : SUCCESS in 6m 34s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/76f73e0aa000461db7c6f8be2ebdaef2 : SUCCESS in 3m 49s (non-voting)\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/df7e0c6de3a041728c5d621416a60eac : SUCCESS in 7m 28s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/7c1729e5e6ff43c89b4d21b0c466e668 : SUCCESS in 7m 15s\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/4ca0adea34cb476080e4b67c57d2ecbf : SUCCESS in 5m 22s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/6a29af0856b04b27a368b77b984cfcac : SUCCESS in 8m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dd0f0f62d8654659b1c9407a1a0ce095 : SUCCESS in 12m 03s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/beecb54c84464d16b3b3d8c3fee7e39e : SUCCESS in 12m 38s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef07358eaefa4b92a89359e9c75f7dd9 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5781a06c91554e21a67c7b7564c7fc51 : SUCCESS in 11m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ae9a61a5e6b48e9aff68cf29503c821 : SUCCESS in 7m 37s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/113146d8ca7c4f76ae383680f38e18fe : SUCCESS in 7m 01s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/cb3c448d09874d289836f129ec62f6ae : SUCCESS in 1h 27m 20s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/f67c79c569f4459eb71c362623a05764 : FAILURE in 49m 21s (non-voting)\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d6a08fa297574cc18dfd1c0067d50281 : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":1},{"id":"7db0e7651b8ddbdb80fbdb434b5958655eb09a83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 16:58:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a8b11c4026e2138826e0d38079ae7e3e5fcb5325","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 18:47:08.000000000","message":"Patch Set 1: Verified-2\n\n(3 comments)\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0b76cde9086d422c8bfc33da1501c70d\n\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/b408ed3778cb4a18acf2a187735d0803 : SUCCESS in 5m 03s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/ee05ed266c3d479ca61ad092173b820a : SUCCESS in 6m 40s\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/677f90dad2454de2986056a1a4fe9a6d : SUCCESS in 4m 45s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/d99edb38d9be47be9700f72f6384fceb : SUCCESS in 6m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ca5ef31dac9e48b99cb53da6a1d81b6e : SUCCESS in 11m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b53cc6a54b245c4a07f110ca850865d : FAILURE in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0de75f0f04842eeb6050df56ff84558 : SUCCESS in 10m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1975a9c4cbda4670836b9bbc425970d8 : SUCCESS in 7m 52s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/73efc8a036e346ddbf60857a97e9cd8a : SUCCESS in 6m 56s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/664c41c40c0d4cf29c1baeab31d166aa : SUCCESS in 1h 44m 13s","accounts_in_message":[],"_revision_number":1},{"id":"cb3de3bddd449c6e355cef8be4d021cc16bdf2fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2024-01-08 07:48:41.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"fd4b105e573f6dec8f91d12a28ad02027e20c96f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-08 09:22:45.000000000","message":"Patch Set 2: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c899d763c9604f74847a562c1df62950\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/3063d2bc05f544428cc7f7e793767f20 : SUCCESS in 5m 18s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/2d8c7890843841b087727aaf675fcd77 : SUCCESS in 5m 47s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/6edba48b10d34b9a92f0aee4c680f41f : SUCCESS in 3m 31s (non-voting)\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/24482e6c540c44739902391158299e7f : SUCCESS in 4m 32s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/fb394234789d445981e1e49f896ebfdb : SUCCESS in 6m 23s\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/3c665d7670f04c77a6a9a31e396f5310 : SUCCESS in 4m 58s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/9d050fc61c6c4a4d9b38561d6f69a3b2 : SUCCESS in 6m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f58b3e30e49948a192b734ecdc0b838c : SUCCESS in 10m 57s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/11558d2e71d84c3d99884e0da5ace430 : SUCCESS in 11m 03s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e66e9dcf15a3459aa170e84056810cbc : FAILURE in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa097be09c44430f9e313c49c965bf22 : SUCCESS in 11m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f26f726e53b43bba107d3ed69b7537b : SUCCESS in 6m 53s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/c67f5e25300e4efdbf65249ee745b26e : SUCCESS in 9m 12s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/b03517f7891645df88bdefe1d8b7f4db : SUCCESS in 1h 27m 00s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/258a11426d9f423b982d61a3ddf9799b : FAILURE in 56m 07s (non-voting)\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d7acb6e04b1841379030c7078ae2450a : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":2},{"id":"2950d2d6c01f42bafbeaf79e1fb845aae9c21885","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2024-01-08 10:55:21.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"75476e4d0d98918cb8aee960ad65afadc7b42c63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-08 12:46:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70d09d06bb2b469e8b6b4e40b91981bb\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/43018bc4859e4a598bf36dd864706662 : SUCCESS in 5m 21s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/9bb0ff8a6a7a4a02a4c9f986715023cf : SUCCESS in 5m 53s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/f740959adb6244f4b9f485632d87a5b0 : SUCCESS in 3m 55s (non-voting)\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/6ea44a7f84c6488aa26c5a911c0c0226 : SUCCESS in 5m 20s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/2b64456714644af5baad7c100ee75686 : SUCCESS in 6m 46s\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/448cd9aa74d24492944a65b2e6be34a1 : SUCCESS in 5m 06s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/f1f9a07834b6468d91fcc3dc5307f57a : SUCCESS in 7m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/da6da0dfff4e446dbf81170e9afa0ade : SUCCESS in 10m 27s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/44a7c96b303a4305bc47cbd74ce44bb4 : SUCCESS in 12m 25s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8d9bd1eb83545cda7838feb70f97869 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2bbe46742ca84e028db4152098068dcd : SUCCESS in 9m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5675a7d11f2249f5a4de49d4283f280a : SUCCESS in 8m 01s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/d8567249b165490ca8483d81b4ac4cfb : SUCCESS in 8m 54s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/55621391dfb444e29c4e8ed3f5908d73 : SUCCESS in 1h 46m 29s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/3c38491b4cf444308706a83e7c378ab0 : FAILURE in 39m 27s (non-voting)\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e9c301d4b51d45f18542efb57da1d9d7 : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":3},{"id":"7210527a51e37866f00a409f042cb44f80e136eb","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2024-01-08 13:58:43.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"bf2f6486af0dc3904808807343ee9a0e70936d61","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2024-01-17 16:49:32.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"832da28f59929e5ddc4469dc96ecc8908c60e844","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 16:50:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"70c6cec04158606c3e32f91fd57421a5738016da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 17:51:11.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/79e95af1d3c04195a792dffcff018dd7\n\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/e3be49f3f1d341dca265e304e06f543b : SUCCESS in 6m 47s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/6ad817a3941c47028ac94a703fd6b703 : SUCCESS in 6m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5a8a72d43e674ea48ff39cea1df83b2b : SUCCESS in 10m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/697d20236ebd4aeebcf69065a358b784 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a6bebdd8a85949779ce5fa5b7f68cc46 : SUCCESS in 10m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e5d8438b97d4e18b7b35ece65b17b8b : SUCCESS in 8m 41s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/0f9cbd30269c4805a0220404b1d8855b : SUCCESS in 5m 59s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/3ec54bffbfec4c88acf7ff8fd3a71403 : FAILURE in 55m 44s","accounts_in_message":[],"_revision_number":3},{"id":"c7d65ace13df52ab00631cac6d720c07b9c458b7","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2024-01-18 08:08:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"79fd98acb495ab06bcf8a8ed8f4efe6aa3dae433","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 10:19:21.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1c47d4622eba48c8aceebe8c8fccae25\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/472f0c9e7d6745c293d9cf7c45f402a0 : SUCCESS in 5m 21s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/7c5ffce02ffe4127901072c908f6f031 : SUCCESS in 4m 04s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/8bc25ac34bba48f79a1e8b1719512a6d : SUCCESS in 5m 25s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/0e31e6910e0444ac9109143d787e426f : SUCCESS in 8m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0780ba2767ac4a37a5c104d31e669d7d : SUCCESS in 9m 49s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/c1f5af60829c4580add952b55c1fc239 : SUCCESS in 13m 29s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7412fd9e9bf8400bbf9d6098a3264333 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a15216b85b2463f905a2fa9e9433bb8 : SUCCESS in 10m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/714a1004a135439092b48e6779cafb12 : SUCCESS in 7m 50s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/6ad8059687f94df1be1b910293877afd : SUCCESS in 7m 09s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/83c34d0743014e9a9882c3b6410f54d3 : TIMED_OUT in 2h 03m 18s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/b5e1b050fa0f4f5f917cbcea1a3a8ac0 : SUCCESS in 58m 22s (non-voting)\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/597df7ec333e40519539ac051ca288d0 : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":3},{"id":"2d8467818a1df3a42c40eaa685f725086fa4b437","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2024-01-18 10:46:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a2ebadef696477110c8d6dbe95022dcaa08c204e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 11:52:09.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e5dc1532fe8c4f9eb51b3a07faf4d03c\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/c8d14cfcc9314218bf0197285bc35c48 : SUCCESS in 5m 40s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/83a7177616a54f22b73e52ee7eedbd55 : SUCCESS in 4m 29s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/3d9fb28699e04f689785ce7514779cdb : SUCCESS in 5m 15s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/67035a6b93544ce59c84956ce26c672e : SUCCESS in 6m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/98e2295ee58e4611bc07a862ec9c3a6a : SUCCESS in 11m 10s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/35148594a1b5407eaf0e029ad8b0b748 : SUCCESS in 11m 16s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b468b0eca3446fa98b717894eb5ec39 : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3978b4c57cba4d92be6fed496b1a097b : SUCCESS in 11m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36fa51b25a3b454f93b9371719973796 : SUCCESS in 7m 36s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/cf315284eb64416887af3df15c499ea4 : SUCCESS in 6m 52s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/74ea910348ac4f669d62f4a841390bc8 : FAILURE in 1h 03m 28s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/de3baf5ec83d4a66be9025817be01e60 : SUCCESS in 57m 20s (non-voting)\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9275c48d2b5446359536ecc005e4f929 : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":3},{"id":"661e2781c3062c2b7ec4f7893beba89ddece378b","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2024-01-18 11:54:09.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a55ec24176ee1d700fb2ae3158ec2e5a0b2674a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 13:54:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d091e978d15648fcb10e9323af4ded3e\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/68f562c52c334493934d6290969a2fa3 : SUCCESS in 4m 58s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/48f7cbecf72c484bad44dff6ed6f78c5 : SUCCESS in 3m 52s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/5b68be20ed5c4fbabdfc6c04da55df41 : SUCCESS in 4m 59s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/e7f1d89a29754d33baeab5067fe0fbc0 : SUCCESS in 6m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3410cd1846e84cd1bad9c2dc08ddea1a : SUCCESS in 11m 44s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/fa1c8a9f2ee944439681fccd2de261ea : SUCCESS in 12m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f5f333ef9884eeeb61c2c2b44dc3cd2 : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f4c03caf79a48b2adacd586bd71c4e1 : SUCCESS in 10m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d6a75b75bd04c50a93a9e2e8e6c182b : SUCCESS in 8m 14s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/705074609ed14fc297c85f4a5f7cc51c : SUCCESS in 6m 10s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/d7804f8491f24ae4861ac166b5598410 : SUCCESS in 1h 38m 48s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/2e7df6b4d54f4dcb90cb97e5a15fce25 : SUCCESS in 1h 00m 39s (non-voting)\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f6d22035838d4e1eaac5a5728a3cc530 : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":3},{"id":"060f6a85b93d7c0eae169dbd36ebbb596f28882b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 13:54:42.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5a683cf96583723933c7bb23f127d4ea57e2a60c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 15:43:23.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3f6db3ebc00413aa5295586078a1d46\n\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/a1884c371e874a7c970f434f9839bb0d : SUCCESS in 7m 16s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/6a934f18ed4e48548fe804e92ba3959a : SUCCESS in 7m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/07c2b129540e46808c2a27d3deb0e694 : SUCCESS in 10m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fbb63128e7d4ea78e2ca29de8399963 : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be6b7dab72b54ef48a2f81ca424bb3d0 : SUCCESS in 10m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58debcc00205472899985e758c447de0 : SUCCESS in 7m 34s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/8d3c361e9d9f4289b1f1b871e03ffc35 : SUCCESS in 6m 21s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/6cce5f27602641e8a51bfa08116adc26 : SUCCESS in 1h 44m 13s","accounts_in_message":[],"_revision_number":3},{"id":"03414ed6f207d7fadcf91926ae2f8ec056cf550e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 15:43:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"7e9b618b64728f53458e3d9ad90cf36a7fb6086e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 15:44:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b5a6a9a84d54f96a90d14e8f596d408\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/049cc1e7c51f453ab0fabb4169f3351f : SUCCESS in 49s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"35da6c52e46cb3be2e4d41479cb15a8a88a790ba","revisions":{"14b50b7c42458e83ff988f5279ad7b6094b39608":{"kind":"REWORK","_number":1,"created":"2023-12-11 18:01:16.000000000","uploader":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"ref":"refs/changes/70/903370/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/70/903370/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/70/903370/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/70/903370/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/70/903370/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/70/903370/1"}}},"commit":{"parents":[{"commit":"185cb40545490ea3592d6f282227e34162c7e31b","subject":"Merge \"pytest-based selenium tests add project members, roles tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/185cb40545490ea3592d6f282227e34162c7e31b"}]}],"author":{"name":"Jan Jasek","email":"jasekjan96@gmail.com","date":"2023-12-11 18:00:08.000000000","tz":60},"committer":{"name":"Jan Jasek","email":"jasekjan96@gmail.com","date":"2023-12-11 18:00:08.000000000","tz":60},"subject":"pytest-based selenium tests add volume actions tests","message":"pytest-based selenium tests add volume actions tests\n\ntest_extend_volume_demo\ntest_volume_launch_as_instance_demo\ntest_volume_upload_to_image_demo\n\nChange-Id: Ic577a94514881e56d32087a4fc4d6a06414ff568\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/14b50b7c42458e83ff988f5279ad7b6094b39608"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/14b50b7c42458e83ff988f5279ad7b6094b39608"}]},"branch":"refs/heads/master"},"fbc34c0e025cc17a356ff9e6fa46ee5396a8a5bd":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-01-08 07:48:41.000000000","uploader":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"ref":"refs/changes/70/903370/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/70/903370/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/70/903370/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/70/903370/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/70/903370/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/70/903370/2"}}},"commit":{"parents":[{"commit":"42915b6a78bb35b030371892665ce7ab30c6def0","subject":"pytest-based selenium tests add volume snapshots admin, edit, action tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/42915b6a78bb35b030371892665ce7ab30c6def0"}]}],"author":{"name":"Jan Jasek","email":"jasekjan96@gmail.com","date":"2023-12-11 18:00:08.000000000","tz":60},"committer":{"name":"Jan Jasek","email":"jjasek@redhat.com","date":"2024-01-08 07:48:41.000000000","tz":0},"subject":"pytest-based selenium tests add volume actions tests","message":"pytest-based selenium tests add volume actions tests\n\ntest_extend_volume_demo\ntest_volume_launch_as_instance_demo\ntest_volume_upload_to_image_demo\n\nChange-Id: Ic577a94514881e56d32087a4fc4d6a06414ff568\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/fbc34c0e025cc17a356ff9e6fa46ee5396a8a5bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/fbc34c0e025cc17a356ff9e6fa46ee5396a8a5bd"}]},"branch":"refs/heads/master","description":"Rebase"},"35da6c52e46cb3be2e4d41479cb15a8a88a790ba":{"kind":"REWORK","_number":3,"created":"2024-01-08 10:55:21.000000000","uploader":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"ref":"refs/changes/70/903370/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/70/903370/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/70/903370/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/70/903370/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/70/903370/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/70/903370/3"}}},"commit":{"parents":[{"commit":"42915b6a78bb35b030371892665ce7ab30c6def0","subject":"pytest-based selenium tests add volume snapshots admin, edit, action tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/42915b6a78bb35b030371892665ce7ab30c6def0"}]}],"author":{"name":"Jan Jasek","email":"jasekjan96@gmail.com","date":"2023-12-11 18:00:08.000000000","tz":60},"committer":{"name":"Jan Jasek","email":"jasekjan96@gmail.com","date":"2024-01-08 10:53:33.000000000","tz":60},"subject":"pytest-based selenium tests add volume actions tests","message":"pytest-based selenium tests add volume actions tests\n\ntest_extend_volume_demo\ntest_volume_launch_as_instance_demo\ntest_volume_upload_to_image_demo\n\nChange-Id: Ic577a94514881e56d32087a4fc4d6a06414ff568\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/35da6c52e46cb3be2e4d41479cb15a8a88a790ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/35da6c52e46cb3be2e4d41479cb15a8a88a790ba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
