)]}'
{"id":"openstack%2Fhorizon~908910","triplet_id":"openstack%2Fhorizon~stable%2F2023.2~I3e71a208a1c7212e168d63a259f2adddf27dbabf","project":"openstack/horizon","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"14567":{"account":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"last_update":"2024-02-16 14:20:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3e71a208a1c7212e168d63a259f2adddf27dbabf","subject":"Extend configurable skippability of neutron calls to project instance detail","status":"MERGED","created":"2024-02-13 15:28:29.000000000","updated":"2024-02-16 14:21:55.000000000","submitted":"2024-02-16 14:20:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"908910","meta_rev_id":"ebe9f8d49ea9d95f2c24c2b6a18b9303a7b4326c","_number":908910,"virtual_id_number":908910,"owner":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-16 14:20:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":2,"date":"2024-02-16 12:27:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":2,"date":"2024-02-14 16:12:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":1,"date":"2024-02-16 12:27:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-13 16:12:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-14 16:12:25.000000000","updated_by":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"reviewer":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"state":"REVIEWER"},{"updated":"2024-02-16 12:27:58.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"}],"messages":[{"id":"cfc89adc02fc064837f1efe2050b83d5208297a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2024-02-13 15:28:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4bec9126ffe9e34a71465134ce252d4572d6639d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 16:12:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f547242ebdad4da1a6868409f1dde208\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/dea16e78940c407fbae89ae886969ed1 : SUCCESS in 4m 56s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/80ba59c993184c1386a58050d93c73a4 : SUCCESS in 5m 42s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/937aa03df00f40698859b98224d2b2b8 : SUCCESS in 4m 34s (non-voting)\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/8f9f511ea0f64d2e9e39704ca6428445 : SUCCESS in 4m 50s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/8a4bfdb143c94e81bd7b1f66cccaa00d : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1fcc846bbe3442ab9f17fb0fc2692642 : SUCCESS in 9m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2da23528ab1a4d648af762673c2e171c : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a04cf43bf8f4aa5b89ec42c8fa85469 : SUCCESS in 10m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3c036566e5b84d9aa3d70e4912e8f2d3 : SUCCESS in 8m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dd4fe4fa246a46e1ac222c03c403db39 : FAILURE in 10m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfee5cfd153b40b0851a15ed0dcd942f : SUCCESS in 6m 29s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/fd74d15a47d14e4693011a43f791434a : SUCCESS in 6m 16s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/ed47d5bbd1084e79b87ee16e34807b13 : FAILURE in 23m 58s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/16db00fbe8dc48d783fc5ae499584881 : SUCCESS in 36m 57s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7eaa82e03bd3438aa457c40e5ab289c9 : SUCCESS in 5m 22s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/f7940193d82549dd8990201a369f2513 : SUCCESS in 37m 41s","accounts_in_message":[],"_revision_number":1},{"id":"d08dc2317deef0ad1cf922bb01d259d9f4a73b50","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2024-02-13 16:39:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"95be1bd70f9b6dd64434243ec1c7e182ad4e6365","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 18:39:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6977110ae5744f8e90bc2503622340e2\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/1058c36bb971465bbcc7042ae3ab9cf8 : SUCCESS in 5m 18s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/475a690e79e345caabf842483a683672 : SUCCESS in 6m 36s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/a9091d5258f241998fbc76ccd901b15c : SUCCESS in 4m 30s (non-voting)\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/46a5998f669648ecab56b723c376436b : SUCCESS in 4m 09s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/394afd67032e461e878c169021cc3d6d : SUCCESS in 7m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/25e3372c1ee44b72ad2d3b27b2511ef8 : SUCCESS in 9m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/884d8b2ede334e42ad3c119a1b81618b : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7cde691d8b31424c8e90c6c34e6c4447 : SUCCESS in 8m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/08367864875944078b60ccf055053db3 : SUCCESS in 9m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9e3e53e2e68d4fcc8e1b856fdc94c31e : FAILURE in 9m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25565496c270428aa870d06d489fa440 : SUCCESS in 6m 33s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/8a32031bd45349158ae231313f134a2b : SUCCESS in 6m 54s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/8b647781f49643a49506f32ed9c8957a : SUCCESS in 1h 53m 25s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/1ee9fbb01062468ba959bc965219ff9c : SUCCESS in 36m 50s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1d9ea611e94647e59b662bb4a1ecfb21 : SUCCESS in 4m 16s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/57041ffbc2c74c7fbb9d4e70e3193b41 : SUCCESS in 39m 11s","accounts_in_message":[],"_revision_number":1},{"id":"0b97a9a1e3dc74ceadd751fa67024cbcf37d2537","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2024-02-14 16:12:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"190c9d3a5379557da2249c63876eed432cd5eade","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2024-02-16 12:27:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ef5f067ee2db11c456a7ca1ddbd50342340366cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 12:29:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0fb900505422cd79d00ca4034fa3e0064f59d223","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 14:20:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95eaa97a45e74284a7014b4e7789234b\n\n- horizon-nodejs18-run-lint https://zuul.opendev.org/t/openstack/build/c96bda7ef79b461f8a36ecb0441b915b : SUCCESS in 4m 34s\n- horizon-nodejs18-run-test https://zuul.opendev.org/t/openstack/build/c3aeeeb786f64cd3a1bbf8bf17636ccd : SUCCESS in 6m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44eb1fe4fc654784938ed0c17d8857d1 : SUCCESS in 11m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51f44456249b47738aca711fa280b900 : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/78902ffe171d41a58d443707308273cf : SUCCESS in 10m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/271ca2d80cb74ff686673f7aa36912ae : SUCCESS in 7m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/632bd1adb3054346bd1128475484bf69 : SUCCESS in 7m 07s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/ef9f7683d07d48e69db72e374067520c : SUCCESS in 35m 37s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/dc4a67d3681948f9890634f896da2b08 : SUCCESS in 5m 25s\n- horizon-integration-tests https://zuul.opendev.org/t/openstack/build/d4d9bf8877f8459780d1670940817820 : SUCCESS in 1h 45m 06s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/2fb7e878739f4bf7b24dbee7bd2ad4da : SUCCESS in 33m 17s","accounts_in_message":[],"_revision_number":1},{"id":"98f29f23f7e24e5a9412b9db616eee4284e83a1a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 14:20:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ebe9f8d49ea9d95f2c24c2b6a18b9303a7b4326c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 14:21:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bda41badaae14cfd99cc1ab72961118d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddd9f11bb95a464ca8a46301cb5ee38f : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4de36bb649c514f50d2a958c9277097a08b23cec","revisions":{"4de36bb649c514f50d2a958c9277097a08b23cec":{"kind":"REWORK","_number":1,"created":"2024-02-13 15:28:29.000000000","uploader":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"ref":"refs/changes/10/908910/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/10/908910/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/10/908910/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/10/908910/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/10/908910/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/10/908910/1"}}},"commit":{"parents":[{"commit":"06e0a3e0447664480f7e8f784125102d54da85ad","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/06e0a3e0447664480f7e8f784125102d54da85ad"}]}],"author":{"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","date":"2024-01-26 18:10:41.000000000","tz":-180},"committer":{"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","date":"2024-02-13 15:28:16.000000000","tz":-180},"subject":"Extend configurable skippability of neutron calls to project instance detail","message":"Extend configurable skippability of neutron calls to project instance detail\n\nThe OPENSTACK_INSTANCE_RETRIEVE_IP_ADDRESSES config aids\nin envs struggling to load the instance list due to having\ntoo many ports or bad neutron plugin performance. However,\nthe config does not apply its effect to the instance detail\npage, which cannot be loaded due to the neutron calls\ntaking too long.\n\nThis patch extends the config option to the instance\ndetail page, allowing the same benefit (and side-effects)\nof the instance list page.\n\nRelated-bug: #2045168\nChange-Id: I3e71a208a1c7212e168d63a259f2adddf27dbabf\n(cherry picked from commit 95089025fda7c8cce6f7195e2a63f7f09efc9e0a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/4de36bb649c514f50d2a958c9277097a08b23cec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/4de36bb649c514f50d2a958c9277097a08b23cec"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
