)]}'
{"id":"openstack%2Fhorizon~936664","triplet_id":"openstack%2Fhorizon~master~I3ea3d0cf42dc8e89a23a763defe84389da56366b","project":"openstack/horizon","branch":"master","topic":"config_ui_tests_different_themes","attention_set":{},"removed_from_attention_set":{"29313":{"account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"last_update":"2024-12-19 10:37:30.000000000","reason":"Change was submitted"},"35153":{"account":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"last_update":"2024-12-19 10:37:30.000000000","reason":"Change was submitted"},"6914":{"account":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"last_update":"2024-12-19 10:37:30.000000000","reason":"Change was submitted"},"8648":{"account":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"last_update":"2024-12-19 08:36:54.000000000","reason":"\u003cGERRIT_ACCOUNT_8648\u003e replied on the change","reason_account":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"}},"35133":{"account":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"last_update":"2024-12-19 10:37:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3ea3d0cf42dc8e89a23a763defe84389da56366b","subject":"Allow configuring separated UI tests for different themes","status":"MERGED","created":"2024-11-29 02:07:48.000000000","updated":"2024-12-19 10:38:54.000000000","submitted":"2024-12-19 10:37:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"936664-config_ui_tests_different_themes","meta_rev_id":"8fb650adbb935285169bb6d474b3d8e979843537","_number":936664,"virtual_id_number":936664,"owner":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-19 10:37:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"all":[{"value":2,"date":"2024-12-19 08:36:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"all":[{"value":1,"date":"2024-12-19 08:36:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2024-12-18 20:39:34.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"}],"reviewers":{"REVIEWER":[{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-29 03:51:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-16 12:22:56.000000000","updated_by":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"reviewer":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"state":"CC"},{"updated":"2024-12-18 20:39:34.000000000","updated_by":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"reviewer":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"state":"REVIEWER"},{"updated":"2024-12-18 20:39:34.000000000","updated_by":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2024-12-18 20:39:34.000000000","updated_by":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"reviewer":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"state":"REVIEWER"},{"updated":"2024-12-18 20:39:34.000000000","updated_by":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"reviewer":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"state":"REVIEWER"}],"messages":[{"id":"78e81457d6e8af6f6fbe46d392774acf0bb70c0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2024-11-29 02:07:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"02afbb7d020f2fc3e2157f7267cbf8c92a7085c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 03:51:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02324b4c4be14dd4978cc3502a3578fc\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/9542fcb56be442a18593fc660f502bdf : SUCCESS in 7m 46s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/04ff4a0d45394da1a1d79c7740153a17 : SUCCESS in 4m 54s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/9ff9f379e3ce4d4ab0e3c1631caf728b : SUCCESS in 6m 17s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/65f355cedfc94abd93613904e1c7e424 : SUCCESS in 9m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cb0ca07e184d417dbb196b88d9a5e48a : SUCCESS in 13m 58s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/dbd894023046409eaca67dc3075a1e06 : SUCCESS in 13m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/263dc158275748028e8c915e9c808de8 : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb38ec405f404b5ebfc66cafba879062 : SUCCESS in 12m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6d3fdcd8c028455fb17c29ecc519a926 : SUCCESS in 13m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4feeaba337864407a240888613190299 : SUCCESS in 7m 22s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/730968cfb8324a4286e6a44322176690 : SUCCESS in 4m 42s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/4d5da8af43cd462f8c1605c153e3d3b2 : SUCCESS in 1h 20m 19s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/a3d6b2fd15394fcb81f11b969c03b715 : SUCCESS in 35m 40s (non-voting)\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2562607dd5e54c389225bbffeef81959 : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":1},{"id":"65200d06b957f40c6df1c9a77e8b94cb1cef6a14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2024-11-30 01:45:23.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"49c7c3f8a07b24b07d2092a2e90a7d09b4bc942b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-30 03:03:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f2221a35a9ab4a888b1f5f616d378885\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/33c13dafdcb34145b4b9d63dacaaff22 : SUCCESS in 5m 01s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/e005284de6c64ec78c681d7d0363267f : SUCCESS in 4m 20s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/97c06f0bfe384a05984a8da75b3cefa2 : SUCCESS in 4m 20s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/02eb4920232f44f69e2f8880e87e5496 : SUCCESS in 5m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8c08099c3bb9492aa082d118e4ecf905 : SUCCESS in 12m 46s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/bbcabe8e552a45fda1ec4dc892483e36 : RETRY_LIMIT in 1m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfe5bb21b6364bc3aba8533ee8c1f786 : FAILURE in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d64d5db9e16c4d4eb0ec2eedbef967d7 : SUCCESS in 13m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/85a7ff9070c14e029d18f03aa8929986 : SUCCESS in 13m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0bd0366786a4177b4687cc75b0bf568 : SUCCESS in 6m 29s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/3a2bd051bdad4a44b279ff78a05106c4 : SUCCESS in 5m 55s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/dd3f66d15d68435fad175c3a086c79d8 : SUCCESS in 1h 10m 54s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/94197dfc45ff42ceb9b52846b5de33d5 : FAILURE in 22m 43s (non-voting)\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/983b1c76268d48ea8a46ff482562bf6f : SUCCESS in 5m 09s\n\nWarning:\n  Comments left for invalid file openstack_dashboard/test/selenium/conftest.py\n  Comments left for invalid file openstack_dashboard/test/selenium/integration/conftest.py\n  Comments left for invalid file openstack_dashboard/test/selenium/integration/test_images.py\n  Comments left for invalid file openstack_dashboard/test/selenium/integration/test_instances.py\n  Comments left for invalid file openstack_dashboard/test/selenium/integration/test_networks.py\n  Comments left for invalid file openstack_dashboard/test/selenium/integration/test_projects.py\n  Comments left for invalid file openstack_dashboard/test/selenium/integration/test_routers.py\n  Comments left for invalid file openstack_dashboard/test/selenium/ui/test_hypervisors.py\n  Comments left for invalid file openstack_dashboard/test/selenium/widgets.py","accounts_in_message":[],"_revision_number":2},{"id":"4b9855355ff84f9d2f2955dc35d679cf841dc79d","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2024-11-30 03:11:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"50f18b2f8a97955d1614e7ac7693e6007fb7912d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-30 04:29:57.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6d33e2b00c604cd1b45ce2267f138fae\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/643efdc2f3ab4020813b29cdd2cc1b82 : SUCCESS in 14m 05s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/dfde43c2a99e41588f04ff38563fe5f5 : SUCCESS in 6m 55s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/321eeb080af44edf86f9e29f167b768a : SUCCESS in 3m 00s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/41e89fd50cb14b18b9b714d6a0416ccd : SUCCESS in 15m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b177a889403c440e804732750f51c136 : SUCCESS in 6m 43s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/6c3e0ae878554ae3b0ba93809eba0a37 : RETRY_LIMIT in 1m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70f86f20711f4a89942605adf47c5fe5 : FAILURE in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b607594dc0a4ad5b70ab8947e4c393c : SUCCESS in 12m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/807a116e68e6449c8735185ef5931300 : SUCCESS in 11m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50da9c6784f1493ba06244c0097cb480 : SUCCESS in 8m 36s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/0ab05f318280440282b28a62b84b9952 : SUCCESS in 5m 59s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/414151f61eb4423aae4f66cfacd4e9dd : SUCCESS in 1h 16m 49s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/4beccb1b50cf4d58ad46902bb2a7562f : SUCCESS in 23m 53s (non-voting)\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9dad3856ecaa4d278eb34ba2060a7426 : SUCCESS in 7m 46s\n\nWarning:\n  Comments left for invalid file openstack_dashboard/test/selenium/conftest.py\n  Comments left for invalid file openstack_dashboard/test/selenium/integration/conftest.py\n  Comments left for invalid file openstack_dashboard/test/selenium/integration/test_images.py\n  Comments left for invalid file openstack_dashboard/test/selenium/integration/test_instances.py\n  Comments left for invalid file openstack_dashboard/test/selenium/integration/test_networks.py\n  Comments left for invalid file openstack_dashboard/test/selenium/integration/test_projects.py\n  Comments left for invalid file openstack_dashboard/test/selenium/integration/test_routers.py\n  Comments left for invalid file openstack_dashboard/test/selenium/ui/test_hypervisors.py\n  Comments left for invalid file openstack_dashboard/test/selenium/widgets.py","accounts_in_message":[],"_revision_number":2},{"id":"785d3c82999d9a1a243fcac3f149892086db6c7d","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2024-12-06 00:40:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d575aa989bfe3937fafe0e8acb7754bb731bc6ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 00:40:22.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/horizon for 936664,2","accounts_in_message":[],"_revision_number":2},{"id":"c7bd32885d58e0bbcc92afa731cbd5ce292cb42e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2024-12-06 17:47:30.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"d41078713d6547dc9cbdd6c8b3624736a454566f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 19:21:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6859ffcc46e84d60846f14737e3c438a\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/a14a3e9b930e439ebe3b018095eb9116 : SUCCESS in 6m 38s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/f24da442e0a74ad1b1cc2d4a55e996ff : SUCCESS in 3m 55s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/0d8ad1daed4542f6853238782ba6cd33 : SUCCESS in 5m 08s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/ae6b2d886ba14e2881aa3f015d2a79bf : SUCCESS in 8m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fa1402e6289e4e48810cfbcea719392e : SUCCESS in 13m 19s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/f56b90c4b28d4608b668fbbb15d23599 : SUCCESS in 14m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/974806eaf67d4e49862e154a30096feb : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/11ae4b9ea1a64237aa69d807ee7adb7f : SUCCESS in 13m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea741e4bc35a4b43bfe3f6c6fc8321bd : SUCCESS in 7m 32s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/bc0d2da11c07473183b35be599959689 : SUCCESS in 6m 44s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/d4c69b365bd047669f76b5409c1b5739 : SUCCESS in 1h 16m 13s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/222ae4c8790c4206b07c34856c2975f3 : SUCCESS in 32m 34s (non-voting)\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/01df98933f664f6caf87fe9825d58c2c : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":3},{"id":"deff63b1f84a0b8cdde8435dd69e97bd5513ea3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2024-12-09 11:39:41.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"71fa815bc9630c3e060ff887cb4161b5eba2c980","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 13:14:57.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4dee99c96ccd40f482b745edaebcda12\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/f963182726234aea9be78cbc88f110d4 : SUCCESS in 7m 22s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/5c1a0beaf69f4196aeeef01eb0eb0709 : SUCCESS in 2m 20s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/baa09527cf37427f934974ce28aa6e46 : SUCCESS in 3m 03s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/64152cabb6e8475f80fafe577e19d375 : SUCCESS in 25m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9853ee9685044a3896d78c6bbdd4bf10 : SUCCESS in 6m 48s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/895eb9f56e1e4b03badcc98a7addac37 : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02943ad7cd48496d918b4117c8dd01ed : SUCCESS in 3m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/819bc9de118d4a8096ed976cc75048e4 : SUCCESS in 12m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/684178c1bc6c4a739808a78ee7fd2999 : SUCCESS in 6m 21s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/dc237ca741744054b5867268c1f85aef : SUCCESS in 6m 31s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/07422d1e6360436ebead299ce39f4f92 : SUCCESS in 1h 15m 59s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/ea9207a395ff418d8f025bbfe9dd58c6 : SUCCESS in 19m 33s (non-voting)\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e29ead29f5ac46bd89fa6fd6fbf1ac57 : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":4},{"id":"1d0c48c2a45f97452c2db794a0ac775c4191a3ad","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2024-12-16 12:22:56.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"90252c6e066deb0bcfdfb8cee2f1dabdaf221f3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2024-12-18 11:16:31.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"46d0e0a08cf29e1cf73e50f60336688231f2b4ee","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2024-12-18 11:18:26.000000000","message":"Patch Set 5: Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"21ef00b79a0481b6cbf8e34e9e58041e583c7dd3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2024-12-18 11:54:16.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Workflow-1\n","accounts_in_message":[],"_revision_number":6},{"id":"690b500938bf92dc5046f883cfb66e233add07f8","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2024-12-18 11:56:19.000000000","message":"Patch Set 6: Workflow-1","accounts_in_message":[],"_revision_number":6},{"id":"85d716defe5f1cd3d78508830fb7ac7ee0a2bb24","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2024-12-18 12:28:16.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"33ef26c5f4cfb6914765e528d1f5d865632ce9aa","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2024-12-18 12:30:43.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7fd544e6bd3822f0b37681b5058926aa13996f5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 13:25:09.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6452aa276ef8468c982d7d931033ec10\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/f8f0bd6f5403433ab22c8e7861be8d0a : SUCCESS in 5m 13s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/5cf74246c6dd41ffb3d09f2b2f23ae31 : SUCCESS in 3m 15s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/11470e3d24684b0f865b72d298102b95 : SUCCESS in 4m 09s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/a5b106aa3f794ce2b14ffb01d00191e9 : SUCCESS in 6m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/eb76f8b991474a01858fb533f11ef13e : SUCCESS in 13m 05s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/d5f2c53aaa634de7b967b365e3b28286 : SUCCESS in 14m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e084763d4ca241438d5350f614edae6d : SUCCESS in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fe1362f3fe254206a4b22fef2c1fcf0b : SUCCESS in 13m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8860c9ba181d41c4b0b25987902d6599 : SUCCESS in 6m 52s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/bfdc8f3604f44844b98c968f843281cd : SUCCESS in 6m 16s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/fe9d7fe35c5b4de0aca0dbfab0d1f748 : SUCCESS in 1h 08m 00s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/d46be48e4dee4f1c8c5201d0e02e7369 : SUCCESS in 21m 42s (non-voting)\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d34e07a5e5474e759d565b2635a0e568 : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":6},{"id":"6420746e870e2ec69cc24fde71eece33bb02042b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2024-12-18 16:03:07.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":7},{"id":"79f356ea297cfb71f2bdd2abfda3f45a38d1fe07","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2024-12-18 16:09:50.000000000","message":"Patch Set 7: Workflow-1","accounts_in_message":[],"_revision_number":7},{"id":"49fc65d43e4bc22c0c274e00beb87873f9d47be2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 17:42:17.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55534346fafd469fb0c0c2391df2ca1b\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/4482d91844b941cc8adaee03736f98e6 : SUCCESS in 6m 01s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/b7d369d1cd654b8d80a01ef6c004d92c : SUCCESS in 3m 01s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/1731fd8c4d804547b85ab7a6c5eefe2a : SUCCESS in 5m 04s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/a1853ff1f9134711960394f1bfff5662 : SUCCESS in 6m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e9b6f8b8ef754b9a97b28c6934df4679 : SUCCESS in 12m 09s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/32f261e977764120a8935b5e454b261d : SUCCESS in 14m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c896cb61c04a4e23aa53eba8e11c909d : SUCCESS in 7m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d7bfa271348c4d62ba0ec92d04bec5b9 : SUCCESS in 12m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44d5c936f7564c67af242fbc8934465f : SUCCESS in 6m 46s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/80347de5838b45c58b9c2ff8bc198a54 : SUCCESS in 6m 11s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/7c12fdd2b1e6497aa80c9e6fedd24ac4 : SUCCESS in 1h 04m 17s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/e66183918c6049f78a3aa0c8df217e44 : SUCCESS in 34m 02s (non-voting)\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ac70fd54919d4448ba4c24f96f43d4f7 : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":7},{"id":"23b7c02bce32e1b9139e7fd068d9dd14635f02ff","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2024-12-18 20:37:33.000000000","message":"Patch Set 7: -Workflow\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"d12e5a93e31cd80cb1b39156eebf36a467c26da2","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2024-12-19 08:36:54.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"665e5870754746bb30e5b29fab37d4da737f3a8a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 08:37:25.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"4fd3bcca8b55a2814c04f944114756d8f4149adf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 10:37:29.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f478bd72cc544129055455e32b922da\n\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/36162396969e409490fb20d583efbd77 : SUCCESS in 4m 17s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/b94811840fac43deab7ec78bcb5f961a : SUCCESS in 6m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d17b0543d9904af8868228b120c4b82c : SUCCESS in 11m 26s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/c3814636fd134f7fb829a581c5802d1c : SUCCESS in 16m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1c81d7878f44a6c967fd783b9f85c57 : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f79bc72e85c49f9b5a1df829c026b32 : SUCCESS in 10m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fd091a8ef0f4047a0b35e2ad81c2a80 : SUCCESS in 5m 57s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/238dece61eb340f59bd945b6e0da333b : SUCCESS in 5m 35s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/5bc2482dc4f7445cbdc176d34b2dcf14 : SUCCESS in 1h 24m 33s","accounts_in_message":[],"_revision_number":7},{"id":"b39de69229b922f578e60a989b4be26346df111b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 10:37:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"8fb650adbb935285169bb6d474b3d8e979843537","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 10:38:54.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da6a10b0f6064bb99e10eafff6696bba\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a29a864c083b44e1ab0e1dc46c4d7605 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"e1fde47a0c2c3a3d9d12747857cbf489158b32ad","revisions":{"58f76928e02e9d240cf824dc93e81fab5dda7a67":{"kind":"REWORK","_number":1,"created":"2024-11-29 02:07:48.000000000","uploader":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"ref":"refs/changes/64/936664/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/64/936664/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/64/936664/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/64/936664/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/64/936664/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/64/936664/1"}}},"commit":{"parents":[{"commit":"46a26e022a15175d9c8234b31370536e7531b6c8","subject":"[WIP] Configuring tests for different themes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/46a26e022a15175d9c8234b31370536e7531b6c8"}]}],"author":{"name":"Jan Jasek","email":"jjasek@redhat.com","date":"2024-11-29 01:59:50.000000000","tz":60},"committer":{"name":"Jan Jasek","email":"jjasek@redhat.com","date":"2024-11-29 02:07:19.000000000","tz":60},"subject":"[WIP] Configuring separated UI tests for different themes","message":"[WIP] Configuring separated UI tests for different themes\n\nChange-Id: I3ea3d0cf42dc8e89a23a763defe84389da56366b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/58f76928e02e9d240cf824dc93e81fab5dda7a67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/58f76928e02e9d240cf824dc93e81fab5dda7a67"}]},"branch":"refs/heads/master"},"f5b141f392cd25727dff9668c0fee06556cfce84":{"kind":"REWORK","_number":2,"created":"2024-11-30 01:45:23.000000000","uploader":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"ref":"refs/changes/64/936664/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/64/936664/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/64/936664/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/64/936664/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/64/936664/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/64/936664/2"}}},"commit":{"parents":[{"commit":"46a26e022a15175d9c8234b31370536e7531b6c8","subject":"[WIP] Configuring tests for different themes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/46a26e022a15175d9c8234b31370536e7531b6c8"}]}],"author":{"name":"Jan Jasek","email":"jjasek@redhat.com","date":"2024-11-29 01:59:50.000000000","tz":60},"committer":{"name":"Jan Jasek","email":"jjasek@redhat.com","date":"2024-11-30 01:42:46.000000000","tz":60},"subject":"[WIP] Configuring separated UI tests for different themes","message":"[WIP] Configuring separated UI tests for different themes\n\nChange-Id: I3ea3d0cf42dc8e89a23a763defe84389da56366b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/f5b141f392cd25727dff9668c0fee06556cfce84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/f5b141f392cd25727dff9668c0fee06556cfce84"}]},"branch":"refs/heads/master"},"d1031b330080c45a7e9d7afb718e387cd0230c0a":{"kind":"REWORK","_number":3,"created":"2024-12-06 17:47:30.000000000","uploader":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"ref":"refs/changes/64/936664/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/64/936664/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/64/936664/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/64/936664/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/64/936664/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/64/936664/3"}}},"commit":{"parents":[{"commit":"0c7e8fb37a09e98a9afbd2776ec67fbd0903f587","subject":"[WIP] Configuring tests for different themes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/0c7e8fb37a09e98a9afbd2776ec67fbd0903f587"}]}],"author":{"name":"Jan Jasek","email":"jjasek@redhat.com","date":"2024-11-29 01:59:50.000000000","tz":60},"committer":{"name":"Jan Jasek","email":"jjasek@redhat.com","date":"2024-12-06 17:45:37.000000000","tz":60},"subject":"[WIP] Configuring separated UI tests for different themes","message":"[WIP] Configuring separated UI tests for different themes\n\nChange-Id: I3ea3d0cf42dc8e89a23a763defe84389da56366b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/d1031b330080c45a7e9d7afb718e387cd0230c0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/d1031b330080c45a7e9d7afb718e387cd0230c0a"}]},"branch":"refs/heads/master"},"06f554a8a2e40215ba6e436918ad722f1efaea48":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":4,"created":"2024-12-09 11:39:41.000000000","uploader":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"ref":"refs/changes/64/936664/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/64/936664/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/64/936664/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/64/936664/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/64/936664/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/64/936664/4"}}},"commit":{"parents":[{"commit":"98ac275580c8788d76da5f15eaece9a7061ab6d0","subject":"Allow configuring tests for different themes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/98ac275580c8788d76da5f15eaece9a7061ab6d0"}]}],"author":{"name":"Jan Jasek","email":"jjasek@redhat.com","date":"2024-11-29 01:59:50.000000000","tz":60},"committer":{"name":"Jan Jasek","email":"jjasek@redhat.com","date":"2024-12-09 11:33:37.000000000","tz":60},"subject":"Allow configuring separated UI tests for different themes","message":"Allow configuring separated UI tests for different themes\n\nAllow configuring of locators/parts of the tests\nthat are theme specific, using a config file.\n\nChange-Id: I3ea3d0cf42dc8e89a23a763defe84389da56366b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/06f554a8a2e40215ba6e436918ad722f1efaea48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/06f554a8a2e40215ba6e436918ad722f1efaea48"}]},"branch":"refs/heads/master"},"78962dc4abfced146b200d0a3b7383b0824cc2f1":{"kind":"REWORK","_number":5,"created":"2024-12-18 11:16:31.000000000","uploader":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"ref":"refs/changes/64/936664/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/64/936664/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/64/936664/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/64/936664/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/64/936664/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/64/936664/5"}}},"commit":{"parents":[{"commit":"426e28773f31cce7f26ba910f2bbd2ac2cee0636","subject":"Allow configuring tests for different themes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/426e28773f31cce7f26ba910f2bbd2ac2cee0636"}]}],"author":{"name":"Jan Jasek","email":"jjasek@redhat.com","date":"2024-11-29 01:59:50.000000000","tz":60},"committer":{"name":"Jan Jasek","email":"jjasek@redhat.com","date":"2024-12-18 10:54:11.000000000","tz":60},"subject":"Allow configuring separated UI tests for different themes","message":"Allow configuring separated UI tests for different themes\n\nAllow configuring of locators/parts of the tests\nthat are theme specific, using a config file.\n\nChange-Id: I3ea3d0cf42dc8e89a23a763defe84389da56366b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/78962dc4abfced146b200d0a3b7383b0824cc2f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/78962dc4abfced146b200d0a3b7383b0824cc2f1"}]},"branch":"refs/heads/master"},"f5f073a3385ee33060c547013a6b6b2ef6c6d7c3":{"kind":"REWORK","_number":6,"created":"2024-12-18 11:54:16.000000000","uploader":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"ref":"refs/changes/64/936664/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/64/936664/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/64/936664/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/64/936664/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/64/936664/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/64/936664/6"}}},"commit":{"parents":[{"commit":"a87ad750ead2a79d0b00cc2dface9b44c2fbf091","subject":"Allow configuring tests for different themes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/a87ad750ead2a79d0b00cc2dface9b44c2fbf091"}]}],"author":{"name":"Jan Jasek","email":"jjasek@redhat.com","date":"2024-11-29 01:59:50.000000000","tz":60},"committer":{"name":"Jan Jasek","email":"jjasek@redhat.com","date":"2024-12-18 11:53:55.000000000","tz":60},"subject":"Allow configuring separated UI tests for different themes","message":"Allow configuring separated UI tests for different themes\n\nAllow configuring of locators/parts of the tests\nthat are theme specific, using a config file.\n\nChange-Id: I3ea3d0cf42dc8e89a23a763defe84389da56366b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/f5f073a3385ee33060c547013a6b6b2ef6c6d7c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/f5f073a3385ee33060c547013a6b6b2ef6c6d7c3"}]},"branch":"refs/heads/master"},"e1fde47a0c2c3a3d9d12747857cbf489158b32ad":{"kind":"REWORK","_number":7,"created":"2024-12-18 16:03:07.000000000","uploader":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"ref":"refs/changes/64/936664/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/64/936664/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/64/936664/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/64/936664/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/64/936664/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/64/936664/7"}}},"commit":{"parents":[{"commit":"a87ad750ead2a79d0b00cc2dface9b44c2fbf091","subject":"Allow configuring tests for different themes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/a87ad750ead2a79d0b00cc2dface9b44c2fbf091"}]}],"author":{"name":"Jan Jasek","email":"jjasek@redhat.com","date":"2024-11-29 01:59:50.000000000","tz":60},"committer":{"name":"Jan Jasek","email":"jjasek@redhat.com","date":"2024-12-18 16:03:04.000000000","tz":60},"subject":"Allow configuring separated UI tests for different themes","message":"Allow configuring separated UI tests for different themes\n\nAllow configuring of locators/parts of the tests\nthat are theme specific, using a config file.\n\nChange-Id: I3ea3d0cf42dc8e89a23a763defe84389da56366b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/e1fde47a0c2c3a3d9d12747857cbf489158b32ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/e1fde47a0c2c3a3d9d12747857cbf489158b32ad"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
