)]}'
{"id":"openstack%2Fhorizon~943821","triplet_id":"openstack%2Fhorizon~master~Ief0b619deefd7c18e882dee3932c5cd90209dbe4","project":"openstack/horizon","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2025-03-11 15:35:49.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"6914":{"account":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"last_update":"2025-03-11 15:16:37.000000000","reason":"\u003cGERRIT_ACCOUNT_6914\u003e replied on the change","reason_account":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"}},"35133":{"account":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"last_update":"2025-03-10 10:41:15.000000000","reason":"\u003cGERRIT_ACCOUNT_35133\u003e replied on the change","reason_account":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}}},"hashtags":[],"change_id":"Ief0b619deefd7c18e882dee3932c5cd90209dbe4","subject":"Improve key-pair description","status":"MERGED","created":"2025-03-07 19:21:59.000000000","updated":"2025-03-11 16:39:50.000000000","submitted":"2025-03-11 16:38:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"943821","meta_rev_id":"bcd51672bc7302ecd58c5f151c84a92523ee6ece","_number":943821,"virtual_id_number":943821,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-11 16:38:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"all":[{"value":2,"date":"2025-03-11 15:16:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-03-10 10:41:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},{"value":1,"date":"2025-03-08 13:27:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"all":[{"value":1,"date":"2025-03-11 15:16:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-07 20:54:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-08 13:27:17.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-03-08 13:27:28.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"state":"REVIEWER"},{"updated":"2025-03-08 13:27:28.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"state":"REVIEWER"}],"messages":[{"id":"17ac97c86da79ca032c4c6598ffd9818172980c3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-03-07 19:21:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d8d606b3525b2ed085f57442ae7d4c15128517a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-03-07 19:51:53.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"8ad1352508259f74704332e617ab8c6af5ac2e6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 20:54:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f4d1e39dc554e9bbc3f5408212c13b6\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/52c35e343a2e4d77910ac71c6c8eebf5 : SUCCESS in 2m 49s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/33c62d76dfe74e3c9be8ee7126e9ea68 : SUCCESS in 2m 04s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/1f968f85bf654aea9cbffd124f271e5a : SUCCESS in 4m 06s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/bbabcc6c89db4923986170273ff7f07c : SUCCESS in 5m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a5c8dfeaeb2f4008bd1a0c4e8d261797 : SUCCESS in 5m 04s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/d9a861f6890943c884c58761fd37440e : SUCCESS in 10m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/111e891aa91c4afb8896bb6440c96c6f : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb44cfa4c7fd433ba0a90e85f114eaf1 : SUCCESS in 14m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7096feb948ee4370843fb910690fecd9 : SUCCESS in 3m 28s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/df16d0ac4277473dba96161ba214deaa : SUCCESS in 6m 00s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/2c1f2194dc1f4b0c9453dc03950c5800 : SUCCESS in 1h 01m 07s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/cce368e697d64c36a5cc036ce0489574 : SUCCESS in 17m 34s (non-voting)\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/4f4e0cb6e9564e76b57463030c4c80bf : SUCCESS in 36m 32s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a15adb800a054d48aba6f765b7d76bde : SUCCESS in 2m 15s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/41dc7f0ef7fd4231a911a4b6e2966aa0 : SUCCESS in 19m 13s\n- horizon-dsvm-tempest-plugin-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d5aab62ca7ed4c178f792b9cf9a9862f : SUCCESS in 15m 42s","accounts_in_message":[],"_revision_number":2},{"id":"0d15e08c6d1143b97421e45f2181e63ab3924e84","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-03-07 22:21:01.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"c75e238d58ae612676e21789e93427aa13b8ee1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 23:12:48.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3eea07ee6ae3454a82850dac1a6e69ca\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/beb8fcca527b4f98a9bbd13169120382 : SUCCESS in 6m 44s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/98c85e4abf65459c9b4b9e56da1cbce0 : SUCCESS in 4m 39s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/c98e37fbf3064110b32e55bc15ba5eb0 : SUCCESS in 6m 09s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/6d625d6beeef433eb0d9ccd6e32c60a4 : SUCCESS in 7m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cc9c851d29a6409b80ffbb39c38f1c6d : SUCCESS in 13m 27s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/ec7c6fbbdac6407a8746c091677cfeba : SUCCESS in 13m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/610693e1a1fb4f98b3ba1feca7d65ca2 : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a5a5907136d4523a7993845125678c2 : SUCCESS in 10m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2f7547d557042279410f2a602cd2c18 : SUCCESS in 7m 34s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/8ac83614ca0548aa92e5250e8cc609e4 : SUCCESS in 3m 52s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/899576b0fb76491c99dff3b18504bb48 : FAILURE in 49m 50s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/5cdd61ed7b9c4999a58e7939085b21ed : SUCCESS in 14m 01s (non-voting)\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/e5af13143f644bc695981a478d1fb585 : SUCCESS in 13m 35s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f02dcb78e3f74609af7278af9eb7b50b : SUCCESS in 2m 12s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/c9580e491286438a8ef2e55c27e4445c : SUCCESS in 20m 08s\n- horizon-dsvm-tempest-plugin-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c8b3d1d78a904da9a2a75138926d274f : SUCCESS in 17m 05s","accounts_in_message":[],"_revision_number":3},{"id":"6226a56da5fa942d56059f08f9cf18a37332f117","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-03-07 23:26:35.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"218ec7cd99987850b22a14371c114037db216a35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-03-08 06:51:06.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"e99ba27a85ba0c9a6852d389c5a5da44cd86eeb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-08 07:54:03.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81eeec76e9e64771996333a7dff245ef\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/88e25319bddb443388a30ac4f24bd612 : SUCCESS in 8m 55s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/acc9b8ebb6f64e47b588794514b912b1 : SUCCESS in 3m 52s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/c5cb6d89c00540ceaceb6709d0ab8367 : SUCCESS in 6m 42s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/38866978491f4f0c8647fdaf96809f78 : SUCCESS in 6m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/67edc1735190460faaae8ace78441d24 : SUCCESS in 14m 09s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/b56fbd65c32d442aa126ea95b0d50cb4 : SUCCESS in 10m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84862e4d716a4f369e995561658457e8 : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77eb78925a4a491faa14ee52df943c9e : SUCCESS in 9m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/533ed8efad6246749adaa05eb9efbb7b : SUCCESS in 4m 35s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/ac906c3fe8674eb0afd6d4959add78cc : SUCCESS in 3m 59s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/a56d8d5d61144d91a06f5faa11fd5b39 : SUCCESS in 1h 01m 58s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/ca88404c4d704035b5e7bd404767256d : SUCCESS in 20m 18s (non-voting)\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/ec0af3f1b21a40d9bcec42922c4a6400 : SUCCESS in 16m 37s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cb6de1467abb43f999b3db233ef390ee : SUCCESS in 4m 48s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/2ea8a4e1442e46788367aa9a9fada40b : SUCCESS in 33m 09s\n- horizon-dsvm-tempest-plugin-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/968b3321844b47ea867d00b30c488e9c : SUCCESS in 30m 47s","accounts_in_message":[],"_revision_number":4},{"id":"1a0b840af9d4762086fbfd5b91c81a088cf51728","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-03-08 13:27:17.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"a40cfbbfa068d3b40a90556d75b64c4922ba83a1","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2025-03-10 10:41:15.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"c5cc8f54b5f9a70c210980cc08ccb113295bbc07","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2025-03-11 15:16:37.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"ed57b70e3e782523d38d6a8295943c4548f94329","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-11 15:17:12.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"9662879a29d86c7200d88d4a865bb63d8a36e7cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-11 16:38:42.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/774f66eecd9140948b9b0e7b7111ae76\n\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/5759c2f44bbe4f5eb0bf6fdcc0ee26ec : SUCCESS in 5m 55s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/00d688564b3540b5addfbc523809d634 : SUCCESS in 7m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/158ad42d7be442f7a8ad7fe2ea025503 : SUCCESS in 14m 40s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/0e5b814bf3914ac38ebea494ed5a3186 : SUCCESS in 7m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b034994ed2e434792131824d7bf456c : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd0d2f0c2e27401995e5066266152ae0 : SUCCESS in 6m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcea3430214b4d7c849eca9ae2aaead8 : SUCCESS in 7m 09s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/dca73f3e2e5f49f28686193e5ccd2fc1 : SUCCESS in 34m 04s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/f62d72816ac547c78a66f2533fd29775 : SUCCESS in 7m 16s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/200bb0f24507482fa50fb1834f9b1102 : SUCCESS in 1h 12m 41s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/e5401b6e4ac54c5091c00fa0fe5837d0 : SUCCESS in 35m 37s\n- horizon-dsvm-tempest-plugin-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d78b7c9aadf44bee9ddfd92a15ab2aa0 : SUCCESS in 26m 40s","accounts_in_message":[],"_revision_number":4},{"id":"b9d6ba037e1edb58f103a44742f8d9d3d0d94aa1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-11 16:38:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"bcd51672bc7302ecd58c5f151c84a92523ee6ece","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-11 16:39:50.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d35554c272014d6db8827a071b5de5c4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e43801b65a6431a8b2d7d543b038078 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"7c94e4af45c6913633e0e69156a994dca2293cef","revisions":{"64c3e002438bbc26c2811f0ecb3d173350280ef0":{"kind":"REWORK","_number":1,"created":"2025-03-07 19:21:59.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/21/943821/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/21/943821/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/21/943821/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/21/943821/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/21/943821/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/21/943821/1"}}},"commit":{"parents":[{"commit":"a11c3d316195f3d19c76c7040e312eb1869fe768","subject":"Don\u0027t modify Resource fields in place","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/a11c3d316195f3d19c76c7040e312eb1869fe768"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-03-07 19:21:59.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-03-07 19:21:59.000000000","tz":0},"subject":"docs: improve key-pair description","message":"docs: improve key-pair description\n\nImproving volume type extra specs description\n\nChange-Id: Ief0b619deefd7c18e882dee3932c5cd90209dbe4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/64c3e002438bbc26c2811f0ecb3d173350280ef0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/64c3e002438bbc26c2811f0ecb3d173350280ef0"}]},"branch":"refs/heads/master"},"a088a5247d5b7bd2c880f6bf05bbf81eba5ad05d":{"kind":"REWORK","_number":2,"created":"2025-03-07 19:51:53.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/21/943821/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/21/943821/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/21/943821/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/21/943821/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/21/943821/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/21/943821/2"}}},"commit":{"parents":[{"commit":"a11c3d316195f3d19c76c7040e312eb1869fe768","subject":"Don\u0027t modify Resource fields in place","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/a11c3d316195f3d19c76c7040e312eb1869fe768"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-03-07 19:21:59.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-03-07 19:51:36.000000000","tz":0},"subject":"desc: Improve key-pair description","message":"desc: Improve key-pair description\n\nImproving volume type extra specs description\n\nChange-Id: Ief0b619deefd7c18e882dee3932c5cd90209dbe4\nCloses-Bug: #2101181\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/a088a5247d5b7bd2c880f6bf05bbf81eba5ad05d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/a088a5247d5b7bd2c880f6bf05bbf81eba5ad05d"}]},"branch":"refs/heads/master"},"25f4b22f4b15c6e483e02023f7dc31e58a0a26c7":{"kind":"REWORK","_number":3,"created":"2025-03-07 22:21:01.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/21/943821/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/21/943821/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/21/943821/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/21/943821/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/21/943821/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/21/943821/3"}}},"commit":{"parents":[{"commit":"a11c3d316195f3d19c76c7040e312eb1869fe768","subject":"Don\u0027t modify Resource fields in place","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/a11c3d316195f3d19c76c7040e312eb1869fe768"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-03-07 19:21:59.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-03-07 22:20:27.000000000","tz":0},"subject":"Improve key-pair description","message":"Improve key-pair description\n\nImproving volume type extra specs description\n\nChange-Id: Ief0b619deefd7c18e882dee3932c5cd90209dbe4\nCloses-Bug: #2101181\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/25f4b22f4b15c6e483e02023f7dc31e58a0a26c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/25f4b22f4b15c6e483e02023f7dc31e58a0a26c7"}]},"branch":"refs/heads/master"},"7c94e4af45c6913633e0e69156a994dca2293cef":{"kind":"REWORK","_number":4,"created":"2025-03-08 06:51:06.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/21/943821/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/21/943821/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/21/943821/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/21/943821/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/21/943821/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/21/943821/4"}}},"commit":{"parents":[{"commit":"a11c3d316195f3d19c76c7040e312eb1869fe768","subject":"Don\u0027t modify Resource fields in place","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/a11c3d316195f3d19c76c7040e312eb1869fe768"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-03-07 19:21:59.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-03-08 06:50:42.000000000","tz":0},"subject":"Improve key-pair description","message":"Improve key-pair description\n\nImproving volume type extra specs description\n\nChange-Id: Ief0b619deefd7c18e882dee3932c5cd90209dbe4\nCloses-Bug: #2101181\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/7c94e4af45c6913633e0e69156a994dca2293cef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/7c94e4af45c6913633e0e69156a994dca2293cef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
