)]}'
{"id":"openstack%2Fhorizon~945083","triplet_id":"openstack%2Fhorizon~master~Iad931f88e58035cbc8fcde3cebba8af24640a3d5","project":"openstack/horizon","branch":"master","attention_set":{},"removed_from_attention_set":{"6914":{"account":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"last_update":"2025-06-03 20:24:03.000000000","reason":"\u003cGERRIT_ACCOUNT_6914\u003e replied on the change","reason_account":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"}},"8648":{"account":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"last_update":"2025-06-03 21:42:28.000000000","reason":"Change was submitted"},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2025-06-03 21:42:28.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-06-03 21:42:28.000000000","reason":"Change was submitted"},"35133":{"account":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"last_update":"2025-06-03 12:26:44.000000000","reason":"\u003cGERRIT_ACCOUNT_35133\u003e replied on the change","reason_account":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}}},"hashtags":[],"change_id":"Iad931f88e58035cbc8fcde3cebba8af24640a3d5","subject":"Fix security group renderring in port overview (4)","status":"MERGED","created":"2025-03-20 10:53:39.000000000","updated":"2025-06-03 21:43:57.000000000","submitted":"2025-06-03 21:42:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"945083","meta_rev_id":"e8de1bd5fac2c6ef0ea77ef10d7964bdadef1651","_number":945083,"virtual_id_number":945083,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-03 21:42:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"recommended":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"all":[{"value":1,"date":"2025-03-20 20:16:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":2,"date":"2025-06-03 20:24:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-06-03 12:26:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},{"value":1,"date":"2025-03-21 10:27:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":1,"date":"2025-06-03 20:24:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}],"reviewers":{"REVIEWER":[{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-20 12:15:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-20 20:16:57.000000000","updated_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2025-03-21 10:27:41.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-03-21 10:27:56.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"state":"REVIEWER"},{"updated":"2025-03-21 10:27:56.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"state":"REVIEWER"},{"updated":"2025-03-25 15:28:04.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"state":"REVIEWER"},{"updated":"2025-04-28 18:32:11.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"66bc6cf6e8fda2816950636b2f0b4db4fc48f67e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-03-20 10:53:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"87ed174e7f7cae59b6d12ce07e59da8135298c2f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-03-20 10:54:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0945fd85a73e73174b29721f5ce1c0d8e53d7baf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 12:15:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7f2baedda9954d9e934702c30c2b25a4\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/5b3e328401cd49a3b2ccead837934c67 : SUCCESS in 5m 46s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/ec9997c8888a424fb6024ad4f6d17aa8 : SUCCESS in 1m 50s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/9be686f2442c4c3fbde4bbbc3b6ac4d0 : SUCCESS in 3m 15s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/dc77fac24c984736ad5e7325106ff511 : SUCCESS in 5m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6d646850990c4fc080ceb8745f41209d : FAILURE in 5m 10s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/e117065d09f34d509ded0f0e93b17bcf : FAILURE in 10m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a01009841444bcd8797746f39de6bea : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c38edb09acbd42929de500e2eb78eb56 : FAILURE in 14m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/560fe8e13550421fa949702e9b68db2c : SUCCESS in 4m 15s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/bf7522316de34c5aabad9f74b072a53c : SUCCESS in 4m 58s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/25afcbc5fd4d41bfb33bb757686650f6 : SUCCESS in 1h 17m 23s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/cad30e673edd4b97bf5b7ebae1d39baf : SUCCESS in 11m 28s (non-voting)\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/ac68306774104aa487d3391ff829b5cd : SUCCESS in 20m 30s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9c7b6ddad8594d458569f3a8acfa81ee : SUCCESS in 2m 14s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/9ae98b6607f5451fb30909a7531f33ee : SUCCESS in 25m 28s\n- horizon-dsvm-tempest-plugin-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6942373bfef742099338d3ff3e67cb8a : SUCCESS in 22m 45s","accounts_in_message":[],"_revision_number":1},{"id":"dfc51a09ceb47fcec34d74b687986ff56f153129","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-03-20 12:27:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dc02f5e6a800367c407aba092a7bcd42274c1ca5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-03-20 16:02:34.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e8e5d1a3d91d220b6db969f584ce72a0351ff21f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-03-20 16:03:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7a1da6ddc045411c02a96113fef9e923a62662fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-03-20 16:04:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7fc6954f9b27681a5a8c908d50358824d3b0df5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 17:15:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d96248dcc9ff441981a4925550da684b\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/55c48c3166a24a96813867d8c0baa9d3 : SUCCESS in 6m 16s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/fb6d898251c34a1aadc71097496d6eaf : SUCCESS in 3m 44s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/996a03b75d2141eb8851924d68bc42c0 : SUCCESS in 4m 29s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/741e5056ca394f5f9f732100a28f97e3 : SUCCESS in 5m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2745874a72f74f679022ab83145d5235 : SUCCESS in 14m 45s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/6463ebc7464f4dedad28433d53ded747 : SUCCESS in 14m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0be4b5ee0e0b4d28b45ca43d550faba4 : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b5806d3e859c45cd8dff824e46bf37c5 : SUCCESS in 10m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb05092640db454390f42765c8725b74 : SUCCESS in 10m 02s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/30fc1293e82f447594282bba629aaf5a : SUCCESS in 6m 13s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/340108c164454f7790eaea8bab91b979 : SUCCESS in 1h 10m 14s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/dae27fba527e4610b3c429d55221d01e : SUCCESS in 21m 26s (non-voting)\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/4d0c2048edff4f22a5e91bfc3e9c5613 : SUCCESS in 22m 54s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/dbf05deba1a94a61bf4b063ff3321af0 : SUCCESS in 6m 25s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/3472b5b1d9684457b4e3677f3643ece5 : SUCCESS in 11m 51s\n- horizon-dsvm-tempest-plugin-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fdc29cf7e585413ebaac95e48aa6c27a : SUCCESS in 38m 52s","accounts_in_message":[],"_revision_number":3},{"id":"871beb23d16284714683b4fdfe9f252eb375755f","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-03-20 20:16:57.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"9733bfd7881625debbea545f0ff8737bb6bcbe39","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-03-20 20:36:17.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"470804298d2337d1b471657eec53c1f3064c2267","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-03-21 10:27:41.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"9031eed3c9fd1bbef59d2e3244c7d0535c0c44a6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-23 06:42:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b27587862875490f18a90b1b4fd597498d3fc2b2","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-23 17:55:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e6918f6ac2914a2dd13a1834fc71a51e87463436","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2025-06-03 12:26:44.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2c7282895102d92b0c8acbb6003156ffda281cfa","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2025-06-03 20:24:03.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"53e863f48f3ce4ae6e0d4b04c6afc6dcba9cf54a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 20:24:21.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2570a9418a7448240422dd0d5e9dbeefaa52c679","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 21:42:28.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ef49696e93041e08c1b9e41c7d48627\n\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/8b5894e9420d48758743099838c7b413 : SUCCESS in 3m 33s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/2aded45cc9ed49e3a4994383464c591e : SUCCESS in 4m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bc55d0b18de74ed2b32e5b34ac26b66b : SUCCESS in 10m 40s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/9c0579aa9eaf41aca94a04d4fb111e70 : SUCCESS in 14m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/306230f277634411bb9d89bc5d06e034 : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ac2b4750bfd2477d87a4859afe663d7a : SUCCESS in 13m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eaa34d1191574365a8da8a14f0de2459 : SUCCESS in 18m 35s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/b3426a451f354456bb3a6cfca8b8a1ac : SUCCESS in 27m 01s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/e66c5218ca804bac87f4b9fbdd976545 : SUCCESS in 5m 56s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/78e96f845e784650a2dcffc8b03f6af7 : SUCCESS in 1h 18m 02s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/5a6b4e02d3fe4791b305269f49863d00 : SUCCESS in 27m 06s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/1a4e8588eb864115abbdd5f3cc51178c : SUCCESS in 25m 14s\n- horizon-dsvm-tempest-plugin-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c9b0b6213c604d73afc0d873e20db0e9 : SUCCESS in 25m 30s","accounts_in_message":[],"_revision_number":3},{"id":"9fb098f3b26633d11e752bed37b8640ec09f4490","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 21:42:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"e8de1bd5fac2c6ef0ea77ef10d7964bdadef1651","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 21:43:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e3e5a13a8214c629e13d2cfd5961510\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/497a65f518fc4c3ab00b3cb8cc9e617c : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8dbaf2ab2787b9d0a2e9fdcd62479656d4e12f58","revisions":{"86a2be32bc93ded8f7d51ee1512c1500123e1605":{"kind":"REWORK","_number":1,"created":"2025-03-20 10:53:39.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/83/945083/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/83/945083/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/83/945083/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/83/945083/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/83/945083/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/83/945083/1"}}},"commit":{"parents":[{"commit":"adad6d69e7ad78d48a954ee0eacf6f2b91dd22d6","subject":"Merge \"Fix for Cancel button in Create Security Group form\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/adad6d69e7ad78d48a954ee0eacf6f2b91dd22d6"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-03-20 10:53:28.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-03-20 10:53:28.000000000","tz":60},"subject":"Fix security group renderring in port overview (4)","message":"Fix security group renderring in port overview (4)\n\nFor the port overview page context from the view is not used.\nInstead data to the _detail_overview template is passed from the\ntabs, where security_groups were not supplied historically.\n\nThis is a follow-up to [1]\n\n[1] https://review.opendev.org/c/openstack/horizon/+/938239\n\nCloses-Bug: #2092276\nChange-Id: Iad931f88e58035cbc8fcde3cebba8af24640a3d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/86a2be32bc93ded8f7d51ee1512c1500123e1605"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/86a2be32bc93ded8f7d51ee1512c1500123e1605"}]},"branch":"refs/heads/master"},"a92b0803c8c2f33afedffcf35f474a0aa9a7cc8f":{"kind":"REWORK","_number":2,"created":"2025-03-20 16:02:34.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/83/945083/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/83/945083/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/83/945083/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/83/945083/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/83/945083/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/83/945083/2"}}},"commit":{"parents":[{"commit":"adad6d69e7ad78d48a954ee0eacf6f2b91dd22d6","subject":"Merge \"Fix for Cancel button in Create Security Group form\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/adad6d69e7ad78d48a954ee0eacf6f2b91dd22d6"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-03-20 10:53:28.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-03-20 15:55:41.000000000","tz":60},"subject":"Fix security group renderring in port overview (4)","message":"Fix security group renderring in port overview (4)\n\nFor the port overview page context from the view is not used.\nInstead data to the _detail_overview template is passed from the\ntabs, where security_groups were not supplied historically.\n\nThis is a follow-up to [1]\n\n[1] https://review.opendev.org/c/openstack/horizon/+/938239\n\nCloses-Bug: #2092276\nChange-Id: Iad931f88e58035cbc8fcde3cebba8af24640a3d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/a92b0803c8c2f33afedffcf35f474a0aa9a7cc8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/a92b0803c8c2f33afedffcf35f474a0aa9a7cc8f"}]},"branch":"refs/heads/master"},"8dbaf2ab2787b9d0a2e9fdcd62479656d4e12f58":{"kind":"REWORK","_number":3,"created":"2025-03-20 16:04:59.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/83/945083/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/83/945083/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/83/945083/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/83/945083/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/83/945083/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/83/945083/3"}}},"commit":{"parents":[{"commit":"adad6d69e7ad78d48a954ee0eacf6f2b91dd22d6","subject":"Merge \"Fix for Cancel button in Create Security Group form\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/adad6d69e7ad78d48a954ee0eacf6f2b91dd22d6"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-03-20 10:53:28.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-03-20 16:04:46.000000000","tz":60},"subject":"Fix security group renderring in port overview (4)","message":"Fix security group renderring in port overview (4)\n\nFor the port overview page context from the view is not used.\nInstead data to the _detail_overview template is passed from the\ntabs, where security_groups were not supplied historically.\n\nThis is a follow-up to [1]\n\n[1] https://review.opendev.org/c/openstack/horizon/+/938239\n\nCloses-Bug: #2092276\nChange-Id: Iad931f88e58035cbc8fcde3cebba8af24640a3d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/8dbaf2ab2787b9d0a2e9fdcd62479656d4e12f58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/8dbaf2ab2787b9d0a2e9fdcd62479656d4e12f58"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
