)]}'
{"id":"openstack%2Fhorizon~949261","triplet_id":"openstack%2Fhorizon~master~Ia4b09da9f3482dd9a7d4aaaf661efc01e6f9a079","project":"openstack/horizon","branch":"master","topic":"2FA-TOTP-gen","attention_set":{},"removed_from_attention_set":{"6914":{"account":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"last_update":"2025-05-16 10:38:54.000000000","reason":"Change was submitted"},"8648":{"account":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"last_update":"2025-05-16 09:20:29.000000000","reason":"\u003cGERRIT_ACCOUNT_8648\u003e replied on the change","reason_account":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"}}},"hashtags":[],"change_id":"Ia4b09da9f3482dd9a7d4aaaf661efc01e6f9a079","subject":"Add TOTP credential generation","status":"MERGED","created":"2025-05-08 19:58:48.000000000","updated":"2025-05-16 10:40:41.000000000","submitted":"2025-05-16 10:38:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949261-2FA-TOTP-gen","meta_rev_id":"bedc133f9cd66fbf2bfd972944479b34c6015693","_number":949261,"virtual_id_number":949261,"owner":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-16 10:38:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"all":[{"value":2,"date":"2025-05-16 09:20:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-05-15 15:38:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"all":[{"value":1,"date":"2025-05-16 09:20:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}],"CC":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-08 21:12:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-15 15:38:28.000000000","updated_by":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"reviewer":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"state":"REVIEWER"},{"updated":"2025-05-15 15:40:45.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2025-05-15 15:40:45.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"state":"REVIEWER"}],"messages":[{"id":"239f576fe7332b2ef743cbc1d3d3b8254c11a7f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2025-05-08 19:58:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"322bf4e803b60bcd2341e5d78821cb570ff4f4b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 21:12:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c562f8a799274ce3bf58c08cd537ca64\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/f120e491b7214c1aa343bc36d4b836f1 : SUCCESS in 6m 08s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/7ff22aec02734e2aae5d7b5b62e17a19 : SUCCESS in 3m 20s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/01a33d6cbb314a2e8f2bbb3c9fbc84c8 : SUCCESS in 4m 16s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/70b5fd05cb5e467abb8541cb330753ad : SUCCESS in 4m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cd4e57439e174348bb10656068292cf0 : SUCCESS in 11m 48s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/0a9e796d6034499da76505a26e684eda : SUCCESS in 8m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65ae85426ed3483ead11314ec946a8c2 : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e01b4bcedc4e49ddb6aa871d17fd2459 : SUCCESS in 7m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3883ec55bd014bf182f0f3dbaa61ed73 : SUCCESS in 17m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/182fe6ecbdd54e5bb49c7d49bbfc6e2e : SUCCESS in 15m 14s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/afffe28039e94c208e19194c21706d28 : SUCCESS in 6m 07s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/7ee6ca886be94470b03b8f51cfcf7cb8 : FAILURE in 1h 10m 16s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/4e40e1c751b64acaa3064e0e268c157d : SUCCESS in 19m 27s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/743ed35db2e840f58b22f71687eb0838 : SUCCESS in 27m 57s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9308b095d39d48a09672427b3768dfe3 : SUCCESS in 4m 24s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/5b08685a2af640e398559ade7ce0298a : SUCCESS in 21m 43s\n- horizon-dsvm-tempest-plugin-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1f709e716d4e44049532231651fb83f2 : SUCCESS in 18m 47s","accounts_in_message":[],"_revision_number":1},{"id":"8e3410b7580ba040b81c632937af5a3529d68548","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2025-05-08 22:32:11.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"00f843994e059997b2cadead7431521e6dc8d9d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 23:37:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee4c5a1ebb944eb4929148dba3b93738\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/3388467da8ba428e96bf10886bfa7bdc : SUCCESS in 4m 42s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/aaf8aa2b52814b5dbc17cfa170d8e074 : SUCCESS in 2m 48s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/0f303020efd14965812d02679c3ac67c : SUCCESS in 3m 38s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/92d78bd4bd6a4cc3ab0109305aeda2d3 : SUCCESS in 4m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d0bbe83eb3e14d2e84c7b941896b8e38 : SUCCESS in 5m 28s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/24d7ca1eb7144baab2752941a375699c : SUCCESS in 10m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6146c12b5eb446c19db8cccd70e942bb : SUCCESS in 2m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3fe49ee2b6a144138d13d3ef991d99f1 : SUCCESS in 9m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4fb6a6215b884ab59342e7d6f43aed52 : SUCCESS in 13m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4954817125094b4391d791d94023253c : SUCCESS in 11m 31s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/68356703cedf4ce4870cde77e93ae5b4 : SUCCESS in 3m 49s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/ff8bd2cea9d94ef58e8bafe610f8aa34 : SUCCESS in 1h 04m 04s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/e81ffea887f8445aa2dec8556f27d2b3 : SUCCESS in 26m 46s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/1b45a4a2776448b5bb93c498d42d46b0 : SUCCESS in 28m 46s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c848cc4f8d944412b1f6f15c789d7198 : SUCCESS in 3m 38s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/1ec11be7c39e42d788c7907a3ec1eff3 : SUCCESS in 20m 20s\n- horizon-dsvm-tempest-plugin-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9a9b0258eb7d44c19007a0e391a8d1e6 : SUCCESS in 17m 40s","accounts_in_message":[],"_revision_number":2},{"id":"7543fae95cccd4ef50d0f55f52f618be1f1b843e","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2025-05-15 15:38:28.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ca90534aa289d2da755d40e2df26c29a654ea6a4","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2025-05-16 09:20:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ec09be7934cd8a9c932a735c721d5a44a2223587","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 09:20:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1a116dc8f95ad243c97572f722a396c676a5827b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 10:38:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5b421809f1449e7816ac324836c6e0e\n\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/5be8edd0a2554342802751c00358fc58 : SUCCESS in 4m 56s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/7141e9c9033d45fc805732b3560faf8c : SUCCESS in 6m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/72be87cfa4ac4bc4bfdad6384edce9a8 : SUCCESS in 14m 42s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/394b6f454a4942c4b12052dd24773ef1 : SUCCESS in 14m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5893f4057e534cef9f6c856e9a1a9727 : SUCCESS in 2m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5b50c439696644a6a74e327c39f3d754 : SUCCESS in 13m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f3f9dcad7e547a9a35c98ed055cbdb6 : SUCCESS in 14m 37s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/d96580ff85e8474baa3d1b1ef7c34496 : SUCCESS in 31m 42s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/ff7021d22a0b4b278554731bccb3c764 : SUCCESS in 9m 42s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/39a9cc10d1e34f699c3460fe64090532 : SUCCESS in 1h 13m 40s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/ccf6db4cc7504fb586693119059ad42e : SUCCESS in 12m 55s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/b7777edb03bc4ceb88bf60747b3bedac : SUCCESS in 27m 51s\n- horizon-dsvm-tempest-plugin-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/34f370631a884fdeb87021ef2fe268ed : SUCCESS in 24m 38s","accounts_in_message":[],"_revision_number":2},{"id":"a8df27b95f03954774fc4c5590f056e73c2e5a7e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 10:38:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"bedc133f9cd66fbf2bfd972944479b34c6015693","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 10:40:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a07d7133ad146efb25fbabfb3a5444a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2689680507444918b308a57f74125d7e : SUCCESS in 1m 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f30764b03403284f88215ae4e11640047bf33adb","revisions":{"bc14b874cb6e310c3f0c2b5d57e3926c2924117a":{"kind":"REWORK","_number":1,"created":"2025-05-08 19:58:48.000000000","uploader":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"ref":"refs/changes/61/949261/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/61/949261/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/61/949261/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/61/949261/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/61/949261/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/61/949261/1"}}},"commit":{"parents":[{"commit":"349bdf83de8ce0d0aba0e996a3e1c46b5544a5eb","subject":"Merge \"RBAC Policies - fix to display title correctly\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/349bdf83de8ce0d0aba0e996a3e1c46b5544a5eb"}]}],"author":{"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","date":"2025-05-08 19:56:28.000000000","tz":-300},"committer":{"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","date":"2025-05-08 19:56:48.000000000","tz":-300},"subject":"Add TOTP credential generation","message":"Add TOTP credential generation\n\nThis patch adds autogeneration of TOTP credential.\n\nChange-Id: Ia4b09da9f3482dd9a7d4aaaf661efc01e6f9a079\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/bc14b874cb6e310c3f0c2b5d57e3926c2924117a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/bc14b874cb6e310c3f0c2b5d57e3926c2924117a"}]},"branch":"refs/heads/master"},"f30764b03403284f88215ae4e11640047bf33adb":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-05-08 22:32:11.000000000","uploader":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"ref":"refs/changes/61/949261/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/61/949261/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/61/949261/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/61/949261/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/61/949261/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/61/949261/2"}}},"commit":{"parents":[{"commit":"b3c881cfbacb7d3f2af6ac92ec745c3edaa3d40f","subject":"Merge \"Create QoS Policy - add description\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/b3c881cfbacb7d3f2af6ac92ec745c3edaa3d40f"}]}],"author":{"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","date":"2025-05-08 19:56:28.000000000","tz":-300},"committer":{"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","date":"2025-05-08 22:32:11.000000000","tz":0},"subject":"Add TOTP credential generation","message":"Add TOTP credential generation\n\nThis patch adds autogeneration of TOTP credential.\n\nChange-Id: Ia4b09da9f3482dd9a7d4aaaf661efc01e6f9a079\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/f30764b03403284f88215ae4e11640047bf33adb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/f30764b03403284f88215ae4e11640047bf33adb"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
