)]}'
{"id":"openstack%2Fhorizon~949656","triplet_id":"openstack%2Fhorizon~master~I269aef21baee6eb4c79da9fd96fff657298ac634","project":"openstack/horizon","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2025-05-14 11:11:15.000000000","reason":"removed on reply"},"6914":{"account":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"last_update":"2025-11-26 20:27:56.000000000","reason":"\u003cGERRIT_ACCOUNT_6914\u003e replied on the change","reason_account":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"}},"8648":{"account":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"last_update":"2025-11-26 21:36:30.000000000","reason":"Change was submitted"},"35133":{"account":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"last_update":"2025-07-23 11:39:13.000000000","reason":"\u003cGERRIT_ACCOUNT_35133\u003e replied on the change","reason_account":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-11-26 21:36:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I269aef21baee6eb4c79da9fd96fff657298ac634","subject":"remove unicode from files and update configuration for documentation","status":"MERGED","created":"2025-05-13 17:06:54.000000000","updated":"2025-11-26 21:37:52.000000000","submitted":"2025-11-26 21:36:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949656","meta_rev_id":"8d44bb3113f9698af98e46a017ffd24c807484c0","_number":949656,"virtual_id_number":949656,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-26 21:36:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"recommended":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"all":[{"value":1,"date":"2025-05-14 11:11:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":2,"date":"2025-11-26 20:27:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-07-23 11:39:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":1,"date":"2025-11-26 20:27:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"}],"reviewers":{"REVIEWER":[{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-13 18:28:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-13 18:31:51.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"state":"REVIEWER"},{"updated":"2025-05-13 18:31:51.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"state":"REVIEWER"},{"updated":"2025-05-13 18:31:51.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2025-05-20 14:33:13.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"state":"REVIEWER"}],"messages":[{"id":"dc2252cfee2cebe919b1d5d59644108b078e4e0c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-13 17:06:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5736fcbbfe39be0fd700f0e742a4cb9f4cd3c379","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-13 17:23:06.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"fdac79dedcdfd890ff87510335b915a75c3cd20a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 18:28:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1db62010fbc342e1996fbf0b4079b20d\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/4487a61261814327a54554bfbd975c3c : SUCCESS in 3m 43s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/f92ec123a2684da483e736c5bc167b20 : SUCCESS in 3m 29s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/71c5a5d8fd8d4e5983d528f27a23e0e7 : SUCCESS in 3m 48s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/af1d34dbc3f64300be88616f6b20a54f : SUCCESS in 5m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b793f1cb1d9643a4a9673f32bb4c6d48 : SUCCESS in 5m 08s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/a44d352b796d4e9aa03088f47c67cbad : SUCCESS in 11m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85bfbe907a504a7e91b44577ff4637e7 : SUCCESS in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/44822d2a0deb45fc9318dcd19def9754 : SUCCESS in 10m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/84502bdad0c841c38b49a8ce19b24a7e : SUCCESS in 18m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22e70c8848e54770bcaf18d0cb474a91 : SUCCESS in 12m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b538f5b97b9546eda43bd96e609ee173 : SUCCESS in 7m 24s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/39031866d7b44c659bef98af6e32a7a2 : SUCCESS in 5m 32s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/c60e7238bec8425db8c1cdd972168c97 : SUCCESS in 1h 04m 27s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/dc93ff98ca854f6b959db34f40239878 : SUCCESS in 29m 57s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/f01d62ffee664aecb8e150b18e3a7b92 : SUCCESS in 11m 12s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/21eee15e972148aaa43658342f7657f8 : SUCCESS in 3m 46s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/7d7e68e659724599aa15a8b05c6faba2 : SUCCESS in 20m 42s\n- horizon-dsvm-tempest-plugin-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7419fc310baf4b1db3d2e3e2b2de76ce : SUCCESS in 13m 53s","accounts_in_message":[],"_revision_number":2},{"id":"37d55b1e4fff51c1c4d5e9e4dac2e633bd8c4dcc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-13 18:31:39.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"8e29c664735138d779f325565d0a307f8bccbe58","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-13 18:31:51.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"44756786d43a813dc07d78d0be1f96e3d327db78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 19:43:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbea2f30f05b4590a2febd175dcff932\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/14ee0c51cd3b40f4934fabdcc2d1ced0 : SUCCESS in 5m 25s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/7bd44af6ce6c46f381a0a2639655f063 : SUCCESS in 4m 21s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/4150c9dd2d944e0fb5d6e9e94c3ce2bf : SUCCESS in 5m 34s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/a5cd20dfdb044598855918f2d903b290 : SUCCESS in 6m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a49acf3563e14e8795872683213c4ea0 : SUCCESS in 10m 55s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/7fb6e020e5cd49cfaaa862664378ac94 : SUCCESS in 14m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de233310d22a441e9e33b254080c675f : SUCCESS in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d33770615c2643ceacc33bea928172cf : SUCCESS in 15m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/057dda323f2d403ba67923ee0a7cd4dd : SUCCESS in 14m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5475b26a6f7645588416157e7fef8052 : SUCCESS in 20m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e018b6d77da144ee98939705252a164a : SUCCESS in 11m 44s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/6dc9bf85d3da418b97040b3101aa141f : SUCCESS in 5m 52s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/d0d4c86dd1af42ae9c79f0a3ac882b8b : SUCCESS in 1h 05m 22s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/1d40bff7ddb1449fa0fd9cc3056013a9 : SUCCESS in 19m 27s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/12cab848998b4425b8d835ea36377356 : SUCCESS in 26m 40s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/215af86fc9114b01b057bc51cb90a3c3 : SUCCESS in 2m 07s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/3573f97839444ef783b7eb4ee31bc2df : SUCCESS in 12m 35s\n- horizon-dsvm-tempest-plugin-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a68010afa8cb44b8bcdb99571850fc0a : SUCCESS in 18m 29s","accounts_in_message":[],"_revision_number":3},{"id":"89b560cc23c6026e5b3acea8a75552f11996725f","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-05-14 11:11:15.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"3cb9e103fc3c0bca04f1c16397718416d3f4f382","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2025-07-23 11:39:13.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0c14d57dafac25cd724970cc4dfc9895ef4316e5","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-11-20 16:14:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"767980ffee9d6f9a3762a148388217f84dbaf4e8","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2025-11-26 20:27:56.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"1b16b303a9ae52015961462085baa96f752dc29e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 20:28:18.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"642cf31327740f94ad78d8c35af70dbb1107ffde","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 21:36:29.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd083ef80c654bd682d1a9a84ceaf4b7\n\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/925e9e627e854fa3a542e3444a05ae77 : SUCCESS in 4m 53s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/2e42620ab8044e779c641ecaa68efdfd : SUCCESS in 5m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ccb885e3877c473799ad0733bc5aabd5 : SUCCESS in 8m 14s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/0a4bf94ee80d43658e4c08b22f183cf5 : SUCCESS in 18m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f4bcf93364f42d78e62f15038ce75e2 : SUCCESS in 4m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8859f64830364bafb529acb85779d214 : SUCCESS in 14m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b12eb0bc75cb435b8aa7724f01157d20 : SUCCESS in 11m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1ace9735fe141f8b39154cff6e44139 : SUCCESS in 7m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ac78de4c7154246ac03e3fd6aa14503 : SUCCESS in 11m 56s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/15ad7021419241df8c5ee660003c101c : SUCCESS in 14m 50s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/437fcce0b4024f8ab7faa98b1023dd08 : SUCCESS in 4m 43s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/dfc6af8053944662bab598a9170d63d4 : SUCCESS in 1h 07m 36s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/f9c1e4d47d7f44d8bc1176e1c7dc4296 : SUCCESS in 30m 04s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/20855fa880904c87a0a5ee322ca7b7a2 : SUCCESS in 12m 16s","accounts_in_message":[],"_revision_number":3},{"id":"6744d0d828082264591d6f181e4f42361939b3de","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 21:36:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"8d44bb3113f9698af98e46a017ffd24c807484c0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 21:37:52.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab2c6b1add0346008a2c38529ca46cc8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ef22df64e9e44df8eb672a78a4e1bf6 : SUCCESS in 1m 13s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f757b3b34a474688bcb5f3886a0e6d0c : SUCCESS in 48s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8bf595775a169a4c4911e4a30720a0c5d9b27543","revisions":{"f2fc7a524400202b730e17f09972ac1cad09ba3f":{"kind":"REWORK","_number":1,"created":"2025-05-13 17:06:54.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/56/949656/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/56/949656/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/56/949656/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/56/949656/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/56/949656/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/56/949656/1"}}},"commit":{"parents":[{"commit":"b3c881cfbacb7d3f2af6ac92ec745c3edaa3d40f","subject":"Merge \"Create QoS Policy - add description\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/b3c881cfbacb7d3f2af6ac92ec745c3edaa3d40f"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-13 17:06:54.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-13 17:06:54.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I269aef21baee6eb4c79da9fd96fff657298ac634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/f2fc7a524400202b730e17f09972ac1cad09ba3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/f2fc7a524400202b730e17f09972ac1cad09ba3f"}]},"branch":"refs/heads/master"},"c5dbf01bfc20c46d4719191f71a2d6aa38ba5eb3":{"kind":"REWORK","_number":2,"created":"2025-05-13 17:23:06.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/56/949656/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/56/949656/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/56/949656/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/56/949656/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/56/949656/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/56/949656/2"}}},"commit":{"parents":[{"commit":"b3c881cfbacb7d3f2af6ac92ec745c3edaa3d40f","subject":"Merge \"Create QoS Policy - add description\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/b3c881cfbacb7d3f2af6ac92ec745c3edaa3d40f"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-13 17:06:54.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-13 17:23:01.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I269aef21baee6eb4c79da9fd96fff657298ac634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/c5dbf01bfc20c46d4719191f71a2d6aa38ba5eb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/c5dbf01bfc20c46d4719191f71a2d6aa38ba5eb3"}]},"branch":"refs/heads/master"},"8bf595775a169a4c4911e4a30720a0c5d9b27543":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-05-13 18:31:39.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/56/949656/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/56/949656/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/56/949656/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/56/949656/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/56/949656/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/56/949656/3"}}},"commit":{"parents":[{"commit":"b3c881cfbacb7d3f2af6ac92ec745c3edaa3d40f","subject":"Merge \"Create QoS Policy - add description\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/b3c881cfbacb7d3f2af6ac92ec745c3edaa3d40f"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-13 17:06:54.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-05-13 18:31:39.000000000","tz":0},"subject":"remove unicode from files and update configuration for documentation","message":"remove unicode from files and update configuration for documentation\n\nChange-Id: I269aef21baee6eb4c79da9fd96fff657298ac634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/8bf595775a169a4c4911e4a30720a0c5d9b27543"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/8bf595775a169a4c4911e4a30720a0c5d9b27543"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
