)]}'
{"id":"openstack%2Fhorizon~950271","triplet_id":"openstack%2Fhorizon~master~I07198f7201296167cdec136d265868091074c2ce","project":"openstack/horizon","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2025-08-15 01:53:09.000000000","reason":"Change was submitted"},"6914":{"account":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"last_update":"2025-08-15 01:53:09.000000000","reason":"Change was submitted"},"8648":{"account":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"last_update":"2025-05-19 14:25:50.000000000","reason":"\u003cGERRIT_ACCOUNT_8648\u003e replied on the change","reason_account":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"}},"35133":{"account":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"last_update":"2025-08-14 13:34:26.000000000","reason":"\u003cGERRIT_ACCOUNT_35133\u003e replied on the change","reason_account":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-05-19 12:29:31.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"I07198f7201296167cdec136d265868091074c2ce","subject":"Language selection in config","status":"MERGED","created":"2025-05-18 21:35:56.000000000","updated":"2025-08-15 01:56:11.000000000","submitted":"2025-08-15 01:53:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950271","meta_rev_id":"221d8400501878ab1636c0ca8ff890d9f36e0520","_number":950271,"virtual_id_number":950271,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-15 01:53:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"all":[{"value":2,"date":"2025-05-19 14:25:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-08-14 13:34:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},{"value":1,"date":"2025-05-19 14:08:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"all":[{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-08-14 13:34:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"}],"reviewers":{"REVIEWER":[{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-18 22:41:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-19 08:13:18.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2025-05-19 14:08:03.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-05-19 14:08:16.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"state":"REVIEWER"},{"updated":"2025-05-19 14:08:16.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"state":"REVIEWER"},{"updated":"2025-05-19 14:08:16.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"state":"REVIEWER"}],"messages":[{"id":"3c60f1a772352ae5936e39447dc1dccf6405e4f2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-05-18 21:35:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0df6acde82c95303305c96658da70e8820411f0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-18 22:41:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/11b4916327ad4ca4a7ab9aaede7a00e1\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/339ff5f80f4045f6bd2775c985d6ca0b : SUCCESS in 1m 54s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c09bf99c8ad14e9eb56b02537fd44d32 : SUCCESS in 1m 58s\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/9e6bb84aee4347138f4abf22f2c1ef69 : SUCCESS in 5m 44s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/ed35d594253a4da399f5ac7185e4d608 : SUCCESS in 1m 46s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/b09e78f1e8d44cf189635ae8912db824 : SUCCESS in 2m 26s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/33eedbe853eb42c7a35a89232396639c : SUCCESS in 3m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ecfe10ad91be4978b6291c69f9aa91c8 : SUCCESS in 5m 19s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/d9a0b4b3184b4cd783155f2ab1893b6a : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f226d765b4814dce8ba2d655d655589a : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/520917480e0445dbabfb39bf9b103a8d : SUCCESS in 4m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/06ef965b58bb467fb9c18b3951e75437 : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5bd7447f3524969a47dbca4c0b3b26e : SUCCESS in 7m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/acd2bad5be7243cd843c2f4f201a887c : SUCCESS in 7m 40s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/317fe0bde5794b5392f0248e5db4590c : SUCCESS in 3m 42s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/eae8b355de7948df9843a5b4e189ef1e : FAILURE in 48m 54s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/c1e99f57798c4438b6805b7e36324606 : SUCCESS in 20m 13s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/83294046a2d84ab5a343ffdb374e8ca3 : SUCCESS in 18m 51s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/713d98cd2c3f499ab95786f7f8f75a5e : SUCCESS in 2m 07s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/210362284b4b410ab29a1b25340ed69a : SUCCESS in 19m 06s\n- horizon-dsvm-tempest-plugin-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f360a9b78caf405ba2f592c58f7a66d6 : SUCCESS in 16m 37s","accounts_in_message":[],"_revision_number":1},{"id":"60b5886af1a2e4b002877464e445aa69d6b639c8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-05-18 22:58:06.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"96612a12860c334084b2c03cffb18655d246d59a","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-05-18 23:05:38.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"92082e3c2868a9eabd085a957d419e7b89211704","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-05-18 23:05:49.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"bf06a2b3876b7c0b0b4205fab30ea95f5d16e8e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 00:37:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac794f912bbd4b71a98b2c9d704efe70\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/8be6423773ae4c7abebb288eba6676af : SUCCESS in 2m 49s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/3e99d767fe8349e7b6fe0387de71baae : SUCCESS in 1m 45s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/329ab6aa0eb14beabc854623c9e16bc4 : SUCCESS in 2m 14s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/78a4ac3d59e64ad78285ff425d156c48 : SUCCESS in 2m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4fef3c2b33ce49debf89047a4add2a02 : SUCCESS in 9m 19s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/dee97d99ece44c42a5d27eb605243dca : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd27ae5c725c4202a08df5c089a92614 : SUCCESS in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6e6fc68d44d74d1cb7be835e8b79033e : SUCCESS in 9m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5044553e8a454811a6ac3834f0225b34 : SUCCESS in 12m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b398286cfa249e3bbdfe20483e97b1d : SUCCESS in 7m 25s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/acf2d06605a441e786757468d027db10 : SUCCESS in 4m 47s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/0b3e94d9802548258bf78cf589e2cdb7 : SUCCESS in 1h 03m 38s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/5c417792e45343d1a1afe968d49a7dd6 : SUCCESS in 16m 55s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/10f1d66a245649ccba68331319c8e5fc : SUCCESS in 10m 57s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0c3cf151b7d14927af6a01fd1dfba006 : SUCCESS in 3m 11s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/9ea6a021fe8d4db6967870f62e91a89e : SUCCESS in 11m 33s\n- horizon-dsvm-tempest-plugin-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c4e0ac6a61904b18b3427b0eb7635f74 : SUCCESS in 18m 11s","accounts_in_message":[],"_revision_number":3},{"id":"8b95fa6ea7747f04d79f72e89c9d576512ae150d","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-19 08:13:18.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"96cbca2c2d31fddaa43daeaa2d49cf9b28f136e4","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-19 08:13:36.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"46dbf5288c993204e9dccaa408894ac76a179ac1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-05-19 09:18:53.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"ab25c13bfb90d18f7167e12f4ed1bec14c538e6a","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-05-19 09:19:40.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9f7d5d34f2b1b917803fda965fabd00db459f23b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 10:28:24.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/19aa5c9c07b843c5b7184a23799e101e\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/1386316de63f45c1b4a44ed531da0557 : SUCCESS in 5m 18s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/be1d9d9a8e2d4913aeb5abf2fd36b893 : SUCCESS in 3m 10s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/05ba953cf335498986e4368106a09a0f : SUCCESS in 4m 27s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/6ba1ad7363ff4580a289afc571fb6a92 : SUCCESS in 5m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/eccb39a01492485bb09a2632970f695f : SUCCESS in 14m 31s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/7f3499caeebd4e048bbd28b78f5c3c2c : SUCCESS in 14m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2663a1dcec04de0a854694d93d24f3e : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/066f03aacbbd4a52a0ebfd8accbd475e : SUCCESS in 13m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f9810e6ee6e9493389af6aa39a0eef0d : SUCCESS in 17m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41f2b6c6c279435aa4241d8d48d207bc : FAILURE in 5m 42s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/b4dc76bb91044804b944bab17e0926da : SUCCESS in 6m 52s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/3ac271fec503462687fec288ce4de019 : SUCCESS in 1h 09m 06s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/34ff970c120349b5a5dd3cc408a93086 : SUCCESS in 21m 28s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/2eae67f1fd74499383bd7864102be95f : SUCCESS in 22m 40s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/33582dec9bc64709a746b7d0e7f6fecc : SUCCESS in 3m 56s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/25e6245badb24d708a64d3bf07e0c7e1 : SUCCESS in 21m 56s\n- horizon-dsvm-tempest-plugin-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dec2202aaf564518b8a24fc0964ab6e6 : SUCCESS in 26m 21s","accounts_in_message":[],"_revision_number":4},{"id":"c1d56233887897580d76e0c56891ea60714fb2c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-05-19 11:27:24.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"e376fe6817787e09264b2b5f3f8e9ce589797214","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-05-19 12:29:22.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"57336961abe2ae19d262b502bc83aaded8092012","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-05-19 12:29:31.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":6},{"id":"8ba5235c02ad0d3e7e261199488268da9e2cb354","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-05-19 12:57:07.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"ea117c105a885db5a0889d4f13b84e4e3d5feb62","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-05-19 13:41:46.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":7},{"id":"49640e5921685a911a83e3472c8654159472662b","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-19 14:08:03.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"eec691a3814bee444dd5f6529fbc09e9bde7800b","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2025-05-19 14:25:50.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"01bc07c4df8ae2b5be75382e1adf9baaf830d772","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 14:32:49.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86a5c8d3977e4c68b9c2d98a56c288b5\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/73de84e7fe464b6790f7bdd0fb278983 : SUCCESS in 6m 44s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/1f89ed558f804b3fb096dc8889919c7c : SUCCESS in 3m 28s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/e9eec70821f847678e6ff445ffc8df49 : SUCCESS in 4m 40s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/44461f39c4be44a182e0f35c9609b5c7 : SUCCESS in 6m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ca2ca2e452ba4f2996905290003c834b : SUCCESS in 14m 18s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/59a49dbc5ad842f59908243f79a79518 : SUCCESS in 14m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28da457750014b69aee46ba73837e1f4 : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/44c8c8a66bda49d4a363a00885421e8b : SUCCESS in 6m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fb02ab49b96f4793a5da684883e874dc : SUCCESS in 18m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0199d4d327784274bd356464a17e5b26 : SUCCESS in 14m 16s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/fb5d365407f14504b0a36209642bcc9d : SUCCESS in 6m 11s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/a53c781823184dfba172143425f90224 : SUCCESS in 1h 19m 43s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/cd8da9278faa4ff787548ad7c15d2f65 : SUCCESS in 23m 52s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/e28825fe0ea04b2596be790aaa8231d4 : SUCCESS in 27m 06s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/246a13fd7f474eb2a32516cf84951dc5 : SUCCESS in 5m 41s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/3ca8b8aab47f4744bff660be71db02a8 : SUCCESS in 27m 44s\n- horizon-dsvm-tempest-plugin-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/85457c212d6a4aaeb0c670d59ce3a4b5 : SUCCESS in 27m 36s","accounts_in_message":[],"_revision_number":7},{"id":"2039a607f49e79a954546c05ff74bcd0e1680d19","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2025-08-14 13:34:26.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"66730c8b085dc973f0b01dfab4128155cdb6964b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-14 13:34:50.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"98c6d0ac925f097294f00596873ca59df811ca03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 01:53:08.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2cd7043f9c243bdab1aa21672810247\n\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/856b77d2eca1434691276b75553b4a08 : SUCCESS in 2m 27s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/7157a122fadd4887aab5816e799563f3 : SUCCESS in 7m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/16e4bacfcff94ed7b57e7a243b199452 : SUCCESS in 7m 49s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/4469ae140e504ba887da48b4399c45f6 : SUCCESS in 8m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c85b31817f414a349286a7524b071194 : SUCCESS in 3m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f1c12d2e8d344f178129e0be5d05ef97 : SUCCESS in 12m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d132f71f87c74c10bed1f33723e9d86d : SUCCESS in 7m 26s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/33b9998e65194ca3a0656d2a1eb6969e : SUCCESS in 11m 47s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/129ca21a715e403e83dfce6d2c0a28c8 : SUCCESS in 3m 51s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/9b8864234f5a431da44b93589e6931a2 : SUCCESS in 55m 56s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/de025f98c2b84dd8a335b25709b4b8f6 : SUCCESS in 22m 35s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/0e51b449228441e29dfcca97d7a4a465 : SUCCESS in 12m 22s\n- horizon-dsvm-tempest-plugin-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/073022fdc69a49f193185f62e2b49d4e : SUCCESS in 27m 03s","accounts_in_message":[],"_revision_number":7},{"id":"cd851fd3172003a1e9ba228d2dc64e31812aee23","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 01:53:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"221d8400501878ab1636c0ca8ff890d9f36e0520","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 01:56:11.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68bf51a9055249d39272b7aa1816e115\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdeb1fab24f64b8bb97c49c5192cc030 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"a98e0ade5a857ed4d479883d36d0bdabf36f6e33","revisions":{"8f2b1403f35aa9726a90632d9b63922e818c03e9":{"kind":"REWORK","_number":1,"created":"2025-05-18 21:35:56.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/71/950271/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/71/950271/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/71/950271/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/71/950271/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/71/950271/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/71/950271/1"}}},"commit":{"parents":[{"commit":"827a566d349020853e0370b6bf90066dd56b6ec2","subject":"Merge \"Fix url for default volume type call to backend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/827a566d349020853e0370b6bf90066dd56b6ec2"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-05-18 21:35:56.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-05-18 21:35:56.000000000","tz":0},"subject":"wip: language choice","message":"wip: language choice\n\nChange-Id: I07198f7201296167cdec136d265868091074c2ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/8f2b1403f35aa9726a90632d9b63922e818c03e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/8f2b1403f35aa9726a90632d9b63922e818c03e9"}]},"branch":"refs/heads/master"},"0e5e4adc4359b8749e39b2cec6196c7108132002":{"kind":"REWORK","_number":2,"created":"2025-05-18 22:58:06.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/71/950271/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/71/950271/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/71/950271/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/71/950271/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/71/950271/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/71/950271/2"}}},"commit":{"parents":[{"commit":"827a566d349020853e0370b6bf90066dd56b6ec2","subject":"Merge \"Fix url for default volume type call to backend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/827a566d349020853e0370b6bf90066dd56b6ec2"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-05-18 21:35:56.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-05-18 22:55:35.000000000","tz":0},"subject":"Language selection in config","message":"Language selection in config\n\nAdded a note to clarify that the list of supported languages can be customized\nvia the local_settings configuration file.\nChange-Id: I07198f7201296167cdec136d265868091074c2ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/0e5e4adc4359b8749e39b2cec6196c7108132002"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/0e5e4adc4359b8749e39b2cec6196c7108132002"}]},"branch":"refs/heads/master"},"696b0a90a01e7738327e80b93875cb4867fdcfc7":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-05-18 23:05:49.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/71/950271/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/71/950271/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/71/950271/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/71/950271/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/71/950271/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/71/950271/3"}}},"commit":{"parents":[{"commit":"827a566d349020853e0370b6bf90066dd56b6ec2","subject":"Merge \"Fix url for default volume type call to backend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/827a566d349020853e0370b6bf90066dd56b6ec2"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-05-18 21:35:56.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-05-18 23:05:49.000000000","tz":0},"subject":"Language selection in config","message":"Language selection in config\n\nAdded a note to clarify that the list of supported languages can be customized via the local_settings configuration file.\n\nChange-Id: I07198f7201296167cdec136d265868091074c2ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/696b0a90a01e7738327e80b93875cb4867fdcfc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/696b0a90a01e7738327e80b93875cb4867fdcfc7"}]},"branch":"refs/heads/master","description":"Edit commit message"},"f4365ebaf0517645294af8c9b4e853b265934244":{"kind":"REWORK","_number":4,"created":"2025-05-19 09:18:53.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/71/950271/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/71/950271/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/71/950271/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/71/950271/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/71/950271/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/71/950271/4"}}},"commit":{"parents":[{"commit":"827a566d349020853e0370b6bf90066dd56b6ec2","subject":"Merge \"Fix url for default volume type call to backend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/827a566d349020853e0370b6bf90066dd56b6ec2"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-05-18 21:35:56.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-05-19 09:18:42.000000000","tz":0},"subject":"Language selection in config","message":"Language selection in config\n\nAdded a note to clarify that the list of supported languages can be customized via the local_settings configuration file.\n\nChange-Id: I07198f7201296167cdec136d265868091074c2ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/f4365ebaf0517645294af8c9b4e853b265934244"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/f4365ebaf0517645294af8c9b4e853b265934244"}]},"branch":"refs/heads/master"},"310170aca7155013901f443222690d6f7b82d59c":{"kind":"REWORK","_number":5,"created":"2025-05-19 11:27:24.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/71/950271/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/71/950271/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/71/950271/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/71/950271/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/71/950271/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/71/950271/5"}}},"commit":{"parents":[{"commit":"827a566d349020853e0370b6bf90066dd56b6ec2","subject":"Merge \"Fix url for default volume type call to backend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/827a566d349020853e0370b6bf90066dd56b6ec2"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-05-18 21:35:56.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-05-19 11:27:16.000000000","tz":0},"subject":"Language selection in config","message":"Language selection in config\n\nAdded a note to clarify that the list of supported languages can be customized via the local_settings configuration file.\n\nChange-Id: I07198f7201296167cdec136d265868091074c2ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/310170aca7155013901f443222690d6f7b82d59c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/310170aca7155013901f443222690d6f7b82d59c"}]},"branch":"refs/heads/master"},"cfe0abd5e4a03a81637193a7221060bd527d2e6c":{"kind":"REWORK","_number":6,"created":"2025-05-19 12:29:22.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/71/950271/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/71/950271/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/71/950271/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/71/950271/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/71/950271/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/71/950271/6"}}},"commit":{"parents":[{"commit":"827a566d349020853e0370b6bf90066dd56b6ec2","subject":"Merge \"Fix url for default volume type call to backend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/827a566d349020853e0370b6bf90066dd56b6ec2"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-05-18 21:35:56.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-05-19 12:29:14.000000000","tz":0},"subject":"Language selection in config","message":"Language selection in config\n\nAdded a note to clarify that the list of supported languages can be customized via the local_settings configuration file.\n\nChange-Id: I07198f7201296167cdec136d265868091074c2ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/cfe0abd5e4a03a81637193a7221060bd527d2e6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/cfe0abd5e4a03a81637193a7221060bd527d2e6c"}]},"branch":"refs/heads/master"},"a98e0ade5a857ed4d479883d36d0bdabf36f6e33":{"kind":"REWORK","_number":7,"created":"2025-05-19 12:57:07.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/71/950271/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/71/950271/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/71/950271/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/71/950271/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/71/950271/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/71/950271/7"}}},"commit":{"parents":[{"commit":"827a566d349020853e0370b6bf90066dd56b6ec2","subject":"Merge \"Fix url for default volume type call to backend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/827a566d349020853e0370b6bf90066dd56b6ec2"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-05-18 21:35:56.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2025-05-19 12:56:59.000000000","tz":0},"subject":"Language selection in config","message":"Language selection in config\n\nAdded a note to clarify that the list of supported languages can be customized via the local_settings configuration file.\n\nChange-Id: I07198f7201296167cdec136d265868091074c2ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/a98e0ade5a857ed4d479883d36d0bdabf36f6e33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/a98e0ade5a857ed4d479883d36d0bdabf36f6e33"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
