)]}'
{"id":"openstack%2Fhorizon~953891","triplet_id":"openstack%2Fhorizon~master~I04f99895c41fb405e1966014fcf4c2751bdfda96","project":"openstack/horizon","branch":"master","topic":"test_manila_create_share","attention_set":{},"removed_from_attention_set":{"6914":{"account":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"last_update":"2025-08-04 07:51:46.000000000","reason":"Change was abandoned"},"8648":{"account":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"last_update":"2025-08-04 07:51:46.000000000","reason":"Change was abandoned"},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2025-08-04 07:51:46.000000000","reason":"Change was abandoned"},"35133":{"account":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"last_update":"2025-07-09 13:39:44.000000000","reason":"\u003cGERRIT_ACCOUNT_35133\u003e replied on the change","reason_account":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}}},"hashtags":[],"change_id":"I04f99895c41fb405e1966014fcf4c2751bdfda96","subject":"Add manila-ui job and first manila integration tests","status":"ABANDONED","created":"2025-07-01 15:42:43.000000000","updated":"2025-08-04 07:51:46.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"41bc141cd354f4a844a8ac93a0f1ad1c44842cc4","_number":953891,"virtual_id_number":953891,"owner":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-07-10 09:00:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2025-07-15 12:01:32.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}],"CC":[{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-01 16:54:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-01 19:28:43.000000000","updated_by":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"reviewer":{"_account_id":35153,"name":"Ashish Gupta","email":"ashigupt@redhat.com","username":"ashigupt","status":"Redhat"},"state":"CC"},{"updated":"2025-07-09 13:39:44.000000000","updated_by":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2025-07-14 08:56:34.000000000","updated_by":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"reviewer":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"state":"REVIEWER"},{"updated":"2025-07-14 08:56:34.000000000","updated_by":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"reviewer":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"state":"REVIEWER"}],"messages":[{"id":"0de1b71ac457578610b91fba7c819fb1d6b8ddaa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2025-07-01 15:42:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"be464b6754725b55a3ec1f10d3246f23f33d4748","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 16:54:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9ac899154621478999ab6af01625ad7b\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/813776157b4245c7aa4ab99e5300607d : SUCCESS in 7m 15s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/32f6803640324998b43e55d8a09e0bae : SUCCESS in 4m 37s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/9bf3c9b4edd24e4d809a895360f18032 : SUCCESS in 6m 17s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/bced177192af469eb4b0f1d0c1ab120c : SUCCESS in 7m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/617e0b5e4d9f46a3a86b89ce4f2e86e3 : SUCCESS in 14m 04s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/83bce37a2ffc434bbe9af779e22efe1b : SUCCESS in 14m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/957797ac40504c7db569539f4eba2862 : SUCCESS in 6m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/feaa5fa9b77744739ca381e35fab8ded : SUCCESS in 15m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/994c9181f6b24949996a5b9cd65a27c1 : SUCCESS in 18m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfe335a527204b2ea4aff8f8dc2b4b4f : SUCCESS in 18m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f5c0ba1e27847b3b68634fdbbbd1884 : SUCCESS in 14m 56s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/5885919a814a4e58aa96761087c3f8ad : SUCCESS in 6m 00s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/f9833e401a89404d8412e129401bcff8 : SUCCESS in 1h 05m 17s\n- horizon-manila-integration-pytest https://zuul.opendev.org/t/openstack/build/4766c4b48e434c36b73d3b79328a8cbb : FAILURE in 29m 36s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/93689bbec82445a8a0d6655191f87b22 : SUCCESS in 28m 04s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/38f02ffacb3a49cf869d349b8fa25ec7 : SUCCESS in 21m 21s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2a1c23e66fdf43b0a3c2cc772f1a2153 : SUCCESS in 4m 14s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/5ba65e743199488aa5a7a2fa3be72a64 : SUCCESS in 23m 51s\n- horizon-dsvm-tempest-plugin-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1b5b20f2a1064b5cbb7acf1cba25e587 : SUCCESS in 21m 56s","accounts_in_message":[],"_revision_number":1},{"id":"4b474c6155cccc89660d5ee0fb13625ad4452ee5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2025-07-03 14:42:24.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"77433052cda671cc4615f4515948d46363a3737b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 14:43:15.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/horizon on branch master.  The\nproblem was:\n\n  expected str for dictionary value @ data[\u0027check\u0027][\u0027jobs\u0027][2][\u0027horizon-manila-integration-pytest\u0027]\n\nThe problem appears in the the following project stanza:\n\n  project:\n      queue: horizon\n      templates:\n        - check-requirements\n        - horizon-cross-jobs\n        - horizon-nodejs-jobs\n        - horizon-non-primary-django-jobs\n        - openstack-python3-jobs\n        - periodic-stable-jobs\n  ...\n\n  in \"openstack/horizon/.zuul.d/project.yaml@master\", line 1","accounts_in_message":[],"_revision_number":2},{"id":"646a36761bb45f876db78cd061b773ddd2626438","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2025-07-03 14:50:34.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"4538e71445dce12bad0d510ccf2ad8b1a225a2e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 14:51:00.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/horizon on branch master.  The\nproblem was:\n\n  expected str for dictionary value @ data[\u0027check\u0027][\u0027jobs\u0027][2][\u0027horizon-manila-integration-pytest\u0027]\n\nThe problem appears in the the following project stanza:\n\n  project:\n      queue: horizon\n      templates:\n        - check-requirements\n        - horizon-cross-jobs\n        - horizon-nodejs-jobs\n        - horizon-non-primary-django-jobs\n        - openstack-python3-jobs\n        - periodic-stable-jobs\n  ...\n\n  in \"openstack/horizon/.zuul.d/project.yaml@master\", line 1","accounts_in_message":[],"_revision_number":3},{"id":"650d294b9857071af8374c3213f7309b1c551166","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2025-07-03 14:54:06.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"1065049ca93e3846709fd29fc4649e08ed79ba37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 16:12:00.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc4854a0ea6f45439a76e3d745d80cc2\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/2efdc2a7b7d04598a295aad0c95ab1ea : SUCCESS in 3m 17s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/d84ad92c6ee04517847083afc1d6089f : SUCCESS in 2m 01s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/eef45dc66fa245afa7899746fa6ac8b9 : SUCCESS in 2m 27s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/64e58911ddd445a3888b22399a1fb655 : SUCCESS in 6m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8d4e71b4eeda411788f1be4f72cc258f : SUCCESS in 5m 53s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/cceef6c39fa141ac86688de0a1f4ebb7 : SUCCESS in 11m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfc6340e7b2742bfb228c7f03d51004d : SUCCESS in 2m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f5b618741dcc48119b3d902a7259a332 : SUCCESS in 15m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bf1871a38b5543f1b7676f7e186f7895 : SUCCESS in 14m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f10b418525ed4adbac884d9d8e652d0b : SUCCESS in 11m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98dd975afc31413b8ef9cadb8bfdfa00 : SUCCESS in 15m 58s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/58d6e37e239e4a9d85c432e652182ddd : SUCCESS in 4m 48s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/a3269d36a5d546f39eaad91324bcc9c8 : SUCCESS in 1h 11m 35s\n- horizon-manila-integration-pytest https://zuul.opendev.org/t/openstack/build/97e1f5d7d1c24125b1325c2fbdecf772 : SUCCESS in 15m 10s (non-voting)\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/b71c34c1bfee4d5a8cc951bf48aa6981 : SUCCESS in 28m 44s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/c78e01e6c3e74314803470926ab3c947 : SUCCESS in 28m 11s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f0a1480a4d5949039c9db390e253be78 : SUCCESS in 6m 18s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/00bcc4d4ff624422bb24926e5471dcf8 : SUCCESS in 24m 30s\n- horizon-dsvm-tempest-plugin-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/afcf7f3af32c466a912783dafc6aa4a8 : SUCCESS in 32m 23s","accounts_in_message":[],"_revision_number":4},{"id":"0eb847b71a9f5d0c3020ad1cdaf8eda925e8d2f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2025-07-04 09:51:07.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"56abe6d329de3098a3713ce77b5bb8e0a2657141","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 11:13:55.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4be07cb53f0c47d992c41a3e475233b8\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/f610da05cdd444d2883bfc85b45cbd8a : SUCCESS in 6m 51s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/ec6c1988224c40979676dadb2841acb7 : SUCCESS in 4m 27s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/95576e6fcb0440edb1c51fd054c2d769 : SUCCESS in 3m 58s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/a5a56ddfc3334c58a47455c7683e5e0d : SUCCESS in 6m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f816c5c868f6404dbe1e9d8b64163a47 : SUCCESS in 15m 15s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/d6998df392114701b7532f3002f0f6ed : SUCCESS in 15m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/307916bb896e455eb8de449b4d7720ff : SUCCESS in 4m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e5f4ee596e944dd488af9eb84088f57f : SUCCESS in 5m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9bca0add0b82483da0fa82ed449f9833 : SUCCESS in 18m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc219e203bf146afb7ab575e25257d1c : SUCCESS in 17m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8468cfb5e3504557b0a54c6f3fb03eed : SUCCESS in 17m 44s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/8951832189884be588e6553e9c16c08b : SUCCESS in 5m 59s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/8a0ce36e44c6431baef791077b1be9c0 : SUCCESS in 1h 20m 44s\n- horizon-manila-integration-pytest https://zuul.opendev.org/t/openstack/build/3f2737c684304a7fa7e03e7b6321ea55 : FAILURE in 13m 59s (non-voting)\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/f1b8b1f4498e44f2b825e5b3dd023c68 : SUCCESS in 26m 58s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/64bf7353ff6e44a7bdd90d85c1c63432 : SUCCESS in 12m 23s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a6d734dfeb7748d094e298c4b6809d1d : SUCCESS in 2m 11s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/38fa2761dda24a93b83e470e6fe0eb60 : SUCCESS in 22m 18s\n- horizon-dsvm-tempest-plugin-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/132868f4082a4d2688c2c317403f15e2 : SUCCESS in 11m 47s","accounts_in_message":[],"_revision_number":5},{"id":"93901ee20146ae7c115bb8865d498cf77f8770a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2025-07-04 12:29:11.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"e8d8f723d55ab35da36c15afed264f84062ccd3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 13:21:21.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df2ef5dc715249a58957fae47e024501\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/9e23f39f44944da28110155efda0661e : SUCCESS in 2m 54s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/f334a8aa73ce4db2a0a0c489fce34608 : SUCCESS in 2m 27s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/d1972a02437f4da7871250d48d73cc1f : SUCCESS in 4m 03s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/5ccbc487f00c4af1823ec9f744ac1d3d : SUCCESS in 7m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b1ace254172647a3ac0ff65465676fd3 : SUCCESS in 16m 59s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/f410e386dcfe4770b4e16c12bbd9a411 : SUCCESS in 7m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13a10f66091d41eabddb94a778fc7647 : SUCCESS in 5m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/936f95823260451c92b57170fdd26edc : SUCCESS in 16m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7cb1f639fa134575b67b633aabffda4a : SUCCESS in 22m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21f390254d5b4f7db9d6e44fc8d7e690 : SUCCESS in 12m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3cd3cda4412548d38ecb080dcbb27eb0 : SUCCESS in 15m 26s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/68d525a3da1b428087066373cb77316c : SUCCESS in 7m 47s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/02714d779c114601af7c3d02b9b3719c : SUCCESS in 50m 46s\n- horizon-manila-integration-pytest https://zuul.opendev.org/t/openstack/build/c3168f1ab52a4778ba1b3c8471b1118b : SUCCESS in 28m 41s (non-voting)\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/b59c58a6990542ccaba155c089a970b2 : SUCCESS in 29m 40s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/bc677b35b08047c5889bdf64eac86eb1 : SUCCESS in 28m 31s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/20639df78d004f0dacb326b51f323ee8 : SUCCESS in 5m 29s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/37693fec2a0a490cb13f0d5d77762ac7 : SUCCESS in 27m 27s\n- horizon-dsvm-tempest-plugin-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e3707214bfe74d01a8a0c4cd27ab29c5 : SUCCESS in 28m 01s","accounts_in_message":[],"_revision_number":6},{"id":"46937a841819185792c1cef58db0e287847256f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2025-07-09 10:37:18.000000000","message":"Uploaded patch set 7: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"2305a2dfd4c58146f765dc281bdfa8f97570f21f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 12:03:19.000000000","message":"Patch Set 7: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dd3c7791ff5f4daaad98e179ef959775\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/755b9526720e4b9d938b21e51f1cc8e6 : SUCCESS in 6m 22s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/068414f6b0e34f57b3c2a4e1c32b23de : SUCCESS in 3m 54s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/f31d358d4fc74df589892f3aebfa8255 : SUCCESS in 5m 08s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/f945b362f9c7456ea8c40bbba91eba03 : SUCCESS in 5m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5e6b3cd8e2004e93a4d7899df56730e4 : SUCCESS in 15m 02s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/d9c94f99e6e14c7c9993e9b178350d49 : SUCCESS in 19m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/124771875b4c4355b58d77dc93ae33dd : FAILURE in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b714e78846d64b6e9171489c632165d1 : SUCCESS in 15m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fdd01957eb664a60b27d8a0b6874cf20 : SUCCESS in 7m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e80fefffc1654645819bc600e770ff23 : SUCCESS in 16m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/278c0a1c9a9443a3ac3038abb9358f52 : SUCCESS in 13m 02s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/3cdb70154eec46c28b2a72d0b237a63e : SUCCESS in 3m 49s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/4f6e88496fa041d28c4497a01c9b2db5 : FAILURE in 1h 20m 27s\n- horizon-manila-integration-pytest https://zuul.opendev.org/t/openstack/build/751598edb1a7498b96358be10bc46f55 : SUCCESS in 31m 18s (non-voting)\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/f48630ec41e7463382384730645046ef : SUCCESS in 29m 46s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/e681523a3190443b98d59f14f046df5c : SUCCESS in 22m 47s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e14fab33a02841a5b1c3f01644759c85 : SUCCESS in 4m 18s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/5f3e60e98aa54e3bbc9aeddfda167101 : SUCCESS in 11m 26s\n- horizon-dsvm-tempest-plugin-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9bf1d403a8cf41d3ad23edc8f4e07f65 : SUCCESS in 26m 56s","accounts_in_message":[],"_revision_number":7},{"id":"f33c6160e849971cfae669eb6fc3c6eb513cb12c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2025-07-09 12:15:21.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"2182b36b90b36161a7f2eff15781023883fb1fec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 13:21:24.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdee5ed1c8aa47a4a603e29b3cb176cd\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/c2daf236b0df4650bb1d432d9fe88598 : SUCCESS in 6m 32s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/cf2f027679f94556b102358b1c50ec57 : SUCCESS in 3m 38s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/735a32776af24a52b1a34b4f0f105dc8 : SUCCESS in 5m 06s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/e12210c9d2204d52b918fa88336fbde5 : SUCCESS in 6m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/864a459a1b80430bb78c649a5daf5e1f : SUCCESS in 10m 52s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/813d2c3371bd40b8a0e36c148e9ddb97 : SUCCESS in 13m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bcf7e5b38294b269989e7a50081d582 : SUCCESS in 5m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c01160aeea6746ca857a69e5f5a6f8fd : SUCCESS in 11m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7b96f76a090949debd3ec32eed84850e : SUCCESS in 16m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89409e67d046490d9e2b4c40ab9163eb : SUCCESS in 14m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c657b356f7e43a88f6333a4bc444638 : SUCCESS in 14m 04s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/be8ae60a8c5b4d2da8e81ba569281524 : SUCCESS in 6m 22s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/c13735096af34314be7ac1cca3a8acd2 : SUCCESS in 1h 05m 06s\n- horizon-manila-integration-pytest https://zuul.opendev.org/t/openstack/build/968b1f0e5f0e4edd8198d8f8c0f8812c : SUCCESS in 28m 57s (non-voting)\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/90b441ffae014f05aec35ecfbd128037 : SUCCESS in 18m 39s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/d41cae03eb364dc1910b9052010d0a1d : SUCCESS in 19m 11s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4ee00d5e95e844159752c6f9fb5c12d7 : SUCCESS in 2m 21s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/065e1c8665eb43db95e77c8233bce2f5 : SUCCESS in 22m 30s\n- horizon-dsvm-tempest-plugin-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7108e353be7741c7a02ec7beb8d35bb0 : SUCCESS in 25m 52s","accounts_in_message":[],"_revision_number":8},{"id":"4febd6b6e4b5f0bedefd52fa7c03b8f790c29f05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2025-07-10 07:50:43.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"6557d6f3a3187327f1b6e1e8674861c2b73add4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 09:00:04.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51a2e0a829054021b1477fa89638de09\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/c0b4367a2095416c8aeb33465e2058f3 : SUCCESS in 6m 40s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/2beaa9ec08c34c2980c6fc01807ce92b : SUCCESS in 3m 24s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/de0825ab42d0481382110ae6ad5acdf5 : SUCCESS in 7m 23s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/575fc521079346c3bc9bdd701cde48bd : SUCCESS in 8m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cb05e137a7ac4ea4aae682e1efde993f : SUCCESS in 12m 09s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/535c7015606c496b971f6cc1dc46a02f : SUCCESS in 14m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a1bafe1dbf34de895255f1d54325bd1 : SUCCESS in 4m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/97cc1e204ce94b25b6c024656b7abf31 : SUCCESS in 12m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bb53db32b7924b3d9633bc8820d51794 : SUCCESS in 17m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b881b8ad78f847e49cb63c4c79ebb0a2 : SUCCESS in 17m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/30e6e3ded7674f0b9f12580bccc52c37 : SUCCESS in 14m 07s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/a91c493fde694208bbc45ff7667e0e3b : SUCCESS in 5m 56s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/d0f1cbd8a3584af2a6a6a14c49001999 : SUCCESS in 1h 06m 06s\n- horizon-manila-integration-pytest https://zuul.opendev.org/t/openstack/build/e8fd0c8f399f479da102f5ff0daea18a : SUCCESS in 21m 22s (non-voting)\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/5034189f003a4baf8667be8a5509b70e : SUCCESS in 30m 13s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/ac4cd56c8a9d4f34b389dd0b0ac44623 : SUCCESS in 28m 36s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d039d120df784093a4fe6e1f98b08ba8 : SUCCESS in 5m 32s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/33b01df7d94a48b38bf698153793abe0 : SUCCESS in 28m 15s\n- horizon-dsvm-tempest-plugin-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2cd17399e441437b82cba2fb1532080b : SUCCESS in 20m 12s","accounts_in_message":[],"_revision_number":9},{"id":"964924c913706eb492c71cb5a791b97cc303ac89","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2025-07-15 12:01:32.000000000","message":"Patch Set 9: Workflow-1","accounts_in_message":[],"_revision_number":9},{"id":"41bc141cd354f4a844a8ac93a0f1ad1c44842cc4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2025-08-04 07:51:46.000000000","message":"Abandoned\n\nMoved to manila-ui repository. Continues here: https://review.opendev.org/c/openstack/manila-ui/+/956098","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"419d2dd6d5400b903365dc1ac8dcdb8c24215e78","revisions":{"c4a52a08dff5bd10bdb2b6a3f0bd31c82197a13f":{"kind":"REWORK","_number":1,"created":"2025-07-01 15:42:43.000000000","uploader":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"ref":"refs/changes/91/953891/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/91/953891/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/91/953891/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/91/953891/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/91/953891/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/91/953891/1"}}},"commit":{"parents":[{"commit":"1c401872713028586b9c1ea5146441389eca8ff0","subject":"Merge \"Make role name validation case insensitive\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/1c401872713028586b9c1ea5146441389eca8ff0"}]}],"author":{"name":"Jan Jasek","email":"jjasek@redhat.com","date":"2025-07-01 15:36:50.000000000","tz":120},"committer":{"name":"Jan Jasek","email":"jjasek@redhat.com","date":"2025-07-01 15:41:38.000000000","tz":120},"subject":"[WIP] Add manila-ui job and first manila integration tests","message":"[WIP] Add manila-ui job and first manila integration tests\n\nChange-Id: I04f99895c41fb405e1966014fcf4c2751bdfda96\nSigned-off-by: Jan Jasek \u003cjjasek@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/c4a52a08dff5bd10bdb2b6a3f0bd31c82197a13f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/c4a52a08dff5bd10bdb2b6a3f0bd31c82197a13f"}]},"branch":"refs/heads/master"},"021b2e7ff9febb96e54e6a8cb687f5cd88278cdb":{"kind":"REWORK","_number":2,"created":"2025-07-03 14:42:24.000000000","uploader":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"ref":"refs/changes/91/953891/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/91/953891/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/91/953891/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/91/953891/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/91/953891/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/91/953891/2"}}},"commit":{"parents":[{"commit":"1c401872713028586b9c1ea5146441389eca8ff0","subject":"Merge \"Make role name validation case insensitive\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/1c401872713028586b9c1ea5146441389eca8ff0"}]}],"author":{"name":"Jan Jasek","email":"jjasek@redhat.com","date":"2025-07-01 15:36:50.000000000","tz":120},"committer":{"name":"Jan Jasek","email":"jjasek@redhat.com","date":"2025-07-03 14:41:44.000000000","tz":120},"subject":"[WIP] Add manila-ui job and first manila integration tests","message":"[WIP] Add manila-ui job and first manila integration tests\n\nChange-Id: I04f99895c41fb405e1966014fcf4c2751bdfda96\nSigned-off-by: Jan Jasek \u003cjjasek@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/021b2e7ff9febb96e54e6a8cb687f5cd88278cdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/021b2e7ff9febb96e54e6a8cb687f5cd88278cdb"}]},"branch":"refs/heads/master"},"10b3ba6e12463a46782274ec8eead438260b3440":{"kind":"REWORK","_number":3,"created":"2025-07-03 14:50:34.000000000","uploader":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"ref":"refs/changes/91/953891/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/91/953891/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/91/953891/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/91/953891/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/91/953891/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/91/953891/3"}}},"commit":{"parents":[{"commit":"1c401872713028586b9c1ea5146441389eca8ff0","subject":"Merge \"Make role name validation case insensitive\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/1c401872713028586b9c1ea5146441389eca8ff0"}]}],"author":{"name":"Jan Jasek","email":"jjasek@redhat.com","date":"2025-07-01 15:36:50.000000000","tz":120},"committer":{"name":"Jan Jasek","email":"jjasek@redhat.com","date":"2025-07-03 14:50:21.000000000","tz":120},"subject":"[WIP] Add manila-ui job and first manila integration tests","message":"[WIP] Add manila-ui job and first manila integration tests\n\nChange-Id: I04f99895c41fb405e1966014fcf4c2751bdfda96\nSigned-off-by: Jan Jasek \u003cjjasek@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/10b3ba6e12463a46782274ec8eead438260b3440"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/10b3ba6e12463a46782274ec8eead438260b3440"}]},"branch":"refs/heads/master"},"b303c8d719e71a67f3cd4dbe3e1d80d2227923c5":{"kind":"REWORK","_number":4,"created":"2025-07-03 14:54:06.000000000","uploader":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"ref":"refs/changes/91/953891/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/91/953891/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/91/953891/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/91/953891/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/91/953891/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/91/953891/4"}}},"commit":{"parents":[{"commit":"1c401872713028586b9c1ea5146441389eca8ff0","subject":"Merge \"Make role name validation case insensitive\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/1c401872713028586b9c1ea5146441389eca8ff0"}]}],"author":{"name":"Jan Jasek","email":"jjasek@redhat.com","date":"2025-07-01 15:36:50.000000000","tz":120},"committer":{"name":"Jan Jasek","email":"jjasek@redhat.com","date":"2025-07-03 14:53:56.000000000","tz":120},"subject":"[WIP] Add manila-ui job and first manila integration tests","message":"[WIP] Add manila-ui job and first manila integration tests\n\nChange-Id: I04f99895c41fb405e1966014fcf4c2751bdfda96\nSigned-off-by: Jan Jasek \u003cjjasek@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/b303c8d719e71a67f3cd4dbe3e1d80d2227923c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/b303c8d719e71a67f3cd4dbe3e1d80d2227923c5"}]},"branch":"refs/heads/master"},"ff41e95277af8046dc881638d9ff4eb395635c7e":{"kind":"REWORK","_number":5,"created":"2025-07-04 09:51:07.000000000","uploader":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"ref":"refs/changes/91/953891/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/91/953891/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/91/953891/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/91/953891/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/91/953891/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/91/953891/5"}}},"commit":{"parents":[{"commit":"1c401872713028586b9c1ea5146441389eca8ff0","subject":"Merge \"Make role name validation case insensitive\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/1c401872713028586b9c1ea5146441389eca8ff0"}]}],"author":{"name":"Jan Jasek","email":"jjasek@redhat.com","date":"2025-07-01 15:36:50.000000000","tz":120},"committer":{"name":"Jan Jasek","email":"jjasek@redhat.com","date":"2025-07-04 09:50:46.000000000","tz":120},"subject":"[WIP] Add manila-ui job and first manila integration tests","message":"[WIP] Add manila-ui job and first manila integration tests\n\nChange-Id: I04f99895c41fb405e1966014fcf4c2751bdfda96\nSigned-off-by: Jan Jasek \u003cjjasek@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/ff41e95277af8046dc881638d9ff4eb395635c7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/ff41e95277af8046dc881638d9ff4eb395635c7e"}]},"branch":"refs/heads/master"},"aa13321faa319f5c552d779f392535a0a2ad07e9":{"kind":"REWORK","_number":6,"created":"2025-07-04 12:29:11.000000000","uploader":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"ref":"refs/changes/91/953891/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/91/953891/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/91/953891/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/91/953891/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/91/953891/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/91/953891/6"}}},"commit":{"parents":[{"commit":"1c401872713028586b9c1ea5146441389eca8ff0","subject":"Merge \"Make role name validation case insensitive\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/1c401872713028586b9c1ea5146441389eca8ff0"}]}],"author":{"name":"Jan Jasek","email":"jjasek@redhat.com","date":"2025-07-01 15:36:50.000000000","tz":120},"committer":{"name":"Jan Jasek","email":"jjasek@redhat.com","date":"2025-07-04 11:55:57.000000000","tz":120},"subject":"[WIP] Add manila-ui job and first manila integration tests","message":"[WIP] Add manila-ui job and first manila integration tests\n\nChange-Id: I04f99895c41fb405e1966014fcf4c2751bdfda96\nSigned-off-by: Jan Jasek \u003cjjasek@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/aa13321faa319f5c552d779f392535a0a2ad07e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/aa13321faa319f5c552d779f392535a0a2ad07e9"}]},"branch":"refs/heads/master"},"7dd5baf82f3f2c785586ad0f725a81d14af3cdbb":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2025-07-09 10:37:18.000000000","uploader":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"ref":"refs/changes/91/953891/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/91/953891/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/91/953891/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/91/953891/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/91/953891/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/91/953891/7"}}},"commit":{"parents":[{"commit":"1c401872713028586b9c1ea5146441389eca8ff0","subject":"Merge \"Make role name validation case insensitive\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/1c401872713028586b9c1ea5146441389eca8ff0"}]}],"author":{"name":"Jan Jasek","email":"jjasek@redhat.com","date":"2025-07-01 15:36:50.000000000","tz":120},"committer":{"name":"Jan Jasek","email":"jjasek@redhat.com","date":"2025-07-09 10:35:55.000000000","tz":120},"subject":"Add manila-ui job and first manila integration tests","message":"Add manila-ui job and first manila integration tests\n\nAdd horizon-manila-integration-pytest job which\ndeploy devstack with enabled manila, manila-ui\nand with a dummy backend.\n\nAdd horizon-manila-integration-pytest job among the\ncheck jobs as a non-voting.\n\nAdd first bunch of manila shares tests:\ntest_create_share_demo\ntest_delete_share_demo\ntest_edit_share_demo\n\nChange-Id: I04f99895c41fb405e1966014fcf4c2751bdfda96\nSigned-off-by: Jan Jasek \u003cjjasek@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/7dd5baf82f3f2c785586ad0f725a81d14af3cdbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/7dd5baf82f3f2c785586ad0f725a81d14af3cdbb"}]},"branch":"refs/heads/master"},"7cac4d1c4b2b9fb915cb26bd8eaa1d56c5188c97":{"kind":"REWORK","_number":8,"created":"2025-07-09 12:15:21.000000000","uploader":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"ref":"refs/changes/91/953891/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/91/953891/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/91/953891/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/91/953891/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/91/953891/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/91/953891/8"}}},"commit":{"parents":[{"commit":"1c401872713028586b9c1ea5146441389eca8ff0","subject":"Merge \"Make role name validation case insensitive\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/1c401872713028586b9c1ea5146441389eca8ff0"}]}],"author":{"name":"Jan Jasek","email":"jjasek@redhat.com","date":"2025-07-01 15:36:50.000000000","tz":120},"committer":{"name":"Jan Jasek","email":"jjasek@redhat.com","date":"2025-07-09 12:10:01.000000000","tz":120},"subject":"Add manila-ui job and first manila integration tests","message":"Add manila-ui job and first manila integration tests\n\nAdd horizon-manila-integration-pytest job which\ndeploy devstack with enabled manila, manila-ui\nand with a dummy backend.\n\nAdd horizon-manila-integration-pytest job among the\ncheck jobs as a non-voting.\n\nAdd first bunch of manila shares tests:\ntest_create_share_demo\ntest_delete_share_demo\ntest_edit_share_demo\n\nChange-Id: I04f99895c41fb405e1966014fcf4c2751bdfda96\nSigned-off-by: Jan Jasek \u003cjjasek@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/7cac4d1c4b2b9fb915cb26bd8eaa1d56c5188c97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/7cac4d1c4b2b9fb915cb26bd8eaa1d56c5188c97"}]},"branch":"refs/heads/master"},"419d2dd6d5400b903365dc1ac8dcdb8c24215e78":{"kind":"REWORK","_number":9,"created":"2025-07-10 07:50:43.000000000","uploader":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"ref":"refs/changes/91/953891/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/91/953891/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/91/953891/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/91/953891/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/91/953891/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/91/953891/9"}}},"commit":{"parents":[{"commit":"c52c9ce653df7048a92421afb7100b69833361ee","subject":"Merge \"Make integration tests stop on first failure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/c52c9ce653df7048a92421afb7100b69833361ee"}]}],"author":{"name":"Jan Jasek","email":"jjasek@redhat.com","date":"2025-07-01 15:36:50.000000000","tz":120},"committer":{"name":"Jan Jasek","email":"jjasek@redhat.com","date":"2025-07-10 07:47:30.000000000","tz":120},"subject":"Add manila-ui job and first manila integration tests","message":"Add manila-ui job and first manila integration tests\n\nAdd horizon-manila-integration-pytest job which\ndeploy devstack with enabled manila, manila-ui\nand with a dummy backend.\n\nAdd horizon-manila-integration-pytest job among the\ncheck jobs as a non-voting.\n\nAdd first bunch of manila shares tests:\ntest_create_share_demo\ntest_delete_share_demo\ntest_edit_share_demo\n\nChange-Id: I04f99895c41fb405e1966014fcf4c2751bdfda96\nSigned-off-by: Jan Jasek \u003cjjasek@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/419d2dd6d5400b903365dc1ac8dcdb8c24215e78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/419d2dd6d5400b903365dc1ac8dcdb8c24215e78"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
