)]}'
{"id":"openstack%2Fhorizon~956006","triplet_id":"openstack%2Fhorizon~master~Ie5fcbe934e4434044b146a724d4cfc907d506b06","project":"openstack/horizon","branch":"master","topic":"bug/2116927","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2025-08-06 08:06:00.000000000","reason":"\u003cGERRIT_ACCOUNT_37632\u003e replied on the change","reason_account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}},"6914":{"account":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"last_update":"2025-08-07 16:33:15.000000000","reason":"\u003cGERRIT_ACCOUNT_6914\u003e replied on the change","reason_account":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"}},"8648":{"account":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"last_update":"2025-08-07 16:52:53.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2025-08-07 16:52:55.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-08-06 08:39:45.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-08-06 08:39:44.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"Ie5fcbe934e4434044b146a724d4cfc907d506b06","subject":"SDK subnetpools: change prefixlen fields as SDK calls them","status":"MERGED","created":"2025-07-28 13:07:49.000000000","updated":"2025-08-07 17:47:39.000000000","submitted":"2025-08-07 17:46:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"956006-bug/2116927","meta_rev_id":"881bdc0d21c24e0226d92826fadfc6e997e5d350","_number":956006,"virtual_id_number":956006,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":34380,"name":"Maximilian Stinsky","email":"maximilian@stinsky.com","username":"mstinsky"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-07 17:45:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"recommended":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"all":[{"value":1,"date":"2025-08-06 08:06:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":2,"date":"2025-08-07 16:33:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":1,"date":"2025-07-29 13:15:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2025-07-30 06:57:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34380,"name":"Maximilian Stinsky","email":"maximilian@stinsky.com","username":"mstinsky"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-07-29 10:37:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},{"value":1,"date":"2025-07-29 08:10:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":1,"date":"2025-08-07 16:33:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":34380,"name":"Maximilian Stinsky","email":"maximilian@stinsky.com","username":"mstinsky"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":34380,"name":"Maximilian Stinsky","email":"maximilian@stinsky.com","username":"mstinsky"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34380,"name":"Maximilian Stinsky","email":"maximilian@stinsky.com","username":"mstinsky"},{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-28 13:50:27.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2025-07-28 13:50:27.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2025-07-28 14:04:47.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-07-28 14:27:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-29 10:37:28.000000000","updated_by":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"reviewer":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"state":"REVIEWER"},{"updated":"2025-07-29 13:15:02.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-07-30 06:57:39.000000000","updated_by":{"_account_id":34380,"name":"Maximilian Stinsky","email":"maximilian@stinsky.com","username":"mstinsky"},"reviewer":{"_account_id":34380,"name":"Maximilian Stinsky","email":"maximilian@stinsky.com","username":"mstinsky"},"state":"REVIEWER"},{"updated":"2025-08-06 08:39:56.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"state":"REVIEWER"},{"updated":"2025-08-06 08:39:56.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"state":"REVIEWER"}],"messages":[{"id":"7f8f52e1104161c5c60ce3d072fb06608c610645","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-07-28 13:07:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f6e5b39e597651810f5bb6c91f49f075e0de19f","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-28 13:50:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"daf112c36f3e9f54e2952c353937748bfb38aa63","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-28 14:04:47.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8e57ba19cc82949ad5880de0a140a9c50c08ce8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-28 14:27:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/687784e3075c45bfb583d0f8464b2f35\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/267a1d73e80546b89969308e6a0998bb : SUCCESS in 5m 58s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/772c5e3c3bb744f59a1327ae0a89490a : SUCCESS in 1m 56s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/db1a08a2e45f4884bf74d809500b6a04 : SUCCESS in 2m 23s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/bfbebdfb38544df1a5169e3752a924f9 : SUCCESS in 6m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/08040ff9432543c98facabc1b7983919 : SUCCESS in 14m 57s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/97571ca5c25f4fa78102648e12722a2c : SUCCESS in 16m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6989c06c5c6a4b209e0590ab59795e19 : SUCCESS in 6m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a98c29ab352146919bc89242555e8a25 : SUCCESS in 16m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b51d02d5389141259f6030407f116346 : SUCCESS in 7m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/477bc9e00f4e4d40b76788dd60f64549 : SUCCESS in 20m 03s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/135c3eb089d64af9a78d4aadd4360e2d : SUCCESS in 6m 18s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/6061ce167f704a5cbfb5f50cecaf358d : SUCCESS in 1h 15m 30s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/30e4526116274653b4f346e53edd4190 : SUCCESS in 13m 33s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/a4141ed187f641a8a1bf7d2ab808d04a : SUCCESS in 26m 59s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fc82f54da3ee416998a6ce5d6e9eae45 : SUCCESS in 4m 56s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/0339d894f576428c968d1e9ca2ea10fb : SUCCESS in 29m 05s\n- horizon-dsvm-tempest-plugin-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a65fb1ca8fb4aeba0f998f39b0e47d2 : SUCCESS in 28m 19s","accounts_in_message":[],"_revision_number":1},{"id":"22edf9c8ad420dc74eadfa2a60d7e0546f067610","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-07-29 07:45:21.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"1c81659acbff8b6fe9d4bf6bc8b9dcc2c12535ab","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-29 08:10:24.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"cae476d314019368c9f715a07734db71911ca072","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-29 09:03:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f1005dc714441ffa5cbe36a87442bfe\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/7219459a4f45410fbd05a70bed06572b : SUCCESS in 7m 08s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/ce5c08cd1f9d446fbdf3deec9132bcd2 : SUCCESS in 3m 01s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/91b871584d1348caa79e81173fba1fdf : SUCCESS in 6m 53s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/d231039ca1bf404d897e3a389cab21e7 : SUCCESS in 6m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5c0c1386f2ac4a1c80a9f89853c70c37 : SUCCESS in 15m 08s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/7f0b3b03b24f41f99d08e7b36fe9a928 : SUCCESS in 15m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ee737ea29c247f18ce0b2a8a4bc806b : SUCCESS in 6m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fa4fa7e412c349b3bb47d5f2c3ffd950 : SUCCESS in 15m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3d3ff7a7fe6646af8ae4285fed2cfb59 : SUCCESS in 12m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/979aa043f12641a58c9d8968d0bc4d17 : SUCCESS in 10m 31s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/b609b39ca598423abca1a1432ed4c191 : SUCCESS in 5m 45s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/e92c23cbb55946dab62cb13bab75b74c : SUCCESS in 1h 13m 56s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/dbbb74b34793412ea24375be463c4da6 : SUCCESS in 25m 58s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/29b4ae40901946dcb84043c79ff2679c : SUCCESS in 28m 06s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/697ab1cc0739464ea76a26123b1322b1 : SUCCESS in 5m 59s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/525e81a3f35f474ba6a61e774a09cc4d : SUCCESS in 28m 16s\n- horizon-dsvm-tempest-plugin-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c3a3966788c4dbe908d4579369fa4ff : SUCCESS in 27m 31s","accounts_in_message":[],"_revision_number":2},{"id":"2f8c7078b576d99b1a088bb71f029f4493031f9e","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2025-07-29 10:37:28.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d4df3004997ea8a491c8852ee67f8f97b2e1a315","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-07-29 13:15:02.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"dff629c3fd4aa8f523a1efbaaa7da6056547feef","author":{"_account_id":34380,"name":"Maximilian Stinsky","email":"maximilian@stinsky.com","username":"mstinsky"},"date":"2025-07-30 06:57:39.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"e36a62b932566a0adf124f20bec6471d1f53ddbe","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-08-05 09:01:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"302f70721ae225a48f4512f366d121cd26ee8895","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-08-06 07:53:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"65fd1d462e3fd9d7d8f3a63b4df475a7d3964ceb","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-08-06 08:06:00.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"151d4a0df4b7759ffb86b453a86949c393f88be2","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-08-07 14:23:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d9199b41c6e73f7ae54a9a4a269683f45d84a890","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2025-08-07 16:33:15.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7834d471ca6e50e7986ee7b72e713ed2c85f9ac8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 16:33:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"906fde5638fcd6f59ff70e4345e31c7d200f6e97","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 17:45:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/210709349166431a8d05549d34877cad\n\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/6c29b3637e9a400683b91083224ed2b7 : SUCCESS in 7m 34s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/182881f33a94415a8ec8c2e8d19cec76 : SUCCESS in 6m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aa73f2573d254ec68bbe36ccac5630fa : SUCCESS in 15m 39s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/783dbf6d0fb04cc2bfe2e58ad75d85ae : SUCCESS in 15m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8febe65af37b49a2bdd217cb1a01be90 : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/83f430754c5e4fac8fce908f8e32bc46 : SUCCESS in 15m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d4dcf9c805b426488685fbcf0fbd13a : SUCCESS in 13m 40s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/f5aa86449532449987b7997645da3e1a : SUCCESS in 22m 41s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/ef1d3a6bd27a4fceb7741bd9cbe02857 : SUCCESS in 6m 32s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/6ffc434c46314d3d8ec0fba29e848e8e : SUCCESS in 1h 10m 44s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/fa1a736c490e4544a5a1b5c647760c7d : SUCCESS in 22m 30s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/1bb1f0102b3e4b40b053c11958bd4d80 : SUCCESS in 22m 40s\n- horizon-dsvm-tempest-plugin-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ed56a4e605943eaa0f3c2c7659df383 : SUCCESS in 20m 40s","accounts_in_message":[],"_revision_number":2},{"id":"798bb2aea54f7b18bbd47990f3fa376de17741a2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 17:46:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"881bdc0d21c24e0226d92826fadfc6e997e5d350","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 17:47:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7cd38d579f164a05ba0b4c938ae2342a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d3a72a4fe5a4fc3bafb4fd8f60f3b4c : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e2edb40c6233c1d82f4d28e7a898e87967351b3b","revisions":{"22ff159bd693f449b715a44dda35b7d870f23332":{"kind":"REWORK","_number":1,"created":"2025-07-28 13:07:49.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/06/956006/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/06/956006/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/06/956006/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/06/956006/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/06/956006/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/06/956006/1"}}},"commit":{"parents":[{"commit":"fb8bacac69cf8a9beadae67a15a26428b4e855e9","subject":"pytest-based selenium test fix test import path","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/fb8bacac69cf8a9beadae67a15a26428b4e855e9"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-07-28 10:03:46.000000000","tz":120},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-07-28 13:01:38.000000000","tz":120},"subject":"SDK subnetpools: change prefixlen fields as SDK calls them","message":"SDK subnetpools: change prefixlen fields as SDK calls them\n\nCloses-Bug: #2116927\nChange-Id: Ie5fcbe934e4434044b146a724d4cfc907d506b06\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/22ff159bd693f449b715a44dda35b7d870f23332"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/22ff159bd693f449b715a44dda35b7d870f23332"}]},"branch":"refs/heads/master"},"e2edb40c6233c1d82f4d28e7a898e87967351b3b":{"kind":"REWORK","_number":2,"created":"2025-07-29 07:45:21.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/06/956006/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/06/956006/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/06/956006/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/06/956006/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/06/956006/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/06/956006/2"}}},"commit":{"parents":[{"commit":"fb8bacac69cf8a9beadae67a15a26428b4e855e9","subject":"pytest-based selenium test fix test import path","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/fb8bacac69cf8a9beadae67a15a26428b4e855e9"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-07-28 10:03:46.000000000","tz":120},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-07-29 07:45:10.000000000","tz":120},"subject":"SDK subnetpools: change prefixlen fields as SDK calls them","message":"SDK subnetpools: change prefixlen fields as SDK calls them\n\nCloses-Bug: #2116927\nChange-Id: Ie5fcbe934e4434044b146a724d4cfc907d506b06\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/e2edb40c6233c1d82f4d28e7a898e87967351b3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/e2edb40c6233c1d82f4d28e7a898e87967351b3b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34380,"name":"Maximilian Stinsky","email":"maximilian@stinsky.com","username":"mstinsky"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
