)]}'
{"id":"openstack%2Fhorizon~956522","triplet_id":"openstack%2Fhorizon~master~I0756ba2146b7825cfe79c282d0b0d7636f24a6fb","project":"openstack/horizon","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2025-08-05 09:05:02.000000000","reason":"removed on reply"},"6914":{"account":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"last_update":"2025-08-07 16:40:13.000000000","reason":"\u003cGERRIT_ACCOUNT_6914\u003e replied on the change","reason_account":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"}},"8648":{"account":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"last_update":"2025-08-08 06:49:55.000000000","reason":"Change was submitted"},"35133":{"account":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"last_update":"2025-08-08 05:38:35.000000000","reason":"\u003cGERRIT_ACCOUNT_35133\u003e replied on the change","reason_account":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-08-08 06:49:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0756ba2146b7825cfe79c282d0b0d7636f24a6fb","subject":"fix forgot space in phrases","status":"MERGED","created":"2025-08-05 00:52:41.000000000","updated":"2025-08-08 06:51:21.000000000","submitted":"2025-08-08 06:49:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"956522","meta_rev_id":"3e99c8c4f09294b17efcbfe5ebcdbe4908b8f320","_number":956522,"virtual_id_number":956522,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-08 06:49:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"recommended":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"all":[{"value":1,"date":"2025-08-05 09:05:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":2,"date":"2025-08-07 16:40:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-08-08 05:38:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-08-08 05:38:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"}],"reviewers":{"REVIEWER":[{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-05 00:55:07.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"state":"REVIEWER"},{"updated":"2025-08-05 00:55:07.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"state":"REVIEWER"},{"updated":"2025-08-05 00:55:07.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"state":"REVIEWER"},{"updated":"2025-08-05 00:55:07.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2025-08-05 02:05:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b48c0b2766ef5fffc4aa9cb7ce23c8dec8772fbe","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-08-05 00:52:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0e224e728e68e8b88f65b46853c0c0a17d1f28d4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-08-05 00:54:00.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"b0b8137935d567e87574993f5efc307c41c94043","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-08-05 00:54:39.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"5efb85259234940b8c797c855f66e735fbde6dad","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-08-05 00:54:45.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"81f542e2f54eb21f686d39c35e7b7cd10fb66d8d","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-08-05 00:55:07.000000000","message":"Patch Set 4:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"90cc6617111fa9cb1823646e07e34bfa30f41fcd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-08-05 01:13:55.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"6f6105b7afe85ff5fe4236e44d88ba5bb85074f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-08-05 01:15:04.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"f83478633b4e8643a5f59898ab603aceafef60d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-05 02:05:28.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24b85451f7274e87a4fe56ba06fa5be0\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/41ed03b354084aa49191048a2bf05a75 : SUCCESS in 5m 47s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/5bb47e1966524cf191cd7ca86614e936 : SUCCESS in 3m 22s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/df7f4d1020ba4881a0c164b6018b8cc0 : SUCCESS in 4m 11s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/e9f9f5b5c3594f878161a376cc2c0f89 : SUCCESS in 7m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/17cd248c931f4548bd22e6137bcfe69a : SUCCESS in 15m 38s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/37e44fdf08b74b08a1b9cd569ca69b62 : SUCCESS in 17m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35d515ce6af2497b80ef87b0ecea2b08 : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/314cad3b7e45443680494f14e89ed11d : SUCCESS in 13m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/45ac492c94b94cfda12527edac6a018f : SUCCESS in 20m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b1c3c97a01041cabfc3804dafdbe3bf : SUCCESS in 14m 28s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/5a2e114c39b44aa680e3e3df322be761 : SUCCESS in 7m 31s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/8cfe660eb68d4a0cb5a9a6c824262d1d : SUCCESS in 49m 48s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/22ee57f090cc4c798ba3cb3b81495d17 : SUCCESS in 28m 09s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/e1b4c83661aa4ac7b57940c5f93ab9ee : SUCCESS in 29m 08s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7b7bc660704d4888b70018f73f2825cd : SUCCESS in 4m 11s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/f5dc50c648164c3683dd858d28d17274 : SUCCESS in 29m 43s\n- horizon-dsvm-tempest-plugin-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4c0684f25dfc4925be81be160ca889dc : SUCCESS in 24m 23s","accounts_in_message":[],"_revision_number":6},{"id":"27a6740f5df6e102dd0920d1bdc2cad3ef406b2a","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-08-05 09:05:02.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"2670ecec802eb399f66658642c465d1206cbc75f","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2025-08-07 16:40:13.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"39aa3781b3e1183e975096374c7f82ab393cde7a","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2025-08-08 05:38:35.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"bb200b0b08a0337fd69cb8ffada6638de185e132","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-08 05:38:57.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"6a4106dcd31194f36077af6c9e38d9af6c4ea7aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-08 06:49:55.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a45943e3d6ad4ca6869abf2f796b8113\n\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/a10c9fc6d9714cc5b87697620e4592d1 : SUCCESS in 5m 08s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/8bcea54fc0d745469fb9ab4e02c20592 : SUCCESS in 6m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/613ac7ef1e974f4baccf223fde3ccd84 : SUCCESS in 17m 17s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/0b6e82ce948b45eebc7a596e87090b69 : SUCCESS in 17m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22ca5bde9acb4832bd2279ba3718026e : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/49a6563534bb4a3f866d45553830e528 : SUCCESS in 13m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9489fd0d520443cdbd464c1432c7eef2 : SUCCESS in 17m 36s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/c39ce462506a4679933def779a51749d : SUCCESS in 23m 24s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/358eac290a8841599259750452fcd7a7 : SUCCESS in 5m 59s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/14c379d9047941428c4ca5c249422701 : SUCCESS in 1h 09m 13s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/539a60d16dd748b09413ffb7cab3849d : SUCCESS in 15m 27s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/20225e53e2f44353b294af466164fe53 : SUCCESS in 11m 34s\n- horizon-dsvm-tempest-plugin-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a6d72abcd9f4bbaa9ede80c354d9986 : SUCCESS in 10m 54s","accounts_in_message":[],"_revision_number":6},{"id":"78e322afaf9bea4d893f64421e7f22968547b5ba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-08 06:49:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"3e99c8c4f09294b17efcbfe5ebcdbe4908b8f320","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-08 06:51:21.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3669f033f7274b6abd1188bde80219f3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2849c540f9c431b8405e8a62fc99331 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"6ca2c873fd12bf5440dc82ff0db7b2244aac8179","revisions":{"d81c90b7f504a79de6c81da88336d4d9c419180c":{"kind":"REWORK","_number":1,"created":"2025-08-05 00:52:41.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/22/956522/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/22/956522/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/22/956522/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/22/956522/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/22/956522/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/22/956522/1"}}},"commit":{"parents":[{"commit":"1f06d743699fe84d1bf2a0dacaa2e035771133bf","subject":"Merge \"pytest-based tests individual timeout for message check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/1f06d743699fe84d1bf2a0dacaa2e035771133bf"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-08-05 00:52:41.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-08-05 00:52:41.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I0756ba2146b7825cfe79c282d0b0d7636f24a6fb\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/d81c90b7f504a79de6c81da88336d4d9c419180c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/d81c90b7f504a79de6c81da88336d4d9c419180c"}]},"branch":"refs/heads/master"},"bac37da6b3686c8e37865221daf987e9468eb44a":{"kind":"REWORK","_number":2,"created":"2025-08-05 00:54:00.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/22/956522/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/22/956522/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/22/956522/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/22/956522/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/22/956522/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/22/956522/2"}}},"commit":{"parents":[{"commit":"1f06d743699fe84d1bf2a0dacaa2e035771133bf","subject":"Merge \"pytest-based tests individual timeout for message check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/1f06d743699fe84d1bf2a0dacaa2e035771133bf"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-08-05 00:52:41.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-08-05 00:54:00.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I0756ba2146b7825cfe79c282d0b0d7636f24a6fb\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/bac37da6b3686c8e37865221daf987e9468eb44a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/bac37da6b3686c8e37865221daf987e9468eb44a"}]},"branch":"refs/heads/master"},"ca031786d8bfbb4e119b9e133a080aead83e61c2":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-08-05 00:54:39.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/22/956522/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/22/956522/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/22/956522/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/22/956522/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/22/956522/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/22/956522/3"}}},"commit":{"parents":[{"commit":"1f06d743699fe84d1bf2a0dacaa2e035771133bf","subject":"Merge \"pytest-based tests individual timeout for message check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/1f06d743699fe84d1bf2a0dacaa2e035771133bf"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-08-05 00:52:41.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-08-05 00:54:39.000000000","tz":0},"subject":"fix forgot space in phrase Change-Id: I0756ba2146b7825cfe79c282d0b0d7636f24a6fb Signed-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e","message":"fix forgot space in phrase\nChange-Id: I0756ba2146b7825cfe79c282d0b0d7636f24a6fb\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/ca031786d8bfbb4e119b9e133a080aead83e61c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/ca031786d8bfbb4e119b9e133a080aead83e61c2"}]},"branch":"refs/heads/master","description":"Edit commit message"},"ceec0112f4c14ac8fb00f5964b350d75927fa61c":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-08-05 00:54:45.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/22/956522/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/22/956522/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/22/956522/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/22/956522/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/22/956522/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/22/956522/4"}}},"commit":{"parents":[{"commit":"1f06d743699fe84d1bf2a0dacaa2e035771133bf","subject":"Merge \"pytest-based tests individual timeout for message check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/1f06d743699fe84d1bf2a0dacaa2e035771133bf"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-08-05 00:52:41.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-08-05 00:54:45.000000000","tz":0},"subject":"fix forgot space in phrase","message":"fix forgot space in phrase\n\nChange-Id: I0756ba2146b7825cfe79c282d0b0d7636f24a6fb\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/ceec0112f4c14ac8fb00f5964b350d75927fa61c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/ceec0112f4c14ac8fb00f5964b350d75927fa61c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"47a6c23d956e042b54e4707e4df399d672ae27de":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2025-08-05 01:13:55.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/22/956522/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/22/956522/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/22/956522/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/22/956522/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/22/956522/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/22/956522/5"}}},"commit":{"parents":[{"commit":"1f06d743699fe84d1bf2a0dacaa2e035771133bf","subject":"Merge \"pytest-based tests individual timeout for message check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/1f06d743699fe84d1bf2a0dacaa2e035771133bf"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-08-05 00:52:41.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-08-05 01:13:55.000000000","tz":0},"subject":"fix forgot space in phrases","message":"fix forgot space in phrases\n\nChange-Id: I0756ba2146b7825cfe79c282d0b0d7636f24a6fb\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/47a6c23d956e042b54e4707e4df399d672ae27de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/47a6c23d956e042b54e4707e4df399d672ae27de"}]},"branch":"refs/heads/master","description":"Edit commit message"},"6ca2c873fd12bf5440dc82ff0db7b2244aac8179":{"kind":"REWORK","_number":6,"created":"2025-08-05 01:15:04.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/22/956522/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/22/956522/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/22/956522/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/22/956522/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/22/956522/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/22/956522/6"}}},"commit":{"parents":[{"commit":"1f06d743699fe84d1bf2a0dacaa2e035771133bf","subject":"Merge \"pytest-based tests individual timeout for message check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/1f06d743699fe84d1bf2a0dacaa2e035771133bf"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-08-05 00:52:41.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-08-05 01:15:04.000000000","tz":0},"subject":"fix forgot space in phrases","message":"fix forgot space in phrases\n\nChange-Id: I0756ba2146b7825cfe79c282d0b0d7636f24a6fb\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/6ca2c873fd12bf5440dc82ff0db7b2244aac8179"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/6ca2c873fd12bf5440dc82ff0db7b2244aac8179"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
