)]}'
{"id":"openstack%2Fhorizon~960204","triplet_id":"openstack%2Fhorizon~master~I013972aac7a6ed998bb33513024e06039232d1d4","project":"openstack/horizon","branch":"master","topic":"integration-tests-cleanup","attention_set":{},"removed_from_attention_set":{"32432":{"account":{"_account_id":32432,"name":"Owen McGonagle","email":"omcgonag@redhat.com","username":"omcgonag"},"last_update":"2025-10-21 17:16:30.000000000","reason":"Change was submitted"},"6914":{"account":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"last_update":"2025-10-21 15:45:56.000000000","reason":"\u003cGERRIT_ACCOUNT_6914\u003e replied on the change","reason_account":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"}},"8648":{"account":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"last_update":"2025-10-21 14:45:21.000000000","reason":"\u003cGERRIT_ACCOUNT_8648\u003e replied on the change","reason_account":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"}},"35133":{"account":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"last_update":"2025-10-21 17:16:30.000000000","reason":"Change was submitted"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-10-21 17:16:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I013972aac7a6ed998bb33513024e06039232d1d4","subject":"Remove all dependencies/connections of old integration test code","status":"MERGED","created":"2025-09-08 21:13:07.000000000","updated":"2025-10-21 17:18:07.000000000","submitted":"2025-10-21 17:16:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"960204-integration-tests-cleanup","meta_rev_id":"12d196bb8184610cc6218befb57b00b5e0af7360","_number":960204,"virtual_id_number":960204,"owner":{"_account_id":32432,"name":"Owen McGonagle","email":"omcgonag@redhat.com","username":"omcgonag"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-21 17:16:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"all":[{"value":2,"date":"2025-10-21 15:45:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":2,"date":"2025-10-21 14:45:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"all":[{"value":1,"date":"2025-10-21 15:46:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}],"reviewers":{"REVIEWER":[{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}],"CC":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-08 21:36:40.000000000","updated_by":{"_account_id":32432,"name":"Owen McGonagle","email":"omcgonag@redhat.com","username":"omcgonag"},"reviewer":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"state":"REVIEWER"},{"updated":"2025-09-08 21:36:40.000000000","updated_by":{"_account_id":32432,"name":"Owen McGonagle","email":"omcgonag@redhat.com","username":"omcgonag"},"reviewer":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"state":"REVIEWER"},{"updated":"2025-09-08 21:36:40.000000000","updated_by":{"_account_id":32432,"name":"Owen McGonagle","email":"omcgonag@redhat.com","username":"omcgonag"},"reviewer":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"state":"REVIEWER"},{"updated":"2025-09-08 22:07:07.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"CC"},{"updated":"2025-09-08 22:07:07.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2025-09-08 22:12:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"655dd9556b5abe33d0b87fc0c926c4de20e60150","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32432,"name":"Owen McGonagle","email":"omcgonag@redhat.com","username":"omcgonag"},"date":"2025-09-08 21:13:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70a143b5751058f29ddb22fd53cc442ba9f6720e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32432,"name":"Owen McGonagle","email":"omcgonag@redhat.com","username":"omcgonag"},"date":"2025-09-08 21:35:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5cd7d8a0ce1c914852f5ed0777d60b4737b45d27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 22:12:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d980f71aa01543568fffee3071c923d2\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9cfe4009719e44b496dcfb40f7976175 : SUCCESS in 1m 56s\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/825f4e7bb26041c69dcf85fc7b3e3e3f : SUCCESS in 6m 41s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/567d3b91e407490fb4a980bfd7adb6ca : SUCCESS in 2m 18s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/e54a33cf0a1c43fb8ddafa9df3d07540 : SUCCESS in 5m 24s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/6748276f50e04bf59f28e22b09bd97e0 : SUCCESS in 4m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/89a42c6a19ee47b1bc5ec42306191e68 : SUCCESS in 8m 05s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/54e64d028fcc42199ff95c84356e4bf1 : SUCCESS in 9m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0be8c4ed32b1477db2dbad0cce431bdb : SUCCESS in 5m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/05c59bdb2c8a44cf9f32dd80e452cf88 : SUCCESS in 9m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/24bd765b0adf4f2da644bc5ba4f961fd : SUCCESS in 16m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/046f0838b0ef433a96264de8b661dcef : SUCCESS in 8m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3acf90c3009e4210b0738b248c212acf : SUCCESS in 11m 50s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/9f3ff880aa224b05b81eb47993790f3e : SUCCESS in 7m 23s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/4e69eb3237ba4b56ab4542de8098c4de : FAILURE in 11m 48s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/0e32d8e92fd64d54bc1adee51615e1e4 : SUCCESS in 31m 53s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/336b89dc901b4ec0856adfe69ed90569 : SUCCESS in 4m 27s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/78445a862ed648178aa2643077ace797 : SUCCESS in 28m 20s","accounts_in_message":[],"_revision_number":2},{"id":"90c499a140d4cbc27120f91025e628e71d003197","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-09-08 22:16:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"58a0bf84420ddf98e2938746940718a85e3aaa64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32432,"name":"Owen McGonagle","email":"omcgonag@redhat.com","username":"omcgonag"},"date":"2025-09-09 01:06:37.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"ccb504014eec0fd2f6a1fd7a5b15a055be50a49a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32432,"name":"Owen McGonagle","email":"omcgonag@redhat.com","username":"omcgonag"},"date":"2025-09-09 01:24:16.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2b5b928fa2e14214a954e78c4474cc3cd3f5198a","author":{"_account_id":32432,"name":"Owen McGonagle","email":"omcgonag@redhat.com","username":"omcgonag"},"date":"2025-09-09 01:30:09.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"530dfb1e4bc5b3bea0063ea0f1dc2d1941995ad2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-09 01:58:02.000000000","message":"Patch Set 4: Verified-1\n\n(35 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/90208b874b5345cba3a34a26033fbe59\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7ba48a1e430448978ffb9a1e8f9e6028 : SUCCESS in 2m 40s\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/7f72f682bc6648dfb41d171f942a39e0 : SUCCESS in 3m 30s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/f8db1c490b9648b9ad77820d6ec2016f : SUCCESS in 3m 40s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/f0beed90ce914af4b8344bc1db56732c : SUCCESS in 6m 37s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/51537ce1667f45c6b048ae54b94114b0 : SUCCESS in 6m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3c6fb4532ea44ed4974d0f9b8a1844b4 : SUCCESS in 19m 13s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/cfe00861916c45909d0587808e78b6a7 : SUCCESS in 19m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88b7b2559a2a4f53994458594337f55a : FAILURE in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/367aec7ce068444a88c2e864988f0afa : SUCCESS in 18m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a27e970dedbd4c0e9b05cefe3d8426fd : SUCCESS in 16m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e19f23710cef4cf99326fd7be70c8bdf : SUCCESS in 15m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c1afc6c1d3e42d5b9e7c9601b004ed2 : SUCCESS in 10m 00s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/2cc8e2ffa59c440cafec270f546b61b3 : SUCCESS in 9m 26s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/762d511bdece44bb88c6117558a82785 : FAILURE in 26m 40s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/57a00d577a694bae919def6f8ee6d8a8 : SUCCESS in 28m 18s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9b48211e9e564ff2ae3525959dcba80f : SUCCESS in 4m 57s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/048791b2a5c143afafcc7d989cfd442f : SUCCESS in 28m 44s","accounts_in_message":[],"_revision_number":4},{"id":"6840f685d4af09da15110f662b78365bb2808496","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32432,"name":"Owen McGonagle","email":"omcgonag@redhat.com","username":"omcgonag"},"date":"2025-09-09 13:15:24.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"78e26714a7033472d014130b4d0f58efe34308ce","author":{"_account_id":32432,"name":"Owen McGonagle","email":"omcgonag@redhat.com","username":"omcgonag"},"date":"2025-09-09 13:16:33.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"676bb02f03f458f0553a8f96702b0fa71f0be6e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-09 13:51:09.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4e84c48878924792a21a2986c9fe38cc\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d6e66fab28d94ee8b1b52eb7f193221a : SUCCESS in 2m 02s\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/ea1133e9227b45cf97b03e1cf99959ab : SUCCESS in 4m 13s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/7ac2117dadfe4411925595cedd041ff5 : SUCCESS in 3m 29s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/822db2ccef59403c8794246a5798482a : SUCCESS in 4m 42s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/69e92ebc55894b6e89734ab17364701f : SUCCESS in 6m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a558868b8189478d9d9c0489d1d804f5 : SUCCESS in 20m 04s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/cd96680d5cf84b1ca6b4e93a5447cfde : SUCCESS in 15m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff5d167f90d54f0488f2dcdfcec5d3fd : SUCCESS in 5m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b21f4bfcb5d6410889f0bf6d631b7ca1 : SUCCESS in 16m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/71183f2f3d6849d5b5e136935579b39e : SUCCESS in 17m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0bd8e5019d547c08784d4249d9fda08 : SUCCESS in 7m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea342813cfb84a65935ab01a2e055821 : SUCCESS in 15m 55s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/2f0625efab4744ceac9e599956002d2d : SUCCESS in 8m 46s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/506c5be090d7419e9dc1c02caa82f689 : FAILURE in 21m 57s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/f73565e5ad7645d88438397d2c41994a : SUCCESS in 25m 20s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/93cd2441337d464dbf1e648b218aa7e7 : SUCCESS in 3m 44s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/913a228b8fac495a89789818748ac9ce : SUCCESS in 30m 25s","accounts_in_message":[],"_revision_number":5},{"id":"4fc85a8f5fcd93fe290b1fb0b9a5ce3cdc9678ef","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2025-09-09 14:03:43.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5dfb46b5e03480083af1e5609db2e61d6614df58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32432,"name":"Owen McGonagle","email":"omcgonag@redhat.com","username":"omcgonag"},"date":"2025-09-09 18:24:53.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"58d989b22e7e895d28a91fdb08a6cc03406751e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-09 19:07:31.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42136333b5494c809c469b7a9f0951e4\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/c93d318e5d44486e99441e96069b8d0c : SUCCESS in 3m 57s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/dc6c6cf01db943a08f97cac4ff31172c : SUCCESS in 3m 32s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/62addda85f4c4e9fa4ed659892995727 : SUCCESS in 4m 42s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/887336ca7a5f4e54a1415dcc18033a38 : SUCCESS in 6m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8c935daf431e464480552f09e7b36b2e : SUCCESS in 17m 55s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/0f37215cd64d48b5b103dbbe42f2efe0 : SUCCESS in 16m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfaec63f853044c48294a45ef29ebe3c : SUCCESS in 6m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/37b8d44b379743889ec2c794c59c987b : SUCCESS in 17m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6fb9bf8ffe0743acaab8786aa013de60 : SUCCESS in 23m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f41671f66ca14a3895a763c6dea73e59 : SUCCESS in 13m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8785edc5836d41389c2fa5671dc6d1c4 : SUCCESS in 8m 52s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/52761d1d07254b6aa3a75527ad341f7e : SUCCESS in 6m 42s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/ad0b85d0dde545d6ae485c84432dceff : SUCCESS in 29m 55s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/bd6717d3805e4da38249fa7cd198ba3f : SUCCESS in 30m 35s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1e4a7228d8854312a7b31bc687191d70 : SUCCESS in 2m 33s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/f64422cd51af4f45bb1964316bf1e505 : SUCCESS in 11m 54s","accounts_in_message":[],"_revision_number":6},{"id":"3d5cc95ce6618fdfe1cb68fcb9abe8166cf32f9c","author":{"_account_id":32432,"name":"Owen McGonagle","email":"omcgonag@redhat.com","username":"omcgonag"},"date":"2025-09-09 19:33:12.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"accb9d17582cdff21964b6bc5c42b5d3cb187ecb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32432,"name":"Owen McGonagle","email":"omcgonag@redhat.com","username":"omcgonag"},"date":"2025-09-10 13:24:08.000000000","message":"Patch Set 7: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"97f6a195b066f50e62b86055f42d336f67c7ea7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 14:03:57.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88094e71b0cb49d297129ca8e6c6f0b0\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/e5a0219458c54dbfbd37f630375f39c5 : SUCCESS in 5m 41s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/f00d5c1f96c74124b3dfa8444689ff04 : SUCCESS in 3m 33s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/d6c7bae19be94d5baa79778bd963d159 : SUCCESS in 6m 50s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/4043f87680274a61a851e9d084aa6128 : SUCCESS in 5m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d4378db03b624bdf96608d3e7a793d49 : SUCCESS in 18m 29s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/348893fd223a41e388c20defcdfda719 : SUCCESS in 17m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c0a775d83f34874b0c9e7db85a59f60 : SUCCESS in 6m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bd7d1185026f4a12929aa35eb8f1df6e : SUCCESS in 16m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fcb9055c15ab4296ae5fca27e4c84ed9 : SUCCESS in 10m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d06bc8cbde8b44b9997450c0afc45339 : SUCCESS in 7m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9984b13d060941b58771da1be0458fee : SUCCESS in 17m 35s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/adc8b8d25ca74e2eb3291f820e377ea9 : SUCCESS in 6m 38s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/3919a9f707d24e09bab10fa811fd867c : SUCCESS in 26m 29s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/c2721303ab3247ae9f66b609310c5712 : SUCCESS in 31m 41s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b3b3f13d7b38448cbc563aa759cc6f09 : SUCCESS in 3m 21s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/9c4d69f84dab4142a7dec369bf118d9b : SUCCESS in 12m 16s","accounts_in_message":[],"_revision_number":7},{"id":"c85c0b27b83709866f93106ed62a78e5c8732323","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32432,"name":"Owen McGonagle","email":"omcgonag@redhat.com","username":"omcgonag"},"date":"2025-09-10 18:57:10.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"c7bb536e711e152be6d5f85c5790ff651c4130e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 19:34:23.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8b4e732c0e7f4773b462212d6b4df5bd\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/91a3ee3574be42fe8e523e58bd3343eb : SUCCESS in 2m 48s\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/1762ccafbf1e4517a2fcfac83589b01f : SUCCESS in 6m 32s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/7d28331c3ed2491f93928caee57db431 : SUCCESS in 3m 42s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/e063964e94a84eeab7f338d90224f671 : SUCCESS in 4m 07s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/6522abd865eb4b9a919c5c222f934c16 : SUCCESS in 6m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/353f3ee26be14117893810b2dfe26a75 : SUCCESS in 18m 59s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/fc2d81f2aa6d4054a65282b3feb5061f : SUCCESS in 8m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/035fe67ed3aa421fa66628c35c0a2bce : SUCCESS in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ad9a3a44d5b343bdb3de28a01241c159 : SUCCESS in 17m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eda3a53a934f41a78cf83abcfb2654fb : SUCCESS in 23m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6311b1922dff45ea94b1951569ba6242 : SUCCESS in 8m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/193baa7e5b13468ba4409a1f1f8f4ee7 : SUCCESS in 12m 17s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/10687166310040148076f679a938a0c4 : SUCCESS in 5m 15s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/1092110f301b4cad92a7e2bfc2e64c28 : FAILURE in 24m 04s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/ff7b4e0601f94114aca379b3715caa91 : SUCCESS in 28m 57s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/311ee923d1914236ab79be2e4eba2434 : SUCCESS in 29m 48s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2328be7fc3134f6686592adeb32e75e1 : SUCCESS in 5m 08s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/f49d08488fc9474386ceb65dc1040b71 : SUCCESS in 32m 17s","accounts_in_message":[],"_revision_number":8},{"id":"e1b7ee2aa720ac52dd60a3cebc9b38afa79e1b69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32432,"name":"Owen McGonagle","email":"omcgonag@redhat.com","username":"omcgonag"},"date":"2025-09-10 20:49:55.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"00fe12aba88a72aabaa71515a62dc5dfb7774e8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 21:45:45.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dafed5386b83441da00c4846e2a5bf92\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8015d9773fa249fcbfa1753ec368aa65 : SUCCESS in 2m 41s\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/d30175f18d0b4024acdec6278249e9c5 : SUCCESS in 6m 11s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/4d5b75e8fedb44fab1c1ce5a18a6a143 : SUCCESS in 3m 44s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/f88798254a524bd591c7c95ef9177c82 : SUCCESS in 5m 04s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/6298ebec85534cf08de39b14132beda8 : SUCCESS in 8m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9d5ec042a5fd4df79e2963bf0a56b8cb : SUCCESS in 13m 06s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/1b3423231f5142a3a5ccd78c1fdf777f : SUCCESS in 18m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4161827f1fd24bd7a9974d12c12fd7ad : SUCCESS in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0df08db41ed14a2b96484d431f11b526 : SUCCESS in 8m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4cfafb0d9feb4263b45ecab6ba713fec : SUCCESS in 24m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3fd29b823bf40dcbdfd7059a8ba537e : SUCCESS in 12m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5893aa1ddde24ebb82d58356803a1aa1 : SUCCESS in 11m 47s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/aefc7e04e6864a8bbeca0b3011540502 : SUCCESS in 8m 21s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/5efb09a4844b45a484980b49683318a8 : SUCCESS in 54m 34s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/632365c9285748c4b3f608f0fc61e7ee : SUCCESS in 19m 10s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/93e381d3dfba4c978be6e61b5cf1c1c2 : SUCCESS in 28m 32s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/84e82633597848cc940338b92668ca86 : SUCCESS in 3m 59s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/d485843332d14af48b3e66ee2edc1ffd : SUCCESS in 14m 11s","accounts_in_message":[],"_revision_number":9},{"id":"4f5b275eb01fcf1cf0fc295f860cbcc7ae9fc858","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32432,"name":"Owen McGonagle","email":"omcgonag@redhat.com","username":"omcgonag"},"date":"2025-10-09 20:53:02.000000000","message":"Patch Set 10: Patch Set 9 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"0c9246b5689ee5a26aa86d03f4b374d1aa9a5f9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32432,"name":"Owen McGonagle","email":"omcgonag@redhat.com","username":"omcgonag"},"date":"2025-10-09 21:08:57.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"4af0475a0e535da29eae050b100e03c50174565c","author":{"_account_id":32432,"name":"Owen McGonagle","email":"omcgonag@redhat.com","username":"omcgonag"},"date":"2025-10-09 21:15:28.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"384fbdaa26ee0a4d3f85c270d39604a2b955aef3","author":{"_account_id":32432,"name":"Owen McGonagle","email":"omcgonag@redhat.com","username":"omcgonag"},"date":"2025-10-09 21:18:01.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"380fbbc07ce3e1d3a0f64da05ff91efa9ae93863","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 22:21:40.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57bd4721388a4e6ba5f9c40f9ef74eff\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/97253ca0ea45416f8e5e7dc0169159ed : SUCCESS in 2m 20s\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/7da536f7eb6941bca6e8cddf0e68ea76 : SUCCESS in 5m 31s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/a845ace0b7e04d99a7c16fc54d86e38a : SUCCESS in 4m 02s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/fcba705ffc7048f8836f3c850a4b14da : SUCCESS in 6m 47s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/36b896bddcb742d9ac3693cc89cd4ecb : SUCCESS in 6m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/736560f214a7437f8f15f0c8ed37064a : SUCCESS in 16m 13s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/e46437045bae49f2ab808ed4636743c3 : SUCCESS in 19m 34s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/1abeccfad70845c593d703b31ed8a5d7 : FAILURE in 18m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8f9c68221394aa1b16372b02c17fd3c : SUCCESS in 6m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c768f767b7064b47a8fe6c626bee40b5 : SUCCESS in 17m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b8da4ac3809743ada08afd8569bd88ba : SUCCESS in 9m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c9984432e374ed3a8a3b0fe11af14a4 : SUCCESS in 7m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86281925a9164ae49f999b458da397e9 : SUCCESS in 8m 48s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/2745c469bbfd45b9ba35e50bb277cfc2 : SUCCESS in 6m 42s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/f3d7e3e0754248ceb3d5414818db9b78 : SUCCESS in 1h 10m 40s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/0ef95d0ae5ce4dcca9f3015159296e74 : SUCCESS in 12m 19s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/0a9d859f5c8242e197de48f2041f5b49 : SUCCESS in 29m 36s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e146732d37ce46a49ac7b149aa836c53 : SUCCESS in 4m 34s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/6eeca9c55ad646b09a4f232dc084c559 : SUCCESS in 20m 52s","accounts_in_message":[],"_revision_number":11},{"id":"dc694b8f01de6146ddf53e35221b2830642d61af","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2025-10-10 07:13:06.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"7f5d74478a595f98a32ff8456d1db0a7a0b9069c","author":{"_account_id":32432,"name":"Owen McGonagle","email":"omcgonag@redhat.com","username":"omcgonag"},"date":"2025-10-10 14:43:18.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"608637f44691044996c68bf3a2a0c85b7cf8460b","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2025-10-13 08:38:35.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"a13cc942c3b4357978a0d590a6ba2fd596a6854c","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2025-10-14 20:14:57.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"48d05c50e0fbce5d77e84f986a6a559ce1b5468e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32432,"name":"Owen McGonagle","email":"omcgonag@redhat.com","username":"omcgonag"},"date":"2025-10-17 19:33:05.000000000","message":"Patch Set 12: Patch Set 11 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"172d1afb0e67c6e647104e2c95f6b43bebc12cda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32432,"name":"Owen McGonagle","email":"omcgonag@redhat.com","username":"omcgonag"},"date":"2025-10-17 19:40:36.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":13},{"id":"c90a0df47e30e33e72fd035c53c94b074ef3c87f","author":{"_account_id":32432,"name":"Owen McGonagle","email":"omcgonag@redhat.com","username":"omcgonag"},"date":"2025-10-17 19:41:58.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"7501bd75c33586b267f0d0b182d6b4ad4eae4afa","author":{"_account_id":32432,"name":"Owen McGonagle","email":"omcgonag@redhat.com","username":"omcgonag"},"date":"2025-10-17 20:10:33.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"20e1085b18847f259cfbf4aba30068b4655f8398","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 21:04:51.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c13959dbd68412ba21db9859af739eb\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/5d40af7e163041b0a89221ced755c209 : SUCCESS in 7m 39s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/a82f66802db54cf5bc140de5a1e3c816 : SUCCESS in 3m 41s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/a9566c1841d04597b82e68252e5fb2e0 : SUCCESS in 3m 00s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/e1deb3a403574308b60b5185df577bf6 : SUCCESS in 7m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9c2cd3a8f49349bd9a00f56a17cfe414 : SUCCESS in 17m 51s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/5971382f037546c3b26805444b805063 : SUCCESS in 8m 08s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/17319968a71548e5a38b642a854f4e49 : FAILURE in 8m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cc983cbdfe84e64b83efe863fecdf9d : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/57c76527a6eb407ca028f810867bec27 : SUCCESS in 19m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/432349878ca44cf6aa8481a5f04ed7e6 : SUCCESS in 15m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc2654399c6f403992947c51dfb151e1 : SUCCESS in 8m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9aa1632e9abb4ec8b61fe43d34aa2c9d : SUCCESS in 18m 23s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/9c4f5ccc11f64f468dba24a365d38c4f : SUCCESS in 6m 20s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/a3a87e8c1a24417b947e173a22c9daff : SUCCESS in 1h 19m 44s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/5beb05c0e5854b389f6df6b0acd9c6c8 : SUCCESS in 31m 56s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/bc5a77122f604b3d99132b35d9bfb768 : SUCCESS in 16m 13s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5181825ba385422c9f12de95f87b2e7c : SUCCESS in 4m 25s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/73e6dfb6f6db41d4bc382dc2a1f16ee4 : SUCCESS in 11m 44s","accounts_in_message":[],"_revision_number":13},{"id":"3f2cf98d1f1c658a3a35ba2721397d61f2d65e07","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2025-10-21 14:45:21.000000000","message":"Patch Set 13: Code-Review+2","accounts_in_message":[],"_revision_number":13},{"id":"7fac3ebeede709a36b63705b00d5581a9f85a67f","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2025-10-21 15:45:56.000000000","message":"Patch Set 13: Code-Review+2","accounts_in_message":[],"_revision_number":13},{"id":"819f2199ddf7ac092e1f0919160594aa9d434c8f","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2025-10-21 15:46:20.000000000","message":"Patch Set 13: Workflow+1","accounts_in_message":[],"_revision_number":13},{"id":"c40634186cfe5dada38be221617f5b2c5d7dbeb0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-21 15:46:43.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"d4d5043a3e99a9fcac8e13514ac883eae3d67137","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-21 17:16:30.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/709b3329db3448bcb45bb8c76cf11176\n\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/0be86f8c3c844f29b2281f7f84daf567 : SUCCESS in 5m 18s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/44232397d0924f2dbfb23cf0b88e956e : SUCCESS in 6m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/036a9ce97eb742b895f50379989eb78d : SUCCESS in 19m 12s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/b07d8dbc2344498ca8ec98b4799881ba : SUCCESS in 13m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc664a87de864846bd6ecc1f7db30b8c : SUCCESS in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eb75f739da444f978c49b416a2509483 : SUCCESS in 19m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/da20285a3bb845ebbcc95180c6a26fa6 : SUCCESS in 10m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e93e52b5ad0148e9bed6d57a5813c008 : SUCCESS in 8m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e0e650cb7ea428e9175424cf98ad491 : SUCCESS in 13m 21s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/d30b02e4a5924b4e91d6b612d28b5286 : SUCCESS in 28m 13s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/dc473359d7a84b1892aa3aa0297aaf48 : SUCCESS in 6m 35s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/385102091c75458582b332e7e621419f : SUCCESS in 1h 22m 57s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/761d56b10efe48bbbf3eaad6b0f34275 : SUCCESS in 32m 07s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/4cd064b5554a40c28739692eef636e8a : SUCCESS in 15m 44s","accounts_in_message":[],"_revision_number":13},{"id":"de519a1b5143d38fcf3ad2f043266df1ac82fdb6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-21 17:16:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":13},{"id":"12d196bb8184610cc6218befb57b00b5e0af7360","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-21 17:18:07.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3b04f71ffcd470681459290d0732144\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28da86be6e554183919ee3b4aa102872 : SUCCESS in 1m 24s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f53cde4c5944acf8c8fea18da97047e : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"49e5fe185a915ab80ccf4c130225371ade323711","revisions":{"c0e05bfdeafaa947217f5cf07d55f2cf751f86db":{"kind":"REWORK","_number":1,"created":"2025-09-08 21:13:07.000000000","uploader":{"_account_id":32432,"name":"Owen McGonagle","email":"omcgonag@redhat.com","username":"omcgonag"},"ref":"refs/changes/04/960204/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/04/960204/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/04/960204/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/04/960204/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/04/960204/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/04/960204/1"}}},"commit":{"parents":[{"commit":"64c47ac369b35b42a17d7f6b00da81c6db46a843","subject":"Merge \"Revert \"SDK: Remove remaining neutronclient calls\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/64c47ac369b35b42a17d7f6b00da81c6db46a843"}]}],"author":{"name":"Owen McGonagle","email":"omcgonag@redhat.com","date":"2025-09-08 20:56:15.000000000","tz":-240},"committer":{"name":"Owen McGonagle","email":"omcgonag@redhat.com","date":"2025-09-08 21:12:52.000000000","tz":-240},"subject":"Remove all dependencies/connections of old integration test code","message":"Remove all dependencies/connections of old integration test code\n\nChange-Id: I013972aac7a6ed998bb33513024e06039232d1d4\nSigned-off-by: Owen McGonagle \u003comcgonag@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/c0e05bfdeafaa947217f5cf07d55f2cf751f86db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/c0e05bfdeafaa947217f5cf07d55f2cf751f86db"}]},"branch":"refs/heads/master"},"347c177935376d344479f84edd52a7c85a070342":{"kind":"REWORK","_number":2,"created":"2025-09-08 21:35:06.000000000","uploader":{"_account_id":32432,"name":"Owen McGonagle","email":"omcgonag@redhat.com","username":"omcgonag"},"ref":"refs/changes/04/960204/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/04/960204/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/04/960204/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/04/960204/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/04/960204/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/04/960204/2"}}},"commit":{"parents":[{"commit":"64c47ac369b35b42a17d7f6b00da81c6db46a843","subject":"Merge \"Revert \"SDK: Remove remaining neutronclient calls\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/64c47ac369b35b42a17d7f6b00da81c6db46a843"}]}],"author":{"name":"Owen McGonagle","email":"omcgonag@redhat.com","date":"2025-09-08 20:56:15.000000000","tz":-240},"committer":{"name":"Owen McGonagle","email":"omcgonag@redhat.com","date":"2025-09-08 21:34:50.000000000","tz":-240},"subject":"Remove all dependencies/connections of old integration test code","message":"Remove all dependencies/connections of old integration test code\n\nChange-Id: I013972aac7a6ed998bb33513024e06039232d1d4\nSigned-off-by: Owen McGonagle \u003comcgonag@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/347c177935376d344479f84edd52a7c85a070342"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/347c177935376d344479f84edd52a7c85a070342"}]},"branch":"refs/heads/master"},"2c66e4b8b8b7d5219ed6fe0d2d71030956e50ade":{"kind":"REWORK","_number":3,"created":"2025-09-09 01:06:37.000000000","uploader":{"_account_id":32432,"name":"Owen McGonagle","email":"omcgonag@redhat.com","username":"omcgonag"},"ref":"refs/changes/04/960204/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/04/960204/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/04/960204/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/04/960204/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/04/960204/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/04/960204/3"}}},"commit":{"parents":[{"commit":"64c47ac369b35b42a17d7f6b00da81c6db46a843","subject":"Merge \"Revert \"SDK: Remove remaining neutronclient calls\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/64c47ac369b35b42a17d7f6b00da81c6db46a843"}]}],"author":{"name":"Owen McGonagle","email":"omcgonag@redhat.com","date":"2025-09-08 20:56:15.000000000","tz":-240},"committer":{"name":"Owen McGonagle","email":"omcgonag@redhat.com","date":"2025-09-09 01:06:30.000000000","tz":-240},"subject":"Remove all dependencies/connections of old integration test code","message":"Remove all dependencies/connections of old integration test code\n\nChange-Id: I013972aac7a6ed998bb33513024e06039232d1d4\nSigned-off-by: Owen McGonagle \u003comcgonag@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/2c66e4b8b8b7d5219ed6fe0d2d71030956e50ade"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/2c66e4b8b8b7d5219ed6fe0d2d71030956e50ade"}]},"branch":"refs/heads/master"},"4912d5f8d38baaab3a8fb442da636538ea399620":{"kind":"REWORK","_number":4,"created":"2025-09-09 01:24:16.000000000","uploader":{"_account_id":32432,"name":"Owen McGonagle","email":"omcgonag@redhat.com","username":"omcgonag"},"ref":"refs/changes/04/960204/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/04/960204/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/04/960204/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/04/960204/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/04/960204/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/04/960204/4"}}},"commit":{"parents":[{"commit":"64c47ac369b35b42a17d7f6b00da81c6db46a843","subject":"Merge \"Revert \"SDK: Remove remaining neutronclient calls\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/64c47ac369b35b42a17d7f6b00da81c6db46a843"}]}],"author":{"name":"Owen McGonagle","email":"omcgonag@redhat.com","date":"2025-09-08 20:56:15.000000000","tz":-240},"committer":{"name":"Owen McGonagle","email":"omcgonag@redhat.com","date":"2025-09-09 01:24:07.000000000","tz":-240},"subject":"Remove all dependencies/connections of old integration test code","message":"Remove all dependencies/connections of old integration test code\n\nChange-Id: I013972aac7a6ed998bb33513024e06039232d1d4\nSigned-off-by: Owen McGonagle \u003comcgonag@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/4912d5f8d38baaab3a8fb442da636538ea399620"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/4912d5f8d38baaab3a8fb442da636538ea399620"}]},"branch":"refs/heads/master"},"8660241a449ea8fdd5973efcacc15e52671a1233":{"kind":"REWORK","_number":5,"created":"2025-09-09 13:15:24.000000000","uploader":{"_account_id":32432,"name":"Owen McGonagle","email":"omcgonag@redhat.com","username":"omcgonag"},"ref":"refs/changes/04/960204/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/04/960204/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/04/960204/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/04/960204/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/04/960204/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/04/960204/5"}}},"commit":{"parents":[{"commit":"64c47ac369b35b42a17d7f6b00da81c6db46a843","subject":"Merge \"Revert \"SDK: Remove remaining neutronclient calls\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/64c47ac369b35b42a17d7f6b00da81c6db46a843"}]}],"author":{"name":"Owen McGonagle","email":"omcgonag@redhat.com","date":"2025-09-08 20:56:15.000000000","tz":-240},"committer":{"name":"Owen McGonagle","email":"omcgonag@redhat.com","date":"2025-09-09 13:15:09.000000000","tz":-240},"subject":"Remove all dependencies/connections of old integration test code","message":"Remove all dependencies/connections of old integration test code\n\nChange-Id: I013972aac7a6ed998bb33513024e06039232d1d4\nSigned-off-by: Owen McGonagle \u003comcgonag@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/8660241a449ea8fdd5973efcacc15e52671a1233"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/8660241a449ea8fdd5973efcacc15e52671a1233"}]},"branch":"refs/heads/master"},"f96bd0da2daef7291358afa6e59319738ce33b7d":{"kind":"REWORK","_number":6,"created":"2025-09-09 18:24:53.000000000","uploader":{"_account_id":32432,"name":"Owen McGonagle","email":"omcgonag@redhat.com","username":"omcgonag"},"ref":"refs/changes/04/960204/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/04/960204/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/04/960204/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/04/960204/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/04/960204/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/04/960204/6"}}},"commit":{"parents":[{"commit":"64c47ac369b35b42a17d7f6b00da81c6db46a843","subject":"Merge \"Revert \"SDK: Remove remaining neutronclient calls\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/64c47ac369b35b42a17d7f6b00da81c6db46a843"}]}],"author":{"name":"Owen McGonagle","email":"omcgonag@redhat.com","date":"2025-09-08 20:56:15.000000000","tz":-240},"committer":{"name":"Owen McGonagle","email":"omcgonag@redhat.com","date":"2025-09-09 18:24:39.000000000","tz":-240},"subject":"Remove all dependencies/connections of old integration test code","message":"Remove all dependencies/connections of old integration test code\n\nChange-Id: I013972aac7a6ed998bb33513024e06039232d1d4\nSigned-off-by: Owen McGonagle \u003comcgonag@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/f96bd0da2daef7291358afa6e59319738ce33b7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/f96bd0da2daef7291358afa6e59319738ce33b7d"}]},"branch":"refs/heads/master"},"78d8112acc23a79639bde251b459a548220ba646":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2025-09-10 13:24:08.000000000","uploader":{"_account_id":32432,"name":"Owen McGonagle","email":"omcgonag@redhat.com","username":"omcgonag"},"ref":"refs/changes/04/960204/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/04/960204/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/04/960204/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/04/960204/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/04/960204/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/04/960204/7"}}},"commit":{"parents":[{"commit":"64c47ac369b35b42a17d7f6b00da81c6db46a843","subject":"Merge \"Revert \"SDK: Remove remaining neutronclient calls\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/64c47ac369b35b42a17d7f6b00da81c6db46a843"}]}],"author":{"name":"Owen McGonagle","email":"omcgonag@redhat.com","date":"2025-09-08 20:56:15.000000000","tz":-240},"committer":{"name":"Owen McGonagle","email":"omcgonag@redhat.com","date":"2025-09-10 13:24:08.000000000","tz":0},"subject":"[WIP] Remove all dependencies/connections of old integration test code","message":"[WIP] Remove all dependencies/connections of old integration test code\n\nChange-Id: I013972aac7a6ed998bb33513024e06039232d1d4\nSigned-off-by: Owen McGonagle \u003comcgonag@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/78d8112acc23a79639bde251b459a548220ba646"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/78d8112acc23a79639bde251b459a548220ba646"}]},"branch":"refs/heads/master","description":"Edit commit message"},"9ee329c94dc4f4991cf0df6bc43b7b360d6d35a9":{"kind":"REWORK","_number":8,"created":"2025-09-10 18:57:10.000000000","uploader":{"_account_id":32432,"name":"Owen McGonagle","email":"omcgonag@redhat.com","username":"omcgonag"},"ref":"refs/changes/04/960204/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/04/960204/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/04/960204/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/04/960204/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/04/960204/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/04/960204/8"}}},"commit":{"parents":[{"commit":"64c47ac369b35b42a17d7f6b00da81c6db46a843","subject":"Merge \"Revert \"SDK: Remove remaining neutronclient calls\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/64c47ac369b35b42a17d7f6b00da81c6db46a843"}]}],"author":{"name":"Owen McGonagle","email":"omcgonag@redhat.com","date":"2025-09-08 20:56:15.000000000","tz":-240},"committer":{"name":"Owen McGonagle","email":"omcgonag@redhat.com","date":"2025-09-10 18:57:02.000000000","tz":-240},"subject":"[WIP] Remove all dependencies/connections of old integration test code","message":"[WIP] Remove all dependencies/connections of old integration test code\n\nChange-Id: I013972aac7a6ed998bb33513024e06039232d1d4\nSigned-off-by: Owen McGonagle \u003comcgonag@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/9ee329c94dc4f4991cf0df6bc43b7b360d6d35a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/9ee329c94dc4f4991cf0df6bc43b7b360d6d35a9"}]},"branch":"refs/heads/master"},"82401b61d354845fd3017f55323c30e94047737e":{"kind":"REWORK","_number":9,"created":"2025-09-10 20:49:55.000000000","uploader":{"_account_id":32432,"name":"Owen McGonagle","email":"omcgonag@redhat.com","username":"omcgonag"},"ref":"refs/changes/04/960204/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/04/960204/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/04/960204/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/04/960204/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/04/960204/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/04/960204/9"}}},"commit":{"parents":[{"commit":"64c47ac369b35b42a17d7f6b00da81c6db46a843","subject":"Merge \"Revert \"SDK: Remove remaining neutronclient calls\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/64c47ac369b35b42a17d7f6b00da81c6db46a843"}]}],"author":{"name":"Owen McGonagle","email":"omcgonag@redhat.com","date":"2025-09-08 20:56:15.000000000","tz":-240},"committer":{"name":"Owen McGonagle","email":"omcgonag@redhat.com","date":"2025-09-10 20:49:48.000000000","tz":-240},"subject":"[WIP] Remove all dependencies/connections of old integration test code","message":"[WIP] Remove all dependencies/connections of old integration test code\n\nChange-Id: I013972aac7a6ed998bb33513024e06039232d1d4\nSigned-off-by: Owen McGonagle \u003comcgonag@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/82401b61d354845fd3017f55323c30e94047737e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/82401b61d354845fd3017f55323c30e94047737e"}]},"branch":"refs/heads/master"},"1ff5de4e8a779ca5a104c4350b8f96d0310fec13":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2025-10-09 20:53:02.000000000","uploader":{"_account_id":32432,"name":"Owen McGonagle","email":"omcgonag@redhat.com","username":"omcgonag"},"ref":"refs/changes/04/960204/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/04/960204/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/04/960204/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/04/960204/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/04/960204/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/04/960204/10"}}},"commit":{"parents":[{"commit":"4b81bd78fc22b846c82b18fc70bec5b49163f514","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/4b81bd78fc22b846c82b18fc70bec5b49163f514"}]}],"author":{"name":"Owen McGonagle","email":"omcgonag@redhat.com","date":"2025-09-08 20:56:15.000000000","tz":-240},"committer":{"name":"Owen McGonagle","email":"omcgonag@redhat.com","date":"2025-10-09 20:53:02.000000000","tz":0},"subject":"[WIP] Remove all dependencies/connections of old integration test code","message":"[WIP] Remove all dependencies/connections of old integration test code\n\nChange-Id: I013972aac7a6ed998bb33513024e06039232d1d4\nSigned-off-by: Owen McGonagle \u003comcgonag@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/1ff5de4e8a779ca5a104c4350b8f96d0310fec13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/1ff5de4e8a779ca5a104c4350b8f96d0310fec13"}]},"branch":"refs/heads/master","description":"Rebase"},"5e02cfe75731625bad8e6b31dca041c51d3cde09":{"kind":"REWORK","_number":11,"created":"2025-10-09 21:08:57.000000000","uploader":{"_account_id":32432,"name":"Owen McGonagle","email":"omcgonag@redhat.com","username":"omcgonag"},"ref":"refs/changes/04/960204/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/04/960204/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/04/960204/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/04/960204/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/04/960204/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/04/960204/11"}}},"commit":{"parents":[{"commit":"64c47ac369b35b42a17d7f6b00da81c6db46a843","subject":"Merge \"Revert \"SDK: Remove remaining neutronclient calls\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/64c47ac369b35b42a17d7f6b00da81c6db46a843"}]}],"author":{"name":"Owen McGonagle","email":"omcgonag@redhat.com","date":"2025-09-08 20:56:15.000000000","tz":-240},"committer":{"name":"Owen McGonagle","email":"omcgonag@redhat.com","date":"2025-10-09 21:08:56.000000000","tz":-240},"subject":"Remove all dependencies/connections of old integration test code","message":"Remove all dependencies/connections of old integration test code\n\nChange-Id: I013972aac7a6ed998bb33513024e06039232d1d4\nSigned-off-by: Owen McGonagle \u003comcgonag@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/5e02cfe75731625bad8e6b31dca041c51d3cde09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/5e02cfe75731625bad8e6b31dca041c51d3cde09"}]},"branch":"refs/heads/master"},"a980072763f40b2c69576d94c53f5a32c59837ac":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2025-10-17 19:33:05.000000000","uploader":{"_account_id":32432,"name":"Owen McGonagle","email":"omcgonag@redhat.com","username":"omcgonag"},"ref":"refs/changes/04/960204/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/04/960204/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/04/960204/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/04/960204/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/04/960204/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/04/960204/12"}}},"commit":{"parents":[{"commit":"197ab193739ec1b98bf7d3e2b627d8ec48939115","subject":"Merge \"Catch callback in network service doesn\u0027t throw\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/197ab193739ec1b98bf7d3e2b627d8ec48939115"}]}],"author":{"name":"Owen McGonagle","email":"omcgonag@redhat.com","date":"2025-09-08 20:56:15.000000000","tz":-240},"committer":{"name":"Owen McGonagle","email":"omcgonag@redhat.com","date":"2025-10-17 19:33:05.000000000","tz":0},"subject":"Remove all dependencies/connections of old integration test code","message":"Remove all dependencies/connections of old integration test code\n\nChange-Id: I013972aac7a6ed998bb33513024e06039232d1d4\nSigned-off-by: Owen McGonagle \u003comcgonag@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/a980072763f40b2c69576d94c53f5a32c59837ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/a980072763f40b2c69576d94c53f5a32c59837ac"}]},"branch":"refs/heads/master","description":"Rebase"},"49e5fe185a915ab80ccf4c130225371ade323711":{"kind":"REWORK","_number":13,"created":"2025-10-17 19:40:36.000000000","uploader":{"_account_id":32432,"name":"Owen McGonagle","email":"omcgonag@redhat.com","username":"omcgonag"},"ref":"refs/changes/04/960204/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/04/960204/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/04/960204/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/04/960204/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/04/960204/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/04/960204/13"}}},"commit":{"parents":[{"commit":"64c47ac369b35b42a17d7f6b00da81c6db46a843","subject":"Merge \"Revert \"SDK: Remove remaining neutronclient calls\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/64c47ac369b35b42a17d7f6b00da81c6db46a843"}]}],"author":{"name":"Owen McGonagle","email":"omcgonag@redhat.com","date":"2025-09-08 20:56:15.000000000","tz":-240},"committer":{"name":"Owen McGonagle","email":"omcgonag@redhat.com","date":"2025-10-17 19:40:35.000000000","tz":-240},"subject":"Remove all dependencies/connections of old integration test code","message":"Remove all dependencies/connections of old integration test code\n\nChange-Id: I013972aac7a6ed998bb33513024e06039232d1d4\nSigned-off-by: Owen McGonagle \u003comcgonag@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/49e5fe185a915ab80ccf4c130225371ade323711"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/49e5fe185a915ab80ccf4c130225371ade323711"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
