)]}'
{"id":"openstack%2Fhorizon~963263","triplet_id":"openstack%2Fhorizon~stable%2F2024.2~I74a10739e3294b9089237c318a4712ef9ddf1372","project":"openstack/horizon","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"6914":{"account":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"last_update":"2025-10-23 16:15:38.000000000","reason":"\u003cGERRIT_ACCOUNT_6914\u003e replied on the change","reason_account":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-10-30 09:32:35.000000000","reason":"Change was submitted"},"38329":{"account":{"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"},"last_update":"2025-10-30 06:59:18.000000000","reason":"\u003cGERRIT_ACCOUNT_38329\u003e replied on the change","reason_account":{"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"}},"35133":{"account":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"last_update":"2025-10-30 09:32:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I74a10739e3294b9089237c318a4712ef9ddf1372","subject":"Fix TOTP view redirection","status":"MERGED","created":"2025-10-07 10:53:54.000000000","updated":"2025-10-30 09:34:17.000000000","submitted":"2025-10-30 09:32:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"963263","meta_rev_id":"0c9b6d10dc5532fd60c821ab5e0ea8b56eb082a1","_number":963263,"virtual_id_number":963263,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-30 09:32:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"all":[{"value":2,"date":"2025-10-23 16:15:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":1,"date":"2025-10-22 14:46:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-10-27 15:23:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":1,"date":"2025-10-07 11:03:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"all":[{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-10-27 15:23:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}],"reviewers":{"REVIEWER":[{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},{"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-07 10:53:54.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"},"state":"REVIEWER"},{"updated":"2025-10-07 12:17:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-22 14:46:25.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"state":"REVIEWER"},{"updated":"2025-10-22 14:46:25.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"state":"REVIEWER"},{"updated":"2025-10-27 15:23:46.000000000","updated_by":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"reviewer":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"state":"REVIEWER"}],"messages":[{"id":"608354cfae89e65a87ffc62ea3a22ed472422cb4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-07 10:53:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"abb0ef823c23432177c0ae429a141b3ee6e18dbf","author":{"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"},"date":"2025-10-07 11:03:57.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5733c1b13b15f502f328eae221e96a558a94a66a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 12:17:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2188f67b698c479c9bd28c6aefd88e19\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/821a69b66e0946eab724738fa746cf25 : SUCCESS in 3m 28s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/6af36594862a47a0b5ae73e6a7c6499f : SUCCESS in 3m 47s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/67ca04b662b7452d9c673f4d5bf38478 : SUCCESS in 3m 55s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/8165000b8e154ffaadabae5c4ec2aab7 : SUCCESS in 5m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b242347f11c94da6b89d3aae1a63718d : SUCCESS in 13m 20s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/4f4a782bd8c842bea11290f6ee2b0d7a : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b86cbe00e25749459775c94beede6905 : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/518921e716604780ab3b77e26c5561d1 : SUCCESS in 14m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/86070f391d3448b59c88817a21d143f0 : SUCCESS in 14m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92bc05f2f72f43bc99d495a00611d5e9 : SUCCESS in 5m 26s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/4c3b8b5d0a8647808d1c3c63345b8afa : SUCCESS in 7m 22s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/7b5b3f543f85416493c5619335bd8329 : SUCCESS in 1h 18m 18s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/6d1de8febecd4b4fae3bfb1068a7f9ce : SUCCESS in 12m 44s (non-voting)\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/fe1c759f091c4cef9d1147b2b10168bd : SUCCESS in 25m 05s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0c9b7a12c45944aab04fd8056391cde7 : SUCCESS in 2m 28s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/55dac08446df4458aedd55887e7482cc : SUCCESS in 26m 28s","accounts_in_message":[],"_revision_number":1},{"id":"32a004f3cdbfae6c92823af8bb6aeb67563eb734","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-22 14:46:25.000000000","message":"Patch Set 1: Code-Review+1\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"7f07e3ecfe7172f3de7d985335a2760e78be66a3","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2025-10-23 16:15:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a1463b5cfa3891ba38c28f416e5e145bc3b74cc8","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2025-10-27 15:23:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5f6e2666685301aa3338006ba215c677413ed9fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 15:24:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f86d26aa2cd946589dc24e8525d1213c358f78cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 16:49:42.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a811c39f0fa84b389f431aabc1729eea\n\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/37650081203b49089a32f9c926a9a5f6 : SUCCESS in 5m 16s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/76c0b134da2c40dab06e3c36ee3a5372 : SUCCESS in 6m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c1d65f84b5e342c1bc7e87b811fc6639 : SUCCESS in 13m 38s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/0ad0cadf70cc431cacf990a8613ec353 : SUCCESS in 14m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3daa15d50e26417ea527afa9c0c019e6 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c1f7c4a7141041ca8fd73be801fd7f9d : SUCCESS in 13m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d69048e743cc4f7eb0a552c48fb7b85c : SUCCESS in 6m 34s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/52b49f227f634689b2d26bd56a038461 : FAILURE in 19m 16s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/2ccf21f921a04e0995e91b67bd82e12b : SUCCESS in 6m 22s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/5b75c51348534fcd8578b99dfaa23af1 : RETRY_LIMIT in 16m 31s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/e3eb3a5fbe8640228535d5d8201fc051 : FAILURE in 16m 56s","accounts_in_message":[],"_revision_number":1},{"id":"c139480fc15900b4281f757a6664db23f7f40b8a","author":{"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"},"date":"2025-10-27 17:32:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6407686b91c57b7f2e2c5baff8622945271014ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 18:56:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ed7340bb7591433690b3b3e63564c594\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/af6ec1a7c1de4590af7b2bfb4ad800c2 : SUCCESS in 5m 23s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/e913079a46a94eac8a3a3a4a0768b899 : SUCCESS in 3m 49s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/e6ec2153fc424fa2a899ab0cb5a34566 : SUCCESS in 5m 01s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/41cd70f58e5e47a3a3f2872a316f9e1f : SUCCESS in 6m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7bb22ed2daee4ed9b3d2958ab17d21b6 : SUCCESS in 10m 07s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/5e64008f86e14654830e03aa2c35f04d : SUCCESS in 13m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6a1f0cd6aae4c569c63689456ee8cc6 : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ddb18a9c144449c89a36b4987c480b6a : SUCCESS in 12m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/90a53ba8e3d24d9b82f7d90e396b5b43 : SUCCESS in 14m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f8599c142984679a7181c8b74fb0f06 : SUCCESS in 8m 32s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/2acc545b8f754eafac26504c3a33a629 : SUCCESS in 6m 37s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/83092526745345f996019bb3a556fef0 : RETRY_LIMIT in 8m 05s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/512f36773a19453d8e43b2b422ed407f : RETRY_LIMIT in 8m 23s (non-voting)\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/31e5d59869c54bd1a1e04f490d210ffb : FAILURE in 18m 20s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/22d8e959c1b342099402baeb92187db9 : SUCCESS in 4m 16s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/73e98bf592294904aa8b90281fd0ade4 : FAILURE in 11m 53s","accounts_in_message":[],"_revision_number":1},{"id":"9cbe857250d75297ebf141b5f5e891c02c69f4e1","author":{"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"},"date":"2025-10-28 22:02:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0d3616b009f30b8dd0c02a4577f91c9cb6332c97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 23:07:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0d4619e83c074870b28309a133140514\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/08e1d55b647743f4b36af04aa0300036 : SUCCESS in 6m 47s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/fdf3ab89860d48eb90e467065a3efc62 : SUCCESS in 5m 14s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/c5aac2e1e5d340c4acf8d88e3c11a618 : SUCCESS in 7m 10s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/71854284bd7d422db7c52a1028018d7c : SUCCESS in 8m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/103dd5b0fc96428893f8d35aa148514f : SUCCESS in 5m 26s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/5a0da3299ff84b53b610cad6e6eb3199 : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8716e8d7aeb4a31a340fc995679b9fb : SUCCESS in 6m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9741ee46ed75461291a770fd7544cee8 : SUCCESS in 14m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/34b930e611ee4cd0b4addd06bf84b34f : SUCCESS in 21m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10a35e0172724095970b1ff8d257f116 : SUCCESS in 6m 52s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/47b1882f842548ebb6500acc0bc099c1 : SUCCESS in 4m 40s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/0ba1b2d3fc40404cae0a2eba0c8ba3e0 : RETRY_LIMIT in 18m 59s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/a1f115317c944fe886a16693f2da6da7 : RETRY_LIMIT in 7m 58s (non-voting)\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/6ecaa986a72d4a04a310c630630bec74 : FAILURE in 17m 23s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ce4c1ba1c90d4606b471641bf3e1e135 : SUCCESS in 3m 39s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/5441468276e44b50b4f7d5e1db1b0461 : FAILURE in 8m 01s","accounts_in_message":[],"_revision_number":1},{"id":"e9160906b126d363fd87b6373203523dd2df2f86","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-29 07:03:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"31a14a30cd1ea4c8ad1429b7f4d1ad7c7d6a243a","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-29 21:12:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e0a650c17b0e333bdfdaabbfa66f8de7ef642b06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-29 22:33:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f06defa3a1dd4d80b224cdb12c007374\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/0e4b2f6ae14d40fa9d143ea88de38df5 : SUCCESS in 5m 04s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/046792ba0e6b4a3ea094c696dd15e4a9 : SUCCESS in 3m 08s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/f471d3b6faaf40a7813053d54b90c0e4 : SUCCESS in 8m 13s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/11dde9bbf2df42ac9d4e1b4e7b817d79 : SUCCESS in 9m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9a1396a8b46648ce8e0614031dba0ffa : SUCCESS in 13m 47s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/7b7c3b0874904492a298a0d0f8723d93 : SUCCESS in 9m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b39dcf007e3405493b6ca392e1ebec3 : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dbb8e705ca854ea0aa57b3a223f1843a : SUCCESS in 13m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2867129b52a54b66b6238f507290599a : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1f85cb21b504592a0778297bbbaab06 : SUCCESS in 7m 45s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/e76d6a5c29c8408b83ad1dafaf4dcde9 : SUCCESS in 7m 41s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/dd82b7a886b14e54b6fea96177fa3961 : SUCCESS in 1h 16m 01s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/12a15c97c9ff486dab90efcb3cb29a64 : SUCCESS in 17m 21s (non-voting)\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/964b5d196e09440e950f1103f5f7027c : SUCCESS in 21m 53s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fa335af74f37432ba77635f259e06364 : SUCCESS in 4m 23s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/70878b04833143dbb5e4f4986f470f85 : SUCCESS in 14m 12s","accounts_in_message":[],"_revision_number":1},{"id":"c31b8effcacfe1bff386af5996e932419fc164e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-29 22:33:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bd4c923efcb6b7c7f30fd58b075ac164fe17d923","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-29 23:56:51.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2aaf088b491147449cf927eaa890ef58\n\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/63eb966e565a4ba6aee8b792164e72c8 : SUCCESS in 4m 54s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/9d7bec92348948eab9598d305bbef195 : SUCCESS in 7m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4c81b1bdb8a141adb075841c532fe2f3 : SUCCESS in 10m 04s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/0802e827407d49c79850f4222c9370a1 : SUCCESS in 13m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c242df1632ea4c8cbc05b6038c94f42f : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab35589148cf4cc88fa70681ee29a1c5 : SUCCESS in 13m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e091bd05915453faa9c59b80c92c3cd : SUCCESS in 4m 53s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/f7856eecf774435bacca724fcef901f8 : SUCCESS in 25m 20s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/e6090efc5b0f40f99cc96886702ef84b : SUCCESS in 7m 31s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/6c12c08c8b2b4d94970c46fe87946d6a : FAILURE in 1h 17m 10s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/9fe4b3874159415a86dec3c22a422b59 : SUCCESS in 19m 45s","accounts_in_message":[],"_revision_number":1},{"id":"bd45437910db255dc74cd783628c074a9120ed49","author":{"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"},"date":"2025-10-30 06:59:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c29cabd1a533d14b196dfa7554fd761f1e4ee49f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 08:15:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d310600b42114984b05e8abd3e746451\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/c567ae10d5374b67821d591c1079be16 : SUCCESS in 5m 40s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/bd2d59743ebd435b864ef9d51411034f : SUCCESS in 3m 24s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/83f9ea23f2cd41d1bb558c5d24011c95 : SUCCESS in 7m 04s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/fd14434bac2b4de984cce68fc47add6b : SUCCESS in 5m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3527113592c3421fa46513f1566d2e87 : SUCCESS in 14m 33s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/f2be8175ea824ccea0e4b1c034413c61 : SUCCESS in 13m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58bfa055998d4d968221bc9188b3de04 : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f89ead1dab8e4fd6b4b9fd81d5c42dbd : SUCCESS in 13m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/41a8486868ca42cba15fc6d4917565d8 : SUCCESS in 13m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fde8b673e414d06be4af14c3a2934b0 : SUCCESS in 7m 13s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/a624f658f95b4bd8a74f6fb5ef323ea5 : SUCCESS in 6m 50s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/3f041f02e716493db8898c54f39d957c : SUCCESS in 1h 11m 24s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/f71259115af542a7996c9e11a36277e9 : SUCCESS in 28m 29s (non-voting)\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/e4b92478662a4807b26b9d17203a5f15 : SUCCESS in 27m 40s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0cbb3a503a444483ba96a3c4cc465462 : SUCCESS in 4m 13s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/b2dffba683b541b1be25b7fe15cefeac : SUCCESS in 27m 53s","accounts_in_message":[],"_revision_number":1},{"id":"a09a1e6e6597cdec383ff559141802d9fb5064a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 08:15:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f4172282bdf2ff2bdbdde62f891695d48b57e049","author":{"_account_id":38329,"name":"Ilia Petrov","display_name":"Ilia Petrov","email":"il.petrov@maxima-int.com","username":"ilpetrov"},"date":"2025-10-30 08:22:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3bd093e0f63909980946ef08faa6dc81cc294013","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 09:32:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40ea41b0c8df4b07bc167fc476c420f0\n\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/dd90c0b4749d4fad98100a2a2748c67f : SUCCESS in 4m 07s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/4058d14fc9ac49239cdf5ee242428c48 : SUCCESS in 6m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/62601f7587ca4311a1903aa95f141956 : SUCCESS in 13m 44s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/bb17256fe6444437a5396a095745337c : SUCCESS in 13m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb3eae61520b4232bb1fd6b0a98d6edd : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/27216b12e2814ba9bbe3a05b1138cf54 : SUCCESS in 13m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4da2511f7d24fdaa519b1503dd40e92 : SUCCESS in 7m 02s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/4e3fa0e9b7f84b00a62121e146c62c25 : SUCCESS in 26m 30s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/d046dbcc8d6d41d999930dc2806ba0ca : SUCCESS in 6m 47s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/2e9963bad11644e9a963ba94b434dd77 : SUCCESS in 1h 12m 33s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/5b4f29c5710146b6b79fcd7004d066e8 : SUCCESS in 24m 51s","accounts_in_message":[],"_revision_number":1},{"id":"1999d139464d884a8951892987bea98ae35bc993","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 09:32:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0c9b6d10dc5532fd60c821ab5e0ea8b56eb082a1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 09:34:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06799babea9346e9b524a35384532adf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee6ceb3b71904cf1b574f622287fc727 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8ddb41dfb50de4f67fe084a3fb796d2ae4971c5e","revisions":{"8ddb41dfb50de4f67fe084a3fb796d2ae4971c5e":{"kind":"REWORK","_number":1,"created":"2025-10-07 10:53:54.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/63/963263/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/63/963263/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/63/963263/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/63/963263/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/63/963263/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/63/963263/1"}}},"commit":{"parents":[{"commit":"f83e6f86cbdd0bd198c376a748537d684e21a860","subject":"Merge \"SDK subnetpools: change prefixlen fields as SDK calls them\" into stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/f83e6f86cbdd0bd198c376a748537d684e21a860"}]}],"author":{"name":"Ilia Petrov","email":"il.petrov@maxima-int.com","date":"2025-08-27 12:52:34.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2025-10-07 10:52:25.000000000","tz":180},"subject":"Fix TOTP view redirection","message":"Fix TOTP view redirection\n\nPreviously, the fail url parameter was used for user redirection,\nwhich could lead to a nonexistent URL. This change ensures proper\nredirects for authenticated users and users without a receipt.\n\nChange-Id: I74a10739e3294b9089237c318a4712ef9ddf1372\nSigned-off-by: Ilia Petrov \u003cil.petrov@maxima-int.com\u003e\n(cherry picked from commit 090108723c378b163a32ec34b0deca36c95b4423)\n(cherry picked from commit fc84d9ab0253a00ce221bdd7d6901c36bff77adc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/8ddb41dfb50de4f67fe084a3fb796d2ae4971c5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/8ddb41dfb50de4f67fe084a3fb796d2ae4971c5e"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
