)]}'
{"id":"openstack%2Fhorizon~970790","triplet_id":"openstack%2Fhorizon~master~I58a4c281716b9640cdcf0d29ddc0d82d7f19efcc","project":"openstack/horizon","branch":"master","hashtags":[],"change_id":"I58a4c281716b9640cdcf0d29ddc0d82d7f19efcc","subject":"[setup-selenium-tests] Add geckodriver_version var","status":"NEW","created":"2025-12-12 07:04:48.000000000","updated":"2025-12-12 07:59:15.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a44210bfd870cec1483f1c692a315c1b54d9e158","_number":970790,"virtual_id_number":970790,"owner":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-12-12 07:59:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-12 07:59:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f0b96026696044f262dec13584d36c6b4d6279d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2025-12-12 07:04:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a44210bfd870cec1483f1c692a315c1b54d9e158","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 07:59:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d2c0289cd3a4422b76b0ee8c2e1d5e1\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/b12c739e40f64f0aba5246afd48c9c70 : SUCCESS in 5m 42s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/b1a174a1d7754d99b14e488ace35b482 : SUCCESS in 6m 08s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/61a3aabe15b54a4b8ea17beed9928ca1 : SUCCESS in 4m 12s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/6eff7dcbb938486c8d98fb60ceb75780 : SUCCESS in 3m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1ca1abc650514612b0825fed200c2e19 : SUCCESS in 17m 44s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/430cc400ad8842cfbf64f875a7cde57e : SUCCESS in 15m 43s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/368866a05e3841a5a59b7d672ee6a02e : FAILURE in 8m 31s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3856b6a6fa54c7d832340730efd38c4 : SUCCESS in 5m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d9a09f5b7e3a4cb3ad120b473fdcd1b5 : SUCCESS in 8m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5389e11a0c4e42bdaa1876692e0db895 : SUCCESS in 9m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c8f1fa1185d44e7956aed91a44b9585 : SUCCESS in 11m 40s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/b94819c196ef4e6fb11be42640d4f402 : SUCCESS in 4m 58s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/03f38a4cb51a490682de104a0285e879 : SUCCESS in 53m 14s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/21314003b52446f8849393df9640c40f : SUCCESS in 29m 08s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/e826a6f373414f868f4d0c07cfe9bd95 : SUCCESS in 12m 12s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9efb0467ce9a435882ac478e265073a4 : SUCCESS in 4m 02s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/c20031f1a5314852bcb9da8ac551f51f : SUCCESS in 27m 32s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2853b586079ecba9550bd894a34c0333489d5a1b","revisions":{"2853b586079ecba9550bd894a34c0333489d5a1b":{"kind":"REWORK","_number":1,"created":"2025-12-12 07:04:48.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/90/970790/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/90/970790/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/90/970790/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/90/970790/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/90/970790/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/90/970790/1"}}},"commit":{"parents":[{"commit":"29e5ad799b510dba0f3359925cd7a25030f2daf8","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/29e5ad799b510dba0f3359925cd7a25030f2daf8"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2025-12-12 07:03:38.000000000","tz":0},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2025-12-12 07:04:46.000000000","tz":0},"subject":"[setup-selenium-tests] Add geckodriver_version var","message":"[setup-selenium-tests] Add geckodriver_version var\n\nIt will help to override the geckodriver_version with different\nfirefox version for selenium testing.\n\nChange-Id: I58a4c281716b9640cdcf0d29ddc0d82d7f19efcc\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/2853b586079ecba9550bd894a34c0333489d5a1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/2853b586079ecba9550bd894a34c0333489d5a1b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
