)]}'
{"id":"openstack%2Fhorizon~973849","triplet_id":"openstack%2Fhorizon~stable%2F2025.2~Id7e6a5bd807e6003eba1c3045aaada86087b8620","project":"openstack/horizon","branch":"stable/2025.2","topic":"filters-admin-volumes-snapshots-stable/2025.2","attention_set":{},"removed_from_attention_set":{"6914":{"account":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"last_update":"2026-01-20 09:35:45.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-01-20 09:35:45.000000000","reason":"Change was submitted"},"35133":{"account":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"last_update":"2026-01-20 08:08:23.000000000","reason":"\u003cGERRIT_ACCOUNT_35133\u003e replied on the change","reason_account":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}}},"hashtags":[],"change_id":"Id7e6a5bd807e6003eba1c3045aaada86087b8620","subject":"Fix filters for admin snapshots table","status":"MERGED","created":"2026-01-19 15:49:59.000000000","updated":"2026-01-20 09:37:27.000000000","submitted":"2026-01-20 09:35:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973849-filters-admin-volumes-snapshots-stable/2025.2","cherry_pick_of_change":972229,"cherry_pick_of_patch_set":1,"meta_rev_id":"a2f9d83acbe75e61a0d363b0e5c668ab47498b01","_number":973849,"virtual_id_number":973849,"owner":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-20 09:35:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-01-20 08:08:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},{"value":1,"date":"2026-01-19 20:12:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-20 08:08:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-19 16:47:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-19 20:12:08.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-01-19 20:12:08.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"state":"REVIEWER"},{"updated":"2026-01-19 20:12:08.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"fc55cd8fd9d282eb050ed27a9342ea9613a050b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2026-01-19 15:49:59.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d16dab276b7c76fd28fe413d5e7e8eefec2c4a82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 16:47:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/144e3ce9a9cc43a6a7edf6c4f9a21894\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/067011a3e8f047f9a2174812905a5fa3 : SUCCESS in 3m 59s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/f66584bf5b3047c98231d5ea5c01bbfb : SUCCESS in 2m 28s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/9f98c6742b9441e6b8c3bc24c4c17b83 : SUCCESS in 3m 05s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/e53b6a0225c64ed4b3a1a47447fa8bd9 : SUCCESS in 3m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/77e8f31c4e17436c81f33f66d0f62ed1 : SUCCESS in 9m 15s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/4be774fd5be043a2ae0a24345f771e1c : SUCCESS in 9m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50316a790f6e49c3bc06afb62e7501a9 : SUCCESS in 2m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e2a4312e4e4a4bd79fc1e7d5a001bb0b : SUCCESS in 9m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4a180541e20d4cdcb59f83f05f4901a6 : SUCCESS in 10m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/722c2202f0b6440d81602442ea3fbd38 : SUCCESS in 13m 56s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/a1017435b104411f8eca344a2c6835f0 : SUCCESS in 4m 32s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/6bcf3d3126554b4ca7d779af17c86afc : SUCCESS in 56m 05s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/7c98716dfbc04f5794d7b2d9806a0df7 : SUCCESS in 29m 56s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/394cd7b5493e4eb392f35441c6b21a7b : SUCCESS in 30m 30s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b452779d6b7e4228a791b82050a03555 : SUCCESS in 2m 53s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/8e46440eb2de4ac7b7fef66ded5ba353 : SUCCESS in 33m 31s","accounts_in_message":[],"_revision_number":1},{"id":"ef0d3000a5d360323757c310297ef54114ce4b7d","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-19 20:12:08.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2bb550cf455f08de56d0bced0cf2b730c128c9fa","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2026-01-20 08:08:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"36a71cc6d5bb5709195175652ed0206f6b7a883b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 08:08:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"facd9ded241930832576401ff0c0eb917b21c67f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 09:35:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e1cac2bb883445ca1deadec7bace949\n\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/0521b143db9e4fbbb4389ac7eff49904 : SUCCESS in 4m 23s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/17bb8701331843c0b8ebf23d5e2d07e5 : SUCCESS in 7m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f3180f2e9e6444078f71850c0e7bdee1 : SUCCESS in 17m 53s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/92e35b106acd4bc4999f45d1a7fa99de : SUCCESS in 14m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/030b73b6aa834c20af0a63ae5603aed5 : SUCCESS in 4m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/65bdf921662e44378f4b09bab4e3b783 : SUCCESS in 16m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5aaa8cf2408549858cc43487fa0e5003 : SUCCESS in 11m 57s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/aee10ec1d19d4ccf88868011a4e52d1a : SUCCESS in 30m 49s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/e24a6ac5de0443639dcbc699f369e045 : SUCCESS in 6m 18s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/9391cb9e8bc64f3bb51944d881b2b236 : SUCCESS in 1h 22m 47s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/c0b2d8709e98420991d2b612b90d8283 : SUCCESS in 29m 37s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/5b271754ce0e4de4a4fc45e384fd5b49 : SUCCESS in 18m 46s","accounts_in_message":[],"_revision_number":1},{"id":"07d8c3257afd8a0af784308232554ac137bb0c70","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 09:35:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a2f9d83acbe75e61a0d363b0e5c668ab47498b01","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 09:37:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23f84109e6c149879e42a0ca18ea2db3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd6a6c0a90e145a0a94606625da05318 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"75e3834bb1eada60f71f33ba12e90bb824b2e5ae","revisions":{"75e3834bb1eada60f71f33ba12e90bb824b2e5ae":{"kind":"REWORK","_number":1,"created":"2026-01-19 15:49:59.000000000","uploader":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"ref":"refs/changes/49/973849/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/49/973849/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/49/973849/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/49/973849/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/49/973849/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/49/973849/1"}}},"commit":{"parents":[{"commit":"2aa44ce6c6e41b7a4a9220d36bce0f987ce57d17","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/2aa44ce6c6e41b7a4a9220d36bce0f987ce57d17"}]}],"author":{"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","date":"2026-01-05 17:59:00.000000000","tz":-360},"committer":{"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","date":"2026-01-19 15:49:59.000000000","tz":0},"subject":"Fix filters for admin snapshots table","message":"Fix filters for admin snapshots table\n\nFollow up for https://review.opendev.org/c/openstack/horizon/+/963468\nThe server side filter was implemented for Project -\u003e Snapshots,\nthis patch adds the filter data to Admin -\u003e Snapshots view, so\nthe server side filtering works for this table as well.\n\nChange-Id: Id7e6a5bd807e6003eba1c3045aaada86087b8620\nSigned-off-by: Tatiana Ovchinnikova \u003ct.v.ovtchinnikova@gmail.com\u003e\n(cherry picked from commit 7d3ee81ae4c5c228643e3f75bf51a165791eba42)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/75e3834bb1eada60f71f33ba12e90bb824b2e5ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/75e3834bb1eada60f71f33ba12e90bb824b2e5ae"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
