)]}'
{"id":"openstack%2Fhorizon~976270","triplet_id":"openstack%2Fhorizon~master~I09a87bdb8463331c018d7d1c443910178879ed75","project":"openstack/horizon","branch":"master","attention_set":{},"removed_from_attention_set":{"6914":{"account":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"last_update":"2026-02-24 15:34:31.000000000","reason":"\u003cGERRIT_ACCOUNT_6914\u003e replied on the change","reason_account":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-02-24 17:05:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I09a87bdb8463331c018d7d1c443910178879ed75","subject":"Remove retired project from plugin registry","status":"MERGED","created":"2026-02-10 15:24:28.000000000","updated":"2026-02-24 17:07:09.000000000","submitted":"2026-02-24 17:05:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"976270","meta_rev_id":"65c5361fbda41fdcf93b5086b2371b3d05e4ea70","_number":976270,"virtual_id_number":976270,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-24 17:05:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"all":[{"value":2,"date":"2026-02-24 15:34:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":2,"date":"2026-02-23 13:44:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"all":[{"value":1,"date":"2026-02-24 15:34:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-10 16:26:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-22 16:06:17.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"state":"REVIEWER"},{"updated":"2026-02-23 13:44:59.000000000","updated_by":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"reviewer":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"state":"REVIEWER"}],"messages":[{"id":"75dd1248d20cad2e7b77d77062f5a3ae3a2aced5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-10 15:24:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4eceecc7da00f2a3c5f9dcd9e164cb9e94c3865f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-10 15:25:18.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"2dc3e7cb43f55f0aad223b205361708154abcedf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 16:26:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f0e4247f4c044f078243642760f8e92e\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/1f000bab0e8e488a9b018906fc2c2faf : SUCCESS in 5m 45s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/94ed0cf697164fb2a1b81d248d9cad84 : SUCCESS in 3m 10s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/77994a4bee8f4c5c9e45ae32c3d42515 : FAILURE in 3m 06s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/cc0e0f0f629a42ec93767d313ae1d45d : FAILURE in 3m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/43c2364980ef46b8a3069b5d0dc4c6df : FAILURE in 5m 03s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/49d08a50f0ff4c1fb0ceadcf9ffbbbc1 : RETRY_LIMIT in 2m 43s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/537d968f7658486f85146d19f7d8dec4 : RETRY_LIMIT in 1m 46s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a0e86bf25cf497bb2b24757bb6db6de : FAILURE in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ba0ea3968c0c4687909d84d996d47369 : SUCCESS in 17m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44cc0b8864fe4c6c8bc41f9b27073f1b : FAILURE in 3m 29s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/b4c7d32e7ae3469596e086f090c5c2d6 : FAILURE in 5m 24s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/29a79203bc6940c79a9d40a87d6c396d : RETRY_LIMIT in 22m 38s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/4f893733508d442a879ebf0d1a8bca41 : RETRY_LIMIT in 12m 27s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/d4fd2ad964ad4653afcbf7483e53e3d6 : FAILURE in 29m 10s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/34e371696687431fb2b69938fe614c02 : FAILURE in 2m 25s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/0e49ffb5b773447e975db7b13078d999 : FAILURE in 27m 22s","accounts_in_message":[],"_revision_number":2},{"id":"d3428e9d26001520c2f2c0095c5f709174b24e76","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-22 16:02:16.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"417fb77322b32847c137a963660a0ee0c45989f9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-22 16:02:43.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"f5f59efb3cbdcdae1cf7435772592b429d179046","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-22 16:05:01.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"16e6b22e710b647ff193c3ecbff93a7c19918cf7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-22 16:05:04.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"6f08d1e769ee2ad04de0a2ffa6bfd4be0b5284fb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-22 16:06:12.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"8cd66ce327ccfea416b59386b2d04bc335015c4d","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-22 16:06:17.000000000","message":"Patch Set 7:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":7},{"id":"f9e41af04cdaabb78ff3674b1bbe15c25422c3ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-22 16:44:01.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a443ae07ea114ed68da1719773b8928c\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/a2d3dcbecfb942dfafbd6ba21e3d33ce : SUCCESS in 7m 39s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/c9b9974a1574486e8ae0fcd056192241 : SUCCESS in 3m 22s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/5c719f77cefe4ab9a6ef3ff69a688425 : SUCCESS in 5m 33s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/9c088c170e0d4f1186872aafb954a9ce : SUCCESS in 3m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5c41133ba1a7443192850c0b7010b24e : SUCCESS in 8m 14s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/4ec00ba1c9344439aae582b65559fe2d : SUCCESS in 15m 35s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/6ed525415b224c0081ecbfff0139a644 : FAILURE in 8m 43s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3b938411ede45ea94285c87478bfe59 : SUCCESS in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/73908a23619b4c439bbf06e171cfcd44 : SUCCESS in 7m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c1b2e2953cf6492bb2f7806687529e2a : FAILURE in 28m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2154e33ffc7f41d59aea44f4882165d5 : SUCCESS in 8m 17s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/901dabdf1deb4adea41c6091a7dbc6c4 : SUCCESS in 5m 39s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/6b26865a0d4d46f380c929956f16300f : FAILURE in 35m 26s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/f578de1ee16d425aa27a950e3b0b82dc : SUCCESS in 15m 31s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/28c9a9ae724347929534c267d62db463 : SUCCESS in 29m 46s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0ab4ffd4a6dd4d718f7b4eab139cc9fb : SUCCESS in 3m 44s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/3dde0e7e23dc4c70ad2de38c586bb895 : SUCCESS in 32m 38s","accounts_in_message":[],"_revision_number":7},{"id":"565f52e585dd945e8e6df6893bba2b72d87d5ffb","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-22 16:51:01.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"d671f324b6c00c6b87a130568b103b73d06b7abf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-22 18:16:49.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e474f11e7de4974be27fdae1f2ea2be\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/8c9c464393d2435f9554a14d1b850720 : SUCCESS in 5m 04s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/e434d94ae3bb4b17be031d16d4be6ec0 : SUCCESS in 2m 09s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/9ac7582a43c042eeb247df5737a72e51 : SUCCESS in 3m 46s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/903bdff59a7844c5b5a9acfe93b4f82b : SUCCESS in 5m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4feddbfc07fc41519eb71f1fbc16d73b : SUCCESS in 8m 20s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/d12e9d46623c4b3fb7014514041525d7 : SUCCESS in 16m 25s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/e5b8dafcdd614b99936cba391da09127 : FAILURE in 17m 38s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3952aa7af55446ffa73fa2e09aa8878c : SUCCESS in 2m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c57cd4b2aa294b7d815457d90e953109 : SUCCESS in 16m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/09ac39c5097249f193b01aa73c7b9bf9 : FAILURE in 23m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b4587c157c34d5c920bde28e9920d8e : SUCCESS in 15m 37s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/3469a69a7a7b45d4910fbcb658d82c14 : SUCCESS in 5m 58s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/52f61eb9254b4116a1307414064d1f3c : SUCCESS in 1h 23m 43s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/062aea4f207a40f39d6b343e70b79048 : SUCCESS in 12m 57s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/3b4ec4684f1148ffbb602d95a2f4af5c : SUCCESS in 29m 04s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e9d92b955c6542f4aa7c839603bbe67e : SUCCESS in 3m 48s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/a9be641dccc4465dba1bcdaaa169c31a : SUCCESS in 30m 59s","accounts_in_message":[],"_revision_number":7},{"id":"45d0204230fb0a10e8fc5c00a5301ccd2b26bc6c","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2026-02-23 13:44:59.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"1fc764ef11b72ae73e52e6c19cbe808bb7703447","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2026-02-24 15:34:31.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"aa8a2f9f3470cf6c75d68afbb639993502d2c574","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 15:35:01.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"f58cb5beba56a6ad1368d111bc7a158b915ba0d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 17:05:37.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23b1bd2526744e9dbb98f3198e7c2cf9\n\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/b7a42f26cdb64ff8b6cf6c2fbd7cd765 : SUCCESS in 4m 22s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/a7615a0718944a9c8827fb6a41c28143 : SUCCESS in 7m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d9095f9d383e43a19139bd0f9e801c11 : SUCCESS in 15m 17s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/74f656f2a52a4d02aea11e467b1ac2ee : SUCCESS in 17m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e377578fdf74bcdb52002d49e7bf4ef : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1ef193c59b944788a40fffd0d38a19bd : SUCCESS in 16m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0843eb67d104491dabe52d922e7d81d5 : SUCCESS in 15m 28s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/1b255b5d0ddf42c896b7bfb86617f973 : SUCCESS in 13m 41s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/570ac65aff60409e844eac67def3b7b6 : SUCCESS in 7m 46s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/eeb1a083337c45f282374bcc0cbcd94d : SUCCESS in 1h 23m 45s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/b6bd3d7fbde445c3bb0c86ebbdf5c126 : SUCCESS in 16m 20s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/30af4dde385f49ed85509b6a318012cc : SUCCESS in 30m 52s","accounts_in_message":[],"_revision_number":7},{"id":"c3288d7d360f8e75ab3d537d562db0f54aa52539","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 17:05:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"65c5361fbda41fdcf93b5086b2371b3d05e4ea70","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 17:07:09.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63d87555b71b413bb11581dac39c6444\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a6350c31ee6492abe2e05a751660ce8 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"aca54d862985f4608db88ac5c0f361f45a825b28","revisions":{"2fbf3736c1bd2148083a9bc6f70d518b265aea5b":{"kind":"REWORK","_number":1,"created":"2026-02-10 15:24:28.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/70/976270/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/70/976270/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/70/976270/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/70/976270/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/70/976270/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/70/976270/1"}}},"commit":{"parents":[{"commit":"47921222008cbe9a230f64f8e3335d203a53d714","subject":"Merge \"Fix DHCP checkbox in subnet edit view\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/47921222008cbe9a230f64f8e3335d203a53d714"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-10 15:24:28.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-10 15:24:28.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I09a87bdb8463331c018d7d1c443910178879ed75\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/2fbf3736c1bd2148083a9bc6f70d518b265aea5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/2fbf3736c1bd2148083a9bc6f70d518b265aea5b"}]},"branch":"refs/heads/master"},"475e296262915a2e34e79ddcde3eaf837c76c6cd":{"kind":"REWORK","_number":2,"created":"2026-02-10 15:25:18.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/70/976270/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/70/976270/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/70/976270/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/70/976270/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/70/976270/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/70/976270/2"}}},"commit":{"parents":[{"commit":"47921222008cbe9a230f64f8e3335d203a53d714","subject":"Merge \"Fix DHCP checkbox in subnet edit view\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/47921222008cbe9a230f64f8e3335d203a53d714"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-10 15:24:28.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-10 15:25:18.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I09a87bdb8463331c018d7d1c443910178879ed75\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/475e296262915a2e34e79ddcde3eaf837c76c6cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/475e296262915a2e34e79ddcde3eaf837c76c6cd"}]},"branch":"refs/heads/master"},"d20771f44e14fc7591c0d2a280d0856565ac093f":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-02-22 16:02:16.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/70/976270/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/70/976270/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/70/976270/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/70/976270/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/70/976270/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/70/976270/3"}}},"commit":{"parents":[{"commit":"47921222008cbe9a230f64f8e3335d203a53d714","subject":"Merge \"Fix DHCP checkbox in subnet edit view\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/47921222008cbe9a230f64f8e3335d203a53d714"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-10 15:24:28.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-22 16:02:16.000000000","tz":0},"subject":"Remove retired project from plugin registry","message":"Remove retired project from plugin registry\n\nChange-Id: I09a87bdb8463331c018d7d1c443910178879ed75\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/d20771f44e14fc7591c0d2a280d0856565ac093f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/d20771f44e14fc7591c0d2a280d0856565ac093f"}]},"branch":"refs/heads/master","description":"Edit commit message"},"0c28c840a50e817a5e58425915a8937595d4d08a":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-02-22 16:02:43.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/70/976270/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/70/976270/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/70/976270/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/70/976270/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/70/976270/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/70/976270/4"}}},"commit":{"parents":[{"commit":"47921222008cbe9a230f64f8e3335d203a53d714","subject":"Merge \"Fix DHCP checkbox in subnet edit view\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/47921222008cbe9a230f64f8e3335d203a53d714"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-10 15:24:28.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-22 16:02:43.000000000","tz":0},"subject":"Remove retired project from plugin registry","message":"Remove retired project from plugin registry\n\nmonasca-ui now retired\n\nChange-Id: I09a87bdb8463331c018d7d1c443910178879ed75\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/0c28c840a50e817a5e58425915a8937595d4d08a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/0c28c840a50e817a5e58425915a8937595d4d08a"}]},"branch":"refs/heads/master","description":"Edit commit message"},"cc9be51560d6f5c266350550dac50639325eacda":{"kind":"REWORK","_number":5,"created":"2026-02-22 16:05:01.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/70/976270/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/70/976270/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/70/976270/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/70/976270/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/70/976270/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/70/976270/5"}}},"commit":{"parents":[{"commit":"47921222008cbe9a230f64f8e3335d203a53d714","subject":"Merge \"Fix DHCP checkbox in subnet edit view\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/47921222008cbe9a230f64f8e3335d203a53d714"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-10 15:24:28.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-22 16:04:59.000000000","tz":0},"subject":"Remove retired project from plugin registry","message":"Remove retired project from plugin registry\n\nmonasca-ui now retired\n\nChange-Id: I09a87bdb8463331c018d7d1c443910178879ed75\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/cc9be51560d6f5c266350550dac50639325eacda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/cc9be51560d6f5c266350550dac50639325eacda"}]},"branch":"refs/heads/master"},"d1cffe97295796bcbebf780c49646a3b37f8935e":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-02-22 16:05:04.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/70/976270/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/70/976270/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/70/976270/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/70/976270/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/70/976270/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/70/976270/6"}}},"commit":{"parents":[{"commit":"af96b50bd0047312c9362350a8be2851c3ded51f","subject":"Merge \"Horizon save URL after okta login\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/af96b50bd0047312c9362350a8be2851c3ded51f"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-10 15:24:28.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-22 16:05:04.000000000","tz":0},"subject":"Remove retired project from plugin registry","message":"Remove retired project from plugin registry\n\nmonasca-ui now retired\n\nChange-Id: I09a87bdb8463331c018d7d1c443910178879ed75\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/d1cffe97295796bcbebf780c49646a3b37f8935e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/d1cffe97295796bcbebf780c49646a3b37f8935e"}]},"branch":"refs/heads/master","description":"Rebase"},"aca54d862985f4608db88ac5c0f361f45a825b28":{"kind":"REWORK","_number":7,"created":"2026-02-22 16:06:12.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/70/976270/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/70/976270/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/70/976270/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/70/976270/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/70/976270/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/70/976270/7"}}},"commit":{"parents":[{"commit":"af96b50bd0047312c9362350a8be2851c3ded51f","subject":"Merge \"Horizon save URL after okta login\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/af96b50bd0047312c9362350a8be2851c3ded51f"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-10 15:24:28.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-22 16:06:11.000000000","tz":0},"subject":"Remove retired project from plugin registry","message":"Remove retired project from plugin registry\n\nmonasca-ui now retired\n\nChange-Id: I09a87bdb8463331c018d7d1c443910178879ed75\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/aca54d862985f4608db88ac5c0f361f45a825b28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/aca54d862985f4608db88ac5c0f361f45a825b28"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
