)]}'
{"id":"openstack%2Fhorizon~978552","triplet_id":"openstack%2Fhorizon~master~I52e7cc36d51406b5b0531ca5c2b42b609504b721","project":"openstack/horizon","branch":"master","attention_set":{},"removed_from_attention_set":{"6914":{"account":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"last_update":"2026-03-09 14:19:35.000000000","reason":"\u003cGERRIT_ACCOUNT_6914\u003e replied on the change","reason_account":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"}},"8648":{"account":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"last_update":"2026-03-05 14:24:04.000000000","reason":"\u003cGERRIT_ACCOUNT_8648\u003e replied on the change","reason_account":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"}},"35133":{"account":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"last_update":"2026-03-09 15:17:34.000000000","reason":"Change was submitted"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-03-04 12:52:53.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I52e7cc36d51406b5b0531ca5c2b42b609504b721","subject":"Remove old Horizon integration tests","status":"MERGED","created":"2026-03-02 15:26:07.000000000","updated":"2026-03-09 15:20:20.000000000","submitted":"2026-03-09 15:17:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"978552","meta_rev_id":"270e3009759c137ccfca8012a62478e989d31527","_number":978552,"virtual_id_number":978552,"owner":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-09 15:17:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"all":[{"value":2,"date":"2026-03-09 14:19:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":2,"date":"2026-03-05 14:24:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-03-04 12:52:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"all":[{"value":1,"date":"2026-03-09 14:19:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-02 17:43:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-03 14:52:16.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-03-03 14:52:16.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"state":"REVIEWER"},{"updated":"2026-03-03 14:52:16.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"state":"REVIEWER"}],"messages":[{"id":"dbc47a6122de7b82a676428782a552e766da4eb6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2026-03-02 15:26:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb4a8ff7d65b75bc865c0419798712714058d739","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2026-03-02 15:57:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"19642f00d172c390e30939da5d1e503f9b7a2f8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 17:43:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa0ba49224df4fb5a4dfa470b50ef67b\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/dceefa43f8be4e0e8a7f53e2c74a35ad : SUCCESS in 9m 47s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/ce194472885b4cdc905ae3c81cbd6a65 : SUCCESS in 3m 49s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/6444a7523d3b4a69b665be12fa729c29 : SUCCESS in 5m 02s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/4c99e201fbb8489497e02a21cae9a335 : SUCCESS in 4m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/aa6aeda7d34047fdb475ad1c53acd5a2 : SUCCESS in 16m 32s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/31f9865b317140e488ac3e32595d1cba : SUCCESS in 9m 23s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/0791e25162934a93b6defd20a07a56e3 : FAILURE in 8m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3e2a219b93b4162817a158a7b9b3c56 : SUCCESS in 6m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f047f0ef3b9c4c888fa1ab73f4987e1e : SUCCESS in 19m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/39b061f2b8d848e8a5e56162ef146037 : FAILURE in 33m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0626ff2e8ea449918f71208999b0e29a : SUCCESS in 25m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea39c9f0300445a492405718fbd5341f : SUCCESS in 15m 29s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/f3f40b6362a54d188642ae58e2fe3f16 : SUCCESS in 8m 44s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/e8affcb12af943c78f17f4f4860e018d : SUCCESS in 1h 13m 06s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/7f8d8c3dd9e94ddd979681ae7765895f : SUCCESS in 34m 31s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/3c6016787ed94060a2f7e346c535a9cb : SUCCESS in 27m 04s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b1fe74f924384f95b0be2424a7d2455c : SUCCESS in 4m 17s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/92e508fa1d0245b7b4e2e143d3f16fee : SUCCESS in 31m 44s","accounts_in_message":[],"_revision_number":2},{"id":"9f33d9b0afab4ada764d3f0bd586d0c041ddc066","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-03 14:52:16.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"6571791cd955dc60b5ca6c117f1be4fa3e1e9104","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2026-03-03 15:32:54.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"e11e30b897cf3ec594ee38bb7b2a4ef9ae124c71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 17:04:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58d1899b52bd43e48105036b2d953046\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/78c4ad38ff8c43a5b4a84772108e03f5 : SUCCESS in 7m 07s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/b41dc6440b234ca8abea780f4a157824 : SUCCESS in 4m 35s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/afc262e12b2d4df0bd5bc437ca89371f : SUCCESS in 2m 22s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/3a6d8552438846bfba7a9d3b5f1a700f : SUCCESS in 6m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4f11c6959e9c446e8986acf8a352bf79 : SUCCESS in 6m 55s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/a779e218d7794dcdbe55b58f5c8927c1 : SUCCESS in 11m 33s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/c08fba7d9563495ebe92f9592d5bb6f5 : FAILURE in 12m 08s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/580952bf7c7b4dfab2e9e4aca09b8c14 : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/55e843e7b0444917be84e31972c2989a : SUCCESS in 7m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/acba46c1676344239c9cc25a8b42d963 : FAILURE in 17m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6b21f467ab84b1f876f0a2ffb8bbaff : SUCCESS in 16m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/94ad8a9dcb154dd1acc9a929529ac674 : SUCCESS in 10m 19s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/51831d6e5b0c4ac6a81b9f10293e0a63 : SUCCESS in 9m 09s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/a411715a360e44f9b8e15f21a373f91d : SUCCESS in 1h 27m 14s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/ebad457958a84577bbef2d1ff193feae : SUCCESS in 32m 25s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/e2320fe7b796451ba6173de7f9d6a085 : SUCCESS in 12m 27s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/34a28cf09bc24e5c9491997fbdb29e42 : SUCCESS in 4m 02s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/2d7f3f3d2f064c8dbf3ab882f08830c3 : SUCCESS in 29m 57s","accounts_in_message":[],"_revision_number":3},{"id":"2ea540a6631d5739b31877b05c2a960b74776036","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2026-03-04 06:12:50.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"f137c2875757195e0399ee924e218fecd2beb285","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-04 10:16:39.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"9d703c1527e8306c4111dac5241c68fe3e42dd5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2026-03-04 11:12:55.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"896d5c39b7ccced0c6260985b7ff7dff333a4bca","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2026-03-04 11:19:04.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6bc396a68272bd83d4facbda25032c0dbb0da864","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-04 12:48:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4553a0eb66e942da85cc72714def259b\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/4938a5978f3a46e2b8f03f3266017909 : SUCCESS in 6m 26s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/851b19998f7845bd97c8fc88723a287f : SUCCESS in 4m 52s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/1620d9d2f2494e42b7348dc516728716 : SUCCESS in 3m 49s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/58c090d997bd463197f140d3c933bd2f : SUCCESS in 3m 54s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/ca5a3e69436b465dab3fdec51da23ef7 : SUCCESS in 5m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a396109ca4c1416e9da9f4695b2273c2 : SUCCESS in 14m 38s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/7f24734b36a04e10b607cf2fda3efee5 : SUCCESS in 16m 07s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/84179c5a9a2a4442b9151b73af5b53f0 : FAILURE in 16m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d450a6c1fa6c49b885ca234580983500 : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/acaf3874ad4a4073b8a7ce9a8b183d13 : SUCCESS in 17m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b57da9f52c1d4b4783ee91aab58e3bc1 : FAILURE in 33m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e9abc1949644c52a934018b7657a5af : SUCCESS in 32m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d9a816ecebe4fdbb5453875acd082c2 : SUCCESS in 17m 47s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/14accd0044704ec58687970342640809 : SUCCESS in 5m 10s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/35b9e64d55594adc990d7a0f5918d798 : SUCCESS in 1h 27m 30s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/681d0dfee01c4dbe962001f60918a75e : SUCCESS in 21m 07s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/dbf99cfb1bc74d86bf89ddd20e492d0c : SUCCESS in 22m 35s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d1750c64251b4bbdb85f723b7c4a9a9a : SUCCESS in 4m 04s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/92c192107f4746558ea5908c262db258 : SUCCESS in 31m 19s","accounts_in_message":[],"_revision_number":4},{"id":"1c6620b383065c03cb5804ebb19168913a2da524","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-04 12:52:53.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"86b92d96bb1fe2d97e6e09d2a24d07f6da0666e4","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2026-03-05 14:24:04.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"65a4f1ce1c90075a1048fa1eca6333b57ef2bc01","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2026-03-09 14:19:35.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"06890b0d52c312d01343ab411d424d2c3d4e5ff0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 14:20:03.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0ba5d9f3c3b643349091c81c910868f120ff0efa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 15:17:34.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd7c3ec55462410b908b0b526c721844\n\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/e9b1f7c961a545cdbdaf817efc45e16f : SUCCESS in 4m 55s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/902cf6cb95fe420aba25c9c32ad92592 : SUCCESS in 5m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/77b2533a4f774cab8ee0f124137a8187 : SUCCESS in 8m 28s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/6153b8af0655427184d9c5f53f820cbe : SUCCESS in 8m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92edb8dba52d45c1b64e74d1a395aab4 : SUCCESS in 5m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f20358b99b1f4c9a870451705e5fc079 : SUCCESS in 18m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba22b68f42ee41cf873a9a1848d4d203 : SUCCESS in 13m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c056578f05f0433eaf4216179c461a81 : SUCCESS in 12m 52s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/6cb2ce9c7f054d368e40577a529d9fd5 : SUCCESS in 15m 31s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/d85a31e6c61d4752b4d7dd1b133260e9 : SUCCESS in 6m 45s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/8350a4472cbc49ea86bd917ec6854caf : SUCCESS in 56m 35s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/e9fcbb5d925045bd8652ac9b901c05cf : SUCCESS in 30m 12s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/1b775bfa236b4432a462bf3d0aefb8cc : SUCCESS in 19m 47s","accounts_in_message":[],"_revision_number":4},{"id":"7e478735cb993cf1d071b2a1c3ee039accb421bf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 15:17:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"270e3009759c137ccfca8012a62478e989d31527","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 15:20:20.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/309edb3b4f824b76a18fc4d773608765\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a85b5a17562b40a1b4e36f219ef95867 : SUCCESS in 1m 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9db8aeac88c486ca3517d3240f836a4 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b5d59181242a8e0476507389afadb6f4283d34e8","revisions":{"f4aedb62d7f55cdd9bf36f2dcbb7799db84d9d38":{"kind":"REWORK","_number":1,"created":"2026-03-02 15:26:07.000000000","uploader":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"ref":"refs/changes/52/978552/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/52/978552/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/52/978552/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/52/978552/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/52/978552/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/52/978552/1"}}},"commit":{"parents":[{"commit":"29ea0122ba96d4fc258b80b54294826630e61657","subject":"Upgrade selenium to 4.41.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/29ea0122ba96d4fc258b80b54294826630e61657"}]}],"author":{"name":"Jan Jasek","email":"jjasek@redhat.com","date":"2026-03-02 15:19:54.000000000","tz":60},"committer":{"name":"Jan Jasek","email":"jjasek@redhat.com","date":"2026-03-02 15:25:54.000000000","tz":60},"subject":"Remove old Horizon integration tests","message":"Remove old Horizon integration tests\n\nOld Horizon integration tests were replaced by\nnew integration-pytest tests some time ago and\nthe old tests are no longer used. However, the\ntests support code (helpers, decorators, etc.)\nis being preserved as it is still used by other\nplugins.\n\nChange-Id: I52e7cc36d51406b5b0531ca5c2b42b609504b721\nSigned-off-by: Jan Jasek \u003cjjasek@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/f4aedb62d7f55cdd9bf36f2dcbb7799db84d9d38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/f4aedb62d7f55cdd9bf36f2dcbb7799db84d9d38"}]},"branch":"refs/heads/master"},"d367ce35de1fb498cbe7b409c1b56b4b25eb3390":{"kind":"REWORK","_number":2,"created":"2026-03-02 15:57:53.000000000","uploader":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"ref":"refs/changes/52/978552/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/52/978552/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/52/978552/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/52/978552/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/52/978552/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/52/978552/2"}}},"commit":{"parents":[{"commit":"29ea0122ba96d4fc258b80b54294826630e61657","subject":"Upgrade selenium to 4.41.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/29ea0122ba96d4fc258b80b54294826630e61657"}]}],"author":{"name":"Jan Jasek","email":"jjasek@redhat.com","date":"2026-03-02 15:19:54.000000000","tz":60},"committer":{"name":"Jan Jasek","email":"jjasek@redhat.com","date":"2026-03-02 15:57:42.000000000","tz":60},"subject":"Remove old Horizon integration tests","message":"Remove old Horizon integration tests\n\nOld Horizon integration tests were replaced by\nnew integration-pytest tests some time ago and\nthe old tests are no longer used. However, the\ntests support code (helpers, decorators, etc.)\nis being preserved as it is still used by other\nplugins.\n\nChange-Id: I52e7cc36d51406b5b0531ca5c2b42b609504b721\nSigned-off-by: Jan Jasek \u003cjjasek@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/d367ce35de1fb498cbe7b409c1b56b4b25eb3390"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/d367ce35de1fb498cbe7b409c1b56b4b25eb3390"}]},"branch":"refs/heads/master"},"e7d5f7d876196253e51a975c37eb193eb20c0b82":{"kind":"REWORK","_number":3,"created":"2026-03-03 15:32:54.000000000","uploader":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"ref":"refs/changes/52/978552/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/52/978552/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/52/978552/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/52/978552/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/52/978552/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/52/978552/3"}}},"commit":{"parents":[{"commit":"29ea0122ba96d4fc258b80b54294826630e61657","subject":"Upgrade selenium to 4.41.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/29ea0122ba96d4fc258b80b54294826630e61657"}]}],"author":{"name":"Jan Jasek","email":"jjasek@redhat.com","date":"2026-03-02 15:19:54.000000000","tz":60},"committer":{"name":"Jan Jasek","email":"jjasek@redhat.com","date":"2026-03-03 15:32:24.000000000","tz":60},"subject":"Remove old Horizon integration tests","message":"Remove old Horizon integration tests\n\nOld Horizon integration tests were replaced by\nnew integration-pytest tests some time ago and\nthe old tests are no longer used. However, the\ntests support code (helpers, decorators, etc.)\nis being preserved as it is still used by other\nplugins.\n\nChange-Id: I52e7cc36d51406b5b0531ca5c2b42b609504b721\nSigned-off-by: Jan Jasek \u003cjjasek@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/e7d5f7d876196253e51a975c37eb193eb20c0b82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/e7d5f7d876196253e51a975c37eb193eb20c0b82"}]},"branch":"refs/heads/master"},"b5d59181242a8e0476507389afadb6f4283d34e8":{"kind":"REWORK","_number":4,"created":"2026-03-04 11:12:55.000000000","uploader":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"ref":"refs/changes/52/978552/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/52/978552/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/52/978552/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/52/978552/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/52/978552/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/52/978552/4"}}},"commit":{"parents":[{"commit":"65cb9ca7f305c5e8a7fb3761f96b3fc06be515e8","subject":"Merge \"Remove references to PhantomJS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/65cb9ca7f305c5e8a7fb3761f96b3fc06be515e8"}]}],"author":{"name":"Jan Jasek","email":"jjasek@redhat.com","date":"2026-03-02 15:19:54.000000000","tz":60},"committer":{"name":"Jan Jasek","email":"jjasek@redhat.com","date":"2026-03-04 11:11:12.000000000","tz":60},"subject":"Remove old Horizon integration tests","message":"Remove old Horizon integration tests\n\nOld Horizon integration tests were replaced by\nnew integration-pytest tests some time ago and\nthe old tests are no longer used. However, the\ntests support code (helpers, decorators, etc.)\nis being preserved as it is still used by other\nplugins.\n\nChange-Id: I52e7cc36d51406b5b0531ca5c2b42b609504b721\nSigned-off-by: Jan Jasek \u003cjjasek@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/b5d59181242a8e0476507389afadb6f4283d34e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/b5d59181242a8e0476507389afadb6f4283d34e8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
