)]}'
{"id":"openstack%2Fhorizon~985776","triplet_id":"openstack%2Fhorizon~stable%2F2025.1~I83d4da47ccf145cb8cfcf21eae2ae971d1c21a41","project":"openstack/horizon","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"33024":{"account":{"_account_id":33024,"name":"Dmitry Krasovskyi","email":"d.krasovskiy@gmail.com","username":"dkrasovskyi"},"last_update":"2026-04-22 11:24:16.000000000","reason":"\u003cGERRIT_ACCOUNT_33024\u003e replied on the change","reason_account":{"_account_id":33024,"name":"Dmitry Krasovskyi","email":"d.krasovskiy@gmail.com","username":"dkrasovskyi"}},"35133":{"account":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"last_update":"2026-04-22 11:24:51.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I83d4da47ccf145cb8cfcf21eae2ae971d1c21a41","subject":"Fix compatibility with openstackSDK for PortForwardingManager","status":"ABANDONED","created":"2026-04-22 08:47:43.000000000","updated":"2026-04-22 11:24:51.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a015f75b401a924fd941c9e3bc8bb6dbcdd429dc","_number":985776,"virtual_id_number":985776,"owner":{"_account_id":33024,"name":"Dmitry Krasovskyi","email":"d.krasovskiy@gmail.com","username":"dkrasovskyi"},"actions":{},"labels":{"Verified":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-22 09:42:15.000000000","updated_by":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"reviewer":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"state":"REVIEWER"},{"updated":"2026-04-22 09:47:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0f14eb10427ac3607ae74eb9c2180523507dc697","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33024,"name":"Dmitry Krasovskyi","email":"d.krasovskiy@gmail.com","username":"dkrasovskyi"},"date":"2026-04-22 08:47:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8cd55b004bd3844a1bdc902621bd85e07125aecf","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2026-04-22 09:42:15.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5c274da185949852f470272c91e5d9f85cb60198","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 09:47:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6de02b51712043b3ac2aa95b1c1cf0db\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/d19ddda8559f4c139bc35b07ea43cfd4 : SUCCESS in 6m 32s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/36a6f1c3586646ccbb5f2f527eb6e19b : SUCCESS in 3m 44s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/5eef3877a1eb4d409e04218e74e44297 : SUCCESS in 4m 51s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/14db7c4a4da04af4808e33f14a7f5ac1 : SUCCESS in 6m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/69a39dcd469148f291e1561509628495 : FAILURE in 7m 31s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/524adb6617c442569ee12c0ad09a13f5 : FAILURE in 15m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/473facb778be4903b1a5ce8818a57bfc : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be065162ff7244d5b3341ef2aa78d6b5 : FAILURE in 13m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73e8f89f61e6410a96cb3ea8f26f1707 : SUCCESS in 11m 12s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/6bf321871d114a1e9672eeb66ca93e88 : SUCCESS in 6m 18s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/0651c6115eed4115bcb295d1f94c0826 : SUCCESS in 57m 53s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/778ce79a20454b30b6873b01706619e4 : SUCCESS in 29m 26s (non-voting)\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/2667d25296214ff0a1091efbd700f717 : SUCCESS in 30m 33s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/84db63a932d24aca9566c5e47c6de98d : SUCCESS in 4m 39s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/da75714029944c429ba50f09360be81f : SUCCESS in 29m 40s\n- horizon-dsvm-tempest-plugin-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5073f3e4bf054bf3a6d50fe1b8c704f8 : SUCCESS in 27m 36s","accounts_in_message":[],"_revision_number":1},{"id":"841393300294246a508e0105345142d715570c7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33024,"name":"Dmitry Krasovskyi","email":"d.krasovskiy@gmail.com","username":"dkrasovskyi"},"date":"2026-04-22 11:14:12.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"b141cc2b90301928522fdb848f0ec6f112fe0dd9","author":{"_account_id":33024,"name":"Dmitry Krasovskyi","email":"d.krasovskiy@gmail.com","username":"dkrasovskyi"},"date":"2026-04-22 11:24:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a015f75b401a924fd941c9e3bc8bb6dbcdd429dc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":33024,"name":"Dmitry Krasovskyi","email":"d.krasovskiy@gmail.com","username":"dkrasovskyi"},"date":"2026-04-22 11:24:51.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cecb8c7a69b4b513609b22a074c29ecdebd912c9","revisions":{"9baaabb82cee3bed4be392a4661a58836bcf3915":{"kind":"REWORK","_number":1,"created":"2026-04-22 08:47:43.000000000","uploader":{"_account_id":33024,"name":"Dmitry Krasovskyi","email":"d.krasovskiy@gmail.com","username":"dkrasovskyi"},"ref":"refs/changes/76/985776/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/76/985776/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/76/985776/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/76/985776/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/76/985776/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/76/985776/1"}}},"commit":{"parents":[{"commit":"9598d44471ffe84d9796bea8f636e62a63aed021","subject":"Don\u0027t request flavors extra specs for Launch Instance dialog","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/9598d44471ffe84d9796bea8f636e62a63aed021"}]}],"author":{"name":"Dmitry Krasovskyi","email":"d.krasovskiy@gmail.com","date":"2026-04-22 08:44:25.000000000","tz":180},"committer":{"name":"Dmitry Krasovskyi","email":"d.krasovskiy@gmail.com","date":"2026-04-22 08:47:14.000000000","tz":180},"subject":"Fix compatibility with openstackSDK for PortForwardingManager","message":"Fix compatibility with openstackSDK for PortForwardingManager\n\nCloses-Bug: #2149874\nChange-Id: I83d4da47ccf145cb8cfcf21eae2ae971d1c21a41\nSigned-off-by: Dmitry Krasovskyi \u003cd.krasovskiy@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/9baaabb82cee3bed4be392a4661a58836bcf3915"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/9baaabb82cee3bed4be392a4661a58836bcf3915"}]},"branch":"refs/heads/stable/2025.1"},"cecb8c7a69b4b513609b22a074c29ecdebd912c9":{"kind":"REWORK","_number":2,"created":"2026-04-22 11:14:12.000000000","uploader":{"_account_id":33024,"name":"Dmitry Krasovskyi","email":"d.krasovskiy@gmail.com","username":"dkrasovskyi"},"ref":"refs/changes/76/985776/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/76/985776/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/76/985776/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/76/985776/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/76/985776/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/76/985776/2"}}},"commit":{"parents":[{"commit":"9598d44471ffe84d9796bea8f636e62a63aed021","subject":"Don\u0027t request flavors extra specs for Launch Instance dialog","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/9598d44471ffe84d9796bea8f636e62a63aed021"}]}],"author":{"name":"Dmitry Krasovskyi","email":"d.krasovskiy@gmail.com","date":"2026-04-22 08:44:25.000000000","tz":180},"committer":{"name":"Dmitry Krasovskyi","email":"d.krasovskiy@gmail.com","date":"2026-04-22 07:34:22.000000000","tz":180},"subject":"Fix compatibility with openstackSDK for PortForwardingManager","message":"Fix compatibility with openstackSDK for PortForwardingManager\n\nCloses-Bug: #2149874\nChange-Id: I83d4da47ccf145cb8cfcf21eae2ae971d1c21a41\nSigned-off-by: Dmitry Krasovskyi \u003cd.krasovskiy@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/cecb8c7a69b4b513609b22a074c29ecdebd912c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/cecb8c7a69b4b513609b22a074c29ecdebd912c9"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
