)]}'
{"id":"openstack%2Fhorizon~987329","triplet_id":"openstack%2Fhorizon~master~Ia0b2358a4b46b93da80df041cc504272153b0abc","project":"openstack/horizon","branch":"master","topic":"selenium-tests","attention_set":{"35133":{"account":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"last_update":"2026-05-05 11:30:15.000000000","reason":"Vote got outdated and was removed: Code-Review-1"}},"removed_from_attention_set":{"8648":{"account":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"last_update":"2026-05-06 14:38:10.000000000","reason":"\u003cGERRIT_ACCOUNT_8648\u003e replied on the change","reason_account":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"}}},"hashtags":[],"change_id":"Ia0b2358a4b46b93da80df041cc504272153b0abc","subject":"Remove legacy selenium-headless job","status":"NEW","created":"2026-05-05 09:39:07.000000000","updated":"2026-05-06 14:38:10.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d6a515e67fb965bc83df1ca9b0a5f2b50129aba0","_number":987329,"virtual_id_number":987329,"owner":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-05-05 15:54:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-05 10:29:53.000000000","updated_by":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"reviewer":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"state":"REVIEWER"},{"updated":"2026-05-05 11:10:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"007d879ded3cea72113de60e4c36f05adf02836b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2026-05-05 09:39:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b82ff1241855d63217b2577f50d0e7b9f428009a","author":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"date":"2026-05-05 10:29:53.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bd7adfe4cf4538b965fcdb9d065aeebb1f721c51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 11:10:54.000000000","message":"Patch Set 1: Verified-1\n\n(13 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/09b3e177d8014dc9a75b284566a212db\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/c53a9feb71d246169100f52f6bdabb8d : FAILURE in 4m 25s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/f210009bafba4dd899c523f3fd351859 : SUCCESS in 4m 04s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/152ec444c95e418ab27c464c772a74bc : SUCCESS in 3m 25s (non-voting)\n- horizon-cross-magnum-ui-python https://zuul.opendev.org/t/openstack/build/8c8cb067896b48c683628bf62f60ea73 : SUCCESS in 3m 20s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/fcbfc2ff69694a108ce142d39589cd1f : SUCCESS in 4m 19s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/33816199bc8f450188c043f0725cc49c : SUCCESS in 4m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/07b9dd8d1ce843fdb2047f8ba99865dc : SUCCESS in 15m 05s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/5ecb607c4f0e46bf806b40bd72ccb88e : SUCCESS in 16m 10s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/b798b7c806494e7498c09f00667474e4 : SUCCESS in 8m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00a8bb803cf843fe836ffded77d86c09 : FAILURE in 3m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/85567fc0643c4996b3bcb5ec5c2059f2 : SUCCESS in 13m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e6b3e702fbe747a38c5b774a2d16d869 : FAILURE in 33m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da8d9093785e4cf097cfa861e02bc7a4 : SUCCESS in 14m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72128e81f1b941258d485c236969e352 : SUCCESS in 18m 20s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/e3ec14e14a30437096e703e1ad5dc22c : SUCCESS in 1h 26m 18s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/dc348d54e6b34d16844571d6a5ca98d4 : SUCCESS in 29m 34s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/cc0298b501e24fb8bcde157448d7082d : SUCCESS in 15m 13s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8bd8c8a453d74468ba51419e93cbba3e : SUCCESS in 3m 42s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/eb9e857d10c24cfb93e9d7dd17347074 : SUCCESS in 14m 56s","accounts_in_message":[],"_revision_number":1},{"id":"84814d158344a6841ee4ea7a3b25ff6be7d851e0","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2026-05-05 11:25:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f796da1fa906eb1ae109ff33fe3b0a38f2fcd44f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2026-05-05 11:30:15.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"a07cce937033e6a9fd037092219d247e4240a99b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 12:45:23.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/56f8892fcdbd40e5a9fe84f3f804eb4e\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/b9bbe439f9774388ba85edf2f31bbf67 : FAILURE in 4m 25s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/03f9e2e38b264a748bcd2737a15db9c4 : SUCCESS in 5m 27s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/e7f9ce6dfa6a4546a3e97a80797745b9 : SUCCESS in 3m 56s (non-voting)\n- horizon-cross-magnum-ui-python https://zuul.opendev.org/t/openstack/build/dde7ffc1c25a479e826f24c1dc05279f : SUCCESS in 4m 51s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/3281aa962bea48779fddf1eabef2abf8 : SUCCESS in 4m 02s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/8089fe1a4d394df2818f1f51981c0355 : SUCCESS in 7m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f63ad97c2fe34ee1a136317cecacc02e : SUCCESS in 11m 50s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/02ad979dfc1d4c83af188aa9b106aa71 : SUCCESS in 15m 27s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/9e50ae97f3b7483c9fd5aa04c4d95f15 : SUCCESS in 9m 00s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ea8728b9a624946a793b9b5aaa3d974 : FAILURE in 5m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b04d6140ab944c0780dc5128aff729e2 : SUCCESS in 16m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7d8fc0f6be75418f98d66f1092f8892c : FAILURE in 16m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf78a8931fbc4f79be7349f7f598ce18 : SUCCESS in 15m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/140ab245b7cb4232929d59a88c822e85 : SUCCESS in 19m 59s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/f026aa2debea4f7ba33dab3a5f5f92ed : SUCCESS in 1h 08m 51s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/f9c9a5ee9b6747cbae62087c942eec80 : SUCCESS in 31m 56s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/17914e79ab40499896347429fa717c83 : SUCCESS in 13m 04s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0b9ae4904ec149ab8cf6b951f3cd5ce9 : SUCCESS in 4m 11s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/2101a26c6af641718cd0a6bfbaea0822 : SUCCESS in 29m 18s","accounts_in_message":[],"_revision_number":2},{"id":"f75b3362969ae6665e047f02a99d774e675a9bd2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2026-05-05 13:40:51.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"780a998584d42fa0a69829d68ce187b8e6163b8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 15:54:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8eb98a27948c48e1bb1d5e3510d964b8\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/5cac8ce195fd45878fbe98ab25e9d6b4 : SUCCESS in 7m 35s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/102c9636a35947108829f257e0f40ac9 : SUCCESS in 5m 29s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/5fc16d931b474926b55e3abd7c722dc4 : SUCCESS in 3m 47s (non-voting)\n- horizon-cross-magnum-ui-python https://zuul.opendev.org/t/openstack/build/a13124cf2297403080daf6ffe7ecd488 : SUCCESS in 3m 46s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/82788c22826d43ea92628a2554ae2b0e : SUCCESS in 4m 32s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/7db4f3547bc849479a8cb5fde99440c1 : SUCCESS in 6m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f8dc59d2ed874309ab3d3b2db460ff45 : SUCCESS in 13m 43s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/04c90d85d72d4964be50ed074949fcbf : SUCCESS in 16m 27s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/cdeb0a9d893f4cb0a52ef5fdcf569231 : SUCCESS in 18m 40s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8164c64d23554b8180988cb3c74775d3 : SUCCESS in 6m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e96c4fbe2afa43ceb1fcdae1e90a5c97 : SUCCESS in 15m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bc7ff23f6c3c489b9bfbbf9fa2ea6961 : FAILURE in 33m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf6087f9214a44299324b8483f2b741a : SUCCESS in 17m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/224c8f92bd4e4aaabb6b87bc19deb769 : SUCCESS in 18m 16s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/960b72c887154c4eb23c217e5ea78c26 : SUCCESS in 1h 27m 24s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/bd9407758aeb4371b7925d4559781d24 : SUCCESS in 15m 29s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/13bb0b74b7a8423da673143e1b778b7c : SUCCESS in 29m 44s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4a181de4b03b4142aca9e48df8e99189 : SUCCESS in 3m 19s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/89c7154732e2482bbc77300a45231e3c : SUCCESS in 20m 26s","accounts_in_message":[],"_revision_number":3},{"id":"d6a515e67fb965bc83df1ca9b0a5f2b50129aba0","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2026-05-06 14:38:10.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6c1e638d10708b470dfe02b1dbff6a7d3bcf5130","revisions":{"6d1b6262c38fad184209aef71575966327b4cc28":{"kind":"REWORK","_number":1,"created":"2026-05-05 09:39:07.000000000","uploader":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"ref":"refs/changes/29/987329/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/29/987329/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/29/987329/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/29/987329/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/29/987329/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/29/987329/1"}}},"commit":{"parents":[{"commit":"6c701e92c7b6f44dd99d07850fdfa6a71843ab06","subject":"Merge \"SDK: Remove remaining neutronclient calls\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/6c701e92c7b6f44dd99d07850fdfa6a71843ab06"}]}],"author":{"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","date":"2026-05-05 09:36:19.000000000","tz":120},"committer":{"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","date":"2026-05-05 09:36:19.000000000","tz":120},"subject":"Remove legacy selenium-headless job","message":"Remove legacy selenium-headless job\n\nThe selenium-headless job only had 7 tests, of which one was skipped.\nI moved the users tests to ui-pytest tests instead, and deleted the\nimmages tests, because they were for an obsolete version of the images\npanel. I kept the one skipped test skipped, because it doesn\u0027t pass\n(and didn\u0027t pass originally).\n\nI also removed the corresponding zuul job and tox environments.\n\nSigned-off-by: Radomir Dopieralski \u003copenstack@dopieralski.pl\u003e\nChange-Id: Ia0b2358a4b46b93da80df041cc504272153b0abc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/6d1b6262c38fad184209aef71575966327b4cc28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/6d1b6262c38fad184209aef71575966327b4cc28"}]},"branch":"refs/heads/master"},"e136cf0cc32a8a9a7a27a3dcb219e10d4eff471b":{"kind":"REWORK","_number":2,"created":"2026-05-05 11:30:15.000000000","uploader":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"ref":"refs/changes/29/987329/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/29/987329/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/29/987329/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/29/987329/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/29/987329/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/29/987329/2"}}},"commit":{"parents":[{"commit":"6c701e92c7b6f44dd99d07850fdfa6a71843ab06","subject":"Merge \"SDK: Remove remaining neutronclient calls\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/6c701e92c7b6f44dd99d07850fdfa6a71843ab06"}]}],"author":{"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","date":"2026-05-05 09:36:19.000000000","tz":120},"committer":{"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","date":"2026-05-05 11:30:07.000000000","tz":120},"subject":"Remove legacy selenium-headless job","message":"Remove legacy selenium-headless job\n\nThe selenium-headless job only had 7 tests, of which one was skipped.\nI moved the users tests to ui-pytest tests instead, and deleted the\nimmages tests, because they were for an obsolete version of the images\npanel. I kept the one skipped test skipped, because it doesn\u0027t pass\n(and didn\u0027t pass originally).\n\nI also removed the corresponding zuul job and tox environments.\n\nSigned-off-by: Radomir Dopieralski \u003copenstack@dopieralski.pl\u003e\nChange-Id: Ia0b2358a4b46b93da80df041cc504272153b0abc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/e136cf0cc32a8a9a7a27a3dcb219e10d4eff471b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/e136cf0cc32a8a9a7a27a3dcb219e10d4eff471b"}]},"branch":"refs/heads/master"},"6c1e638d10708b470dfe02b1dbff6a7d3bcf5130":{"kind":"REWORK","_number":3,"created":"2026-05-05 13:40:51.000000000","uploader":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"ref":"refs/changes/29/987329/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/29/987329/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/29/987329/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/29/987329/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/29/987329/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/29/987329/3"}}},"commit":{"parents":[{"commit":"6c701e92c7b6f44dd99d07850fdfa6a71843ab06","subject":"Merge \"SDK: Remove remaining neutronclient calls\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/6c701e92c7b6f44dd99d07850fdfa6a71843ab06"}]}],"author":{"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","date":"2026-05-05 09:36:19.000000000","tz":120},"committer":{"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","date":"2026-05-05 13:40:42.000000000","tz":120},"subject":"Remove legacy selenium-headless job","message":"Remove legacy selenium-headless job\n\nThe selenium-headless job only had 7 tests, of which one was skipped.\nI moved the users tests to ui-pytest tests instead, and deleted the\nimmages tests, because they were for an obsolete version of the images\npanel. I kept the one skipped test skipped, because it doesn\u0027t pass\n(and didn\u0027t pass originally).\n\nI also removed the corresponding zuul job and tox environments.\n\nSigned-off-by: Radomir Dopieralski \u003copenstack@dopieralski.pl\u003e\nChange-Id: Ia0b2358a4b46b93da80df041cc504272153b0abc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/6c1e638d10708b470dfe02b1dbff6a7d3bcf5130"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/6c1e638d10708b470dfe02b1dbff6a7d3bcf5130"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
