)]}'
{"id":"openstack%2Fhorizon~991039","triplet_id":"openstack%2Fhorizon~stable%2F2025.2~Iec01e820a7ff0d4d3bed7f8929cb9c3229deb502","project":"openstack/horizon","branch":"stable/2025.2","topic":"rc-script-quotes-stable/2026.1-stable/2025.2","hashtags":["CVE-2026-55748"],"change_id":"Iec01e820a7ff0d4d3bed7f8929cb9c3229deb502","subject":"Escape $ character in shellfilter, and use it consistently","status":"MERGED","created":"2026-06-02 09:06:17.000000000","updated":"2026-06-18 20:50:44.000000000","submitted":"2026-06-02 17:05:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"991039-rc-script-quotes-stable/2026.1-stable/2025.2","cherry_pick_of_change":991038,"cherry_pick_of_patch_set":1,"meta_rev_id":"5ba141e1ad3c28f749471457c2bffed53dbbd36f","_number":991039,"virtual_id_number":991039,"owner":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-02 17:05:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"all":[{"value":2,"date":"2026-06-02 16:01:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"all":[{"value":1,"date":"2026-06-02 16:01:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-02 10:43:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2d640cb93c27d3b232f76ba2598f00da117a56e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2026-06-02 09:06:17.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2026.1.","accounts_in_message":[],"_revision_number":1},{"id":"5e3abf85bc359204ec0c819cc85191ac0ffa417c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 10:43:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c63c06e77f8436393791130f33c6f9d\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/8003200b90fa433ebebbbe0a9a970c4a : SUCCESS in 6m 24s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/0cdb8d82dd094892ac38e00d634bf4e8 : SUCCESS in 3m 46s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/6697c0feaef04f99bdc01ff773fde286 : SUCCESS in 3m 34s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/c2fd6e3083994f8cb7a55e506336a248 : SUCCESS in 11m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1f7f6b11a42f418d9fe5d97c269f90a9 : SUCCESS in 17m 58s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/9cf78bb8e97a4e969334bb4d1009bce9 : SUCCESS in 19m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57bf82a640884aaba326ad013dc10ba5 : SUCCESS in 5m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/617df9c7b67e492fa6d18a9064a9a806 : SUCCESS in 14m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c87c6f6838a74f7cbb8adaca074995b0 : SUCCESS in 14m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c0ce1d64dd241ffa42a572cab46d369 : SUCCESS in 14m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45e68122635a40e2a424faf76c24b013 : SUCCESS in 8m 25s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/22405fecc54844858ad5c77903f6e5fc : SUCCESS in 6m 20s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/7ee1fb8a58a1444bacf0515406d7c334 : SUCCESS in 1h 28m 21s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/59bb72d9120f4e5391edadf54c1492c2 : SUCCESS in 19m 49s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/7eeed87fae5e4b17ba15b9d28b430213 : SUCCESS in 28m 32s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b55b5bb93b5f4ac0850379ece70603e1 : SUCCESS in 5m 37s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/a019a5d93580411a9d5bb0e3fc0380b4 : SUCCESS in 30m 38s","accounts_in_message":[],"_revision_number":1},{"id":"6d3027abdf3ba74f2effedff2a81baedf407abe3","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2026-06-02 16:01:31.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"78e9e521745b20019cf8b7adc758529e6344fa05","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2026-06-02 16:01:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"51090b8129f5f284446fb87aaed92045c554430b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 16:02:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f5b68b10cae3d723a01c468579a4fe88a3eef9e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 17:05:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c3eb2959ece423fb7a52e554f193c57\n\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/4f6e23a2bb7d4117abea6deefbd9bbb3 : SUCCESS in 5m 57s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/697441ed61244a28a761c1e0a1091b73 : SUCCESS in 9m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/24aad125dad241d19df68ba8ec7fbf48 : SUCCESS in 16m 02s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/d80864ae7320407c81927490a2e990f8 : SUCCESS in 18m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d19979009664f42833766ae180ec04b : SUCCESS in 6m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e8783e5442f146ab829f680f1fffbb3c : SUCCESS in 20m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72564393242140d08c41907086ea6cd1 : SUCCESS in 11m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c5dadbccaa96446d9e8bc062c9253a7f : SUCCESS in 13m 02s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/09aff98788c84501b4ad95a05ecc8eaf : SUCCESS in 30m 05s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/7e9b1e19fba54ef6b773a5e92ed0b09e : SUCCESS in 7m 29s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/543c984df3574fa2adc9508357be1e76 : SUCCESS in 1h 00m 19s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/0f977a363eb14f15ae716d3bad5e51c5 : SUCCESS in 30m 47s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/b2f9697af9be460ba73905a247920951 : SUCCESS in 34m 19s","accounts_in_message":[],"_revision_number":1},{"id":"a2c042d7973b7b5b4f7f3c7361b5cad63f979437","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 17:05:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"949ade578f313cd5e482b49cd8d1bd98208d211c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 17:08:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79e40638f67e4d63936f974e29fffd9d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55e0469500034650ba25bacfa5b0b848 : SUCCESS in 1m 55s","accounts_in_message":[],"_revision_number":1},{"id":"5ba141e1ad3c28f749471457c2bffed53dbbd36f","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-06-18 20:50:44.000000000","message":"Hashtag added: CVE-2026-55748","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4a4995f46f16a5ae3fb40fae528d896150ff27f3","revisions":{"4a4995f46f16a5ae3fb40fae528d896150ff27f3":{"kind":"REWORK","_number":1,"created":"2026-06-02 09:06:17.000000000","uploader":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"ref":"refs/changes/39/991039/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/39/991039/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/39/991039/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/39/991039/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/39/991039/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/39/991039/1"}}},"commit":{"parents":[{"commit":"dbe5dbb1144a0c2ac60455072d5b2a52e65c5d6f","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/dbe5dbb1144a0c2ac60455072d5b2a52e65c5d6f"}]}],"author":{"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","date":"2026-05-12 17:18:50.000000000","tz":120},"committer":{"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","date":"2026-06-02 09:06:17.000000000","tz":0},"subject":"Escape $ character in shellfilter, and use it consistently","message":"Escape $ character in shellfilter, and use it consistently\n\nMake sure the special characters included in user-provided values\ndon\u0027t break the shell scripts generated by Horizon.\n\nFixes-bug: 2152240\n\nChange-Id: Iec01e820a7ff0d4d3bed7f8929cb9c3229deb502\nSigned-off-by: Radomir Dopieralski \u003copenstack@dopieralski.pl\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/4a4995f46f16a5ae3fb40fae528d896150ff27f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/4a4995f46f16a5ae3fb40fae528d896150ff27f3"}]},"branch":"refs/heads/stable/2025.2","conflicts":{"ours":"dbe5dbb1144a0c2ac60455072d5b2a52e65c5d6f","theirs":"4acf34f1a3326aa1c3a5b60c132b3fbf5c354092","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
