)]}'
{"id":"openstack%2Fhorizon~992502","triplet_id":"openstack%2Fhorizon~stable%2F2025.2~I8fca9b1a5e12dcf35ef58bfa79d2bfa044d618ce","project":"openstack/horizon","branch":"stable/2025.2","topic":"bug/description-limit-stable/2026.1-stable/2025.2","hashtags":[],"change_id":"I8fca9b1a5e12dcf35ef58bfa79d2bfa044d618ce","subject":"Enforce max length on user description fields to prevent DoS","status":"NEW","created":"2026-06-09 18:07:07.000000000","updated":"2026-06-09 19:47:23.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":992501,"cherry_pick_of_patch_set":1,"meta_rev_id":"f2b84833f23f2adc8b894ef0754cf2edd7616a87","_number":992502,"virtual_id_number":992502,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-06-09 19:47:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":39177,"name":"Bhavana","display_name":"Bhavana","email":"sortebhavna55@gmail.com","username":"Bhavana"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-09 18:07:07.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":39177,"name":"Bhavana","display_name":"Bhavana","email":"sortebhavna55@gmail.com","username":"Bhavana"},"state":"CC"},{"updated":"2026-06-09 19:47:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"618e2a07fcb6c2cf04b026ca1d6a7935e3422a87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-09 18:07:07.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2026.1.","accounts_in_message":[],"_revision_number":1},{"id":"f2b84833f23f2adc8b894ef0754cf2edd7616a87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 19:47:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a06ae713b6b944e99115bc91a8c1413d\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/be97f29a58d549efa3c6bec5ea090d5b : SUCCESS in 7m 19s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/da3dcad25bed44cd82bf788f9f1c1a05 : SUCCESS in 4m 27s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/8806d2155a5b46a783a11a4b839b4e52 : SUCCESS in 7m 30s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/f5f0dcdf67d54843b6334b9c71b81999 : SUCCESS in 6m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/062a560b205b4c5f98b4b64414dde738 : SUCCESS in 22m 51s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/2a4d21d04109452fb26e47c37f0b952c : SUCCESS in 20m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32fc61ca0f3b41789642baff1af07347 : SUCCESS in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f2ec50db39604057a80a2cd0e9963437 : SUCCESS in 17m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4344c0ba7b7947ecaf0ff61213b9e04a : SUCCESS in 24m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7680fbf45bf4e11a3444c531b565046 : SUCCESS in 16m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/02d94b2281ba4b61abf2a7982f5f2cc3 : SUCCESS in 13m 06s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/97ef6e3cee584acdb20ed3a98830acd5 : SUCCESS in 7m 29s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/13c8b74435674729a74239328888a046 : SUCCESS in 1h 27m 33s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/9ce4a9ebdeaa4b7a9d654577a8158da6 : SUCCESS in 31m 25s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/07c48a0ae6b249c1a909ab8956d77ea2 : SUCCESS in 32m 19s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/60e79e4ee20b4e24bd84c32684abf01f : SUCCESS in 5m 11s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/dc44c6b3eeca494f9fa6a41d4dc50416 : SUCCESS in 32m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1657664aff618106275b2586c756f62e51367334","revisions":{"1657664aff618106275b2586c756f62e51367334":{"kind":"REWORK","_number":1,"created":"2026-06-09 18:07:07.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/02/992502/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/02/992502/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/02/992502/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/02/992502/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/02/992502/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/02/992502/1"}}},"commit":{"parents":[{"commit":"4a4995f46f16a5ae3fb40fae528d896150ff27f3","subject":"Escape $ character in shellfilter, and use it consistently","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/4a4995f46f16a5ae3fb40fae528d896150ff27f3"}]}],"author":{"name":"Bhavana Sorte","email":"sortebhavna55@gmail.com","date":"2026-06-09 07:14:21.000000000","tz":330},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-09 18:07:07.000000000","tz":0},"subject":"Enforce max length on user description fields to prevent DoS","message":"Enforce max length on user description fields to prevent DoS\n\nThe user creation and update forms accept an unrestricted string length\nfor the description parameter. A malicious user could send a massive\npayload via this field, causing high resource consumption and potential\nDenial of Service (DoS) conditions on the Horizon web server.\n\nThis patch adds a safe standard max_length\u003d255 constraint to the\ndescription form fields in the identity dashboard module.\n\nCloses-Bug: #2048111\nChange-Id: I8fca9b1a5e12dcf35ef58bfa79d2bfa044d618ce\nSigned-off-by: Bhavana Sorte \u003csortebhavna55@gmail.com\u003e\n(cherry picked from commit 24a2cf75a5833c09468aa3908b3f65b6195ecc60)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/1657664aff618106275b2586c756f62e51367334"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/1657664aff618106275b2586c756f62e51367334"}]},"branch":"refs/heads/stable/2025.2","conflicts":{"base":"4acf34f1a3326aa1c3a5b60c132b3fbf5c354092","ours":"4a4995f46f16a5ae3fb40fae528d896150ff27f3","theirs":"254a6fc478c4b1b9792782cdaaf59fae721a86f5","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
