)]}'
{"id":"openstack%2Fhorizon~992503","triplet_id":"openstack%2Fhorizon~stable%2F2025.1~I8fca9b1a5e12dcf35ef58bfa79d2bfa044d618ce","project":"openstack/horizon","branch":"stable/2025.1","topic":"bug/description-limit-stable/2026.1-stable/2025.2-stable/2025.1","hashtags":[],"change_id":"I8fca9b1a5e12dcf35ef58bfa79d2bfa044d618ce","subject":"Enforce max length on user description fields to prevent DoS","status":"NEW","created":"2026-06-09 18:07:29.000000000","updated":"2026-06-09 19:37:44.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":992502,"cherry_pick_of_patch_set":1,"meta_rev_id":"00dfe81e3b6c1c780fd427ef7843f2728dd2d8a8","_number":992503,"virtual_id_number":992503,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-06-09 19:37:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":39177,"name":"Bhavana","display_name":"Bhavana","email":"sortebhavna55@gmail.com","username":"Bhavana"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-09 18:07:29.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":39177,"name":"Bhavana","display_name":"Bhavana","email":"sortebhavna55@gmail.com","username":"Bhavana"},"state":"CC"},{"updated":"2026-06-09 19:37:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"44a8e2ecd260aa037c0bd9c1be4d9a9cf988f3be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-09 18:07:29.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2025.2.","accounts_in_message":[],"_revision_number":1},{"id":"00dfe81e3b6c1c780fd427ef7843f2728dd2d8a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 19:37:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4bb46960054d426c97023f46b9137326\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/c3c1ce55e57342a09ab0ab748ea754bb : SUCCESS in 7m 36s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/e6a7af339f044afb9f2d2414220aadcf : SUCCESS in 7m 24s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/6bb48c24629c4bb5abd99ffbe9303721 : SUCCESS in 4m 37s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/87737a9a41ab4c0cbda7876f9245716c : SUCCESS in 6m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7abc6c82263049128ac7d002ea7e313f : SUCCESS in 12m 54s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/34b514592faa4eaeb1a5331297fc9c7a : SUCCESS in 15m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19e97e870ac84759a881edf0f6ab73da : SUCCESS in 6m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/16c6b86440b14f9ba048f7125c22893c : SUCCESS in 13m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ebc6d4756aa4038a23d0c120adc2b0e : SUCCESS in 10m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a02eef8f6e36421e919b0c5c8dd0eed5 : SUCCESS in 10m 38s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/0fcffc662afc4588995952e2ae4a9059 : SUCCESS in 7m 22s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/a26d7f6b9f014fd59653dac2fb48bdfa : SUCCESS in 1h 19m 09s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/9d0c9ab57e6040c69a31d887b9071e53 : SUCCESS in 34m 57s (non-voting)\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/111fa884f3144be58cd1333d2a207255 : SUCCESS in 30m 31s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/08a7402ec4604dbc8bdcd6f0bf75f5e8 : SUCCESS in 5m 08s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/750cdd9863474773b6646933236b0188 : SUCCESS in 15m 08s\n- horizon-dsvm-tempest-plugin-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/683f785bfa664d85ada4331c1e4f5eb8 : SUCCESS in 29m 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c3108ed3fa2f10ab29a1572298ac7f8fa8a18127","revisions":{"c3108ed3fa2f10ab29a1572298ac7f8fa8a18127":{"kind":"REWORK","_number":1,"created":"2026-06-09 18:07:29.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/03/992503/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/03/992503/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/03/992503/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/03/992503/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/03/992503/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/03/992503/1"}}},"commit":{"parents":[{"commit":"039850556d0516e52b94b28f95762f310d779f16","subject":"Merge \"Escape $ character in shellfilter, and use it consistently\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/039850556d0516e52b94b28f95762f310d779f16"}]}],"author":{"name":"Bhavana Sorte","email":"sortebhavna55@gmail.com","date":"2026-06-09 07:14:21.000000000","tz":330},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-09 18:07:29.000000000","tz":0},"subject":"Enforce max length on user description fields to prevent DoS","message":"Enforce max length on user description fields to prevent DoS\n\nThe user creation and update forms accept an unrestricted string length\nfor the description parameter. A malicious user could send a massive\npayload via this field, causing high resource consumption and potential\nDenial of Service (DoS) conditions on the Horizon web server.\n\nThis patch adds a safe standard max_length\u003d255 constraint to the\ndescription form fields in the identity dashboard module.\n\nCloses-Bug: #2048111\nChange-Id: I8fca9b1a5e12dcf35ef58bfa79d2bfa044d618ce\nSigned-off-by: Bhavana Sorte \u003csortebhavna55@gmail.com\u003e\n(cherry picked from commit 24a2cf75a5833c09468aa3908b3f65b6195ecc60)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/c3108ed3fa2f10ab29a1572298ac7f8fa8a18127"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/c3108ed3fa2f10ab29a1572298ac7f8fa8a18127"}]},"branch":"refs/heads/stable/2025.1","conflicts":{"base":"4a4995f46f16a5ae3fb40fae528d896150ff27f3","ours":"039850556d0516e52b94b28f95762f310d779f16","theirs":"1657664aff618106275b2586c756f62e51367334","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
