)]}'
{"id":"openstack%2Fhorizon~994151","triplet_id":"openstack%2Fhorizon~stable%2F2026.1~I95be5ea7b59aac0005deec2ba0734d146424bfa6","project":"openstack/horizon","branch":"stable/2026.1","attention_set":{"6914":{"account":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"last_update":"2026-09-17 14:25:31.000000000","reason":"Added by \u003cGERRIT_ACCOUNT_37598\u003e using the hovercard menu","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"35133":{"account":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"last_update":"2026-06-20 08:54:37.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I95be5ea7b59aac0005deec2ba0734d146424bfa6","subject":"Add translation for Locked status for Instances","status":"NEW","created":"2026-06-19 18:32:30.000000000","updated":"2026-09-17 14:25:31.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":961289,"cherry_pick_of_patch_set":10,"meta_rev_id":"0db9d6f2ec96856742a992300dbf4952f1621011","_number":994151,"virtual_id_number":994151,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-06-19 19:33:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-19 19:33:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-20 08:54:37.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"state":"REVIEWER"},{"updated":"2026-06-20 08:54:37.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":35133,"name":"Jan Jasek","email":"jjasek@redhat.com","username":"janjasek"},"state":"REVIEWER"}],"messages":[{"id":"b2944acf951363c63f0e78fe11b03ed60691bfb9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-19 18:32:30.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9cdb1258c7c150aa215f321bb7a8c7e76cd22912","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-19 19:33:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/855eaa064eeb4f1fb7992ad5754a5963\n\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/2bc14bff44554901b1ea7c6841b6db3f : SUCCESS in 8m 48s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/eb6f9b6d428347d2909fa126fae37de7 : SUCCESS in 5m 11s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/83179975777a4f5cb68a4fea121cc6e2 : SUCCESS in 3m 26s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/a886c10886ab4664972cbf6f00474545 : SUCCESS in 3m 06s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/2a285cc445334f74a835b2f7f0c5fb6a : SUCCESS in 12m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/319bd06af36348e9949659836c1661cb : SUCCESS in 12m 11s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/b29ab3d422804efcbfed2b4341e775dc : SUCCESS in 16m 38s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/9f1b39048b0e4298b9e35f6a306e8132 : SUCCESS in 12m 54s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0bd8a5741f542e798cce65257bda46f : SUCCESS in 4m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9e307031569c446fa93cda563bd3b7ff : SUCCESS in 16m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/25d474b32e884438a4dbb7f28d3efd15 : FAILURE in 24m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db96d2707cde4ab18d81e34809687f56 : SUCCESS in 15m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e164adcf89a4da1a5831853554f500b : SUCCESS in 4m 58s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/3abacba396f9450f8a794c5d1f159e46 : SUCCESS in 16m 32s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/f25fd53b603b407faa8c67a0dc783fc2 : SUCCESS in 58m 35s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/eab1d2465ea54a6c843000e324cb7779 : SUCCESS in 31m 56s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/c8469eb177984086aba2372f553744af : SUCCESS in 15m 31s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/383a6073cd9a427c997128cec7d4356b : SUCCESS in 4m 17s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/aa61ae69479a4e1db76186e0fe0fb762 : SUCCESS in 13m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2726f10cb0fe5ee7ebf45a781539d1c9e1e398df","revisions":{"2726f10cb0fe5ee7ebf45a781539d1c9e1e398df":{"kind":"REWORK","_number":1,"created":"2026-06-19 18:32:30.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/51/994151/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/51/994151/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/51/994151/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/51/994151/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/51/994151/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/51/994151/1"}}},"commit":{"parents":[{"commit":"3faa48b6e3573b83123a16810594fb0133b99617","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/3faa48b6e3573b83123a16810594fb0133b99617"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-09-15 20:11:48.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-19 18:32:30.000000000","tz":0},"subject":"Add translation for Locked status for Instances","message":"Add translation for Locked status for Instances\n\nChange-Id: I95be5ea7b59aac0005deec2ba0734d146424bfa6\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n(cherry picked from commit 41e4eb6d015187e6a33232d8bffff5aee1aba7ec)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/2726f10cb0fe5ee7ebf45a781539d1c9e1e398df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/2726f10cb0fe5ee7ebf45a781539d1c9e1e398df"}]},"branch":"refs/heads/stable/2026.1","conflicts":{"base":"8d4ede745ce969c9aad6984c6bf6b078828f9829","ours":"3faa48b6e3573b83123a16810594fb0133b99617","theirs":"41e4eb6d015187e6a33232d8bffff5aee1aba7ec","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
