)]}'
{"id":"openstack%2Fhorizon~995839","triplet_id":"openstack%2Fhorizon~master~If42897664b16955274926ff48fa634ffc35c1d10","project":"openstack/horizon","branch":"master","topic":"login-splash-text","hashtags":[],"change_id":"If42897664b16955274926ff48fa634ffc35c1d10","subject":"Add a setting for a login splash text","status":"NEW","created":"2026-07-02 15:23:09.000000000","updated":"2026-07-02 16:58:49.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"4d50c4ad7ecef3e9b409b598ebf1ded37ee65e5e","_number":995839,"virtual_id_number":995839,"owner":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-07-02 16:58:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2026-07-02 16:58:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"80dab61929bb2bc58858d104a563350a96c6d39c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"date":"2026-07-02 15:23:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4d50c4ad7ecef3e9b409b598ebf1ded37ee65e5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 16:58:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ab7838f199d47e8aa8cd14bae119513\n\n- horizon-cross-designate-dashboard-python https://zuul.opendev.org/t/openstack/build/b114e3e20ecd440eb4fcfdbde94e35b2 : SUCCESS in 4m 19s (non-voting)\n- horizon-cross-heat-dashboard-python https://zuul.opendev.org/t/openstack/build/e8847dbce3154453adc54fb9302e652c : SUCCESS in 6m 07s (non-voting)\n- horizon-cross-ironic-ui-npm https://zuul.opendev.org/t/openstack/build/bd609196c7f049199d9cb4b0e82d31e6 : SUCCESS in 5m 17s (non-voting)\n- horizon-cross-octavia-dashboard-python https://zuul.opendev.org/t/openstack/build/8d2acb24c60241628c22bfe3866cb6fe : SUCCESS in 3m 37s (non-voting)\n- horizon-cross-magnum-ui-python https://zuul.opendev.org/t/openstack/build/acc3b862111d47d28170ae5d20fc1716 : SUCCESS in 2m 43s (non-voting)\n- horizon-cross-manila-ui-python https://zuul.opendev.org/t/openstack/build/e3b5cfb1e6704f7f8b84d80801a6a511 : SUCCESS in 7m 40s (non-voting)\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/3dd51c82ae084b898c91798f1095611c : SUCCESS in 4m 16s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/7911e1cbb169462da30256190fc184ca : SUCCESS in 5m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/31c1c7c7b5ee45e1813c185700e8ddfc : SUCCESS in 14m 52s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/a2322a34050a4896a3b662af5dbfdb1e : SUCCESS in 17m 25s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/cb8645ed9d4c4d28a7f365cb406900e3 : SUCCESS in 8m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c7af52e34484b3385c15973b707dd73 : SUCCESS in 3m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f8e4a8e0acbe44978902883f8c46ed25 : SUCCESS in 19m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f788b27cce5542d49921b67a712dd2a1 : SUCCESS in 13m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98d285be18d1423183210095c3d2e890 : SUCCESS in 19m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6bfd56e85784bfba8b9db730fdd49da : SUCCESS in 11m 48s\n- horizon-selenium-headless https://zuul.opendev.org/t/openstack/build/9406dd8a811d49fe981322bba8f7df03 : SUCCESS in 6m 20s\n- horizon-integration-pytest https://zuul.opendev.org/t/openstack/build/987c63b3904644028491b758d14ca90b : SUCCESS in 1h 29m 04s\n- horizon-ui-pytest https://zuul.opendev.org/t/openstack/build/fca616bc74d54291aafaa3c07378d110 : SUCCESS in 22m 14s\n- horizon-dsvm-tempest-plugin https://zuul.opendev.org/t/openstack/build/b45d8682229f459a84a070c3ca9c5224 : SUCCESS in 29m 07s\n- horizon-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/88a03c5d3d5547c3a282feaa3c5aea31 : SUCCESS in 4m 15s\n- horizon-tempest-plugin-ipv6 https://zuul.opendev.org/t/openstack/build/6af73e0423ca4863b9fcbb9888abc224 : SUCCESS in 34m 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6cc6b79811d37da99a13a424bbe344882db1754a","revisions":{"6cc6b79811d37da99a13a424bbe344882db1754a":{"kind":"REWORK","_number":1,"created":"2026-07-02 15:23:09.000000000","uploader":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"ref":"refs/changes/39/995839/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/horizon","ref":"refs/changes/39/995839/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/horizon refs/changes/39/995839/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/horizon refs/changes/39/995839/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/horizon refs/changes/39/995839/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/horizon refs/changes/39/995839/1"}}},"commit":{"parents":[{"commit":"b7ed1aba1c5d0cd8d8a1303b096d94f720787805","subject":"Merge \"Correcting import to allow using datetime.timezone.utc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/b7ed1aba1c5d0cd8d8a1303b096d94f720787805"}]}],"author":{"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","date":"2026-07-02 15:21:39.000000000","tz":120},"committer":{"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","date":"2026-07-02 15:22:56.000000000","tz":120},"subject":"Add a setting for a login splash text","message":"Add a setting for a login splash text\n\nThere is often a legal requirement to display a certain text on the\nlogin screen. Normally we would do this by customizing the theme, but\nit\u0027s such a common requirement, that it would make sense to have a\nsetting for that.\n\nChange-Id: If42897664b16955274926ff48fa634ffc35c1d10\nSigned-off-by: Radomir Dopieralski \u003copenstack@dopieralski.pl\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/6cc6b79811d37da99a13a424bbe344882db1754a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/horizon/commit/6cc6b79811d37da99a13a424bbe344882db1754a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
