)]}'
{"id":"openstack%2Fi18n~672602","triplet_id":"openstack%2Fi18n~master~Ic0c12c29f0260873faa8c4d386fb8ad33d7e9265","project":"openstack/i18n","branch":"master","topic":"fix-translations","hashtags":[],"change_id":"Ic0c12c29f0260873faa8c4d386fb8ad33d7e9265","subject":"Add bindep.txt","status":"MERGED","created":"2019-07-24 22:13:16.000000000","updated":"2019-07-25 05:48:29.000000000","submitted":"2019-07-25 05:48:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672602-1564033709194-c3a5ce02","meta_rev_id":"43aa0dff4664c37bb5db14df127191d2e3c90bd4","_number":672602,"virtual_id_number":672602,"owner":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":2,"date":"2019-07-25 05:48:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":2,"date":"2019-07-25 05:39:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":1,"date":"2019-07-25 05:39:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-25 05:39:37.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-07-25 05:48:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2463a719f787d91cf61468424e9436c3696b66ca","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2019-07-24 22:13:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2fa9e721980840bee416e1405fcfeedd5dbeab7d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 22:23:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/02/672602/1/check/openstack-tox-docs/e79fa8e/html/ : SUCCESS in 6m 06s\n- openstack-tox-pep8 http://logs.openstack.org/02/672602/1/check/openstack-tox-pep8/060727f/ : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":1},{"id":"c65258a705ccfaa3b2cd6ebb81700fad8cea13af","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-07-25 05:39:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"84b26dbdbde0d35173e1793711b6151c17acbc02","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 05:39:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5d466f47a27665fe21588d361b3ffc61b1d4ce6e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 05:48:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/02/672602/1/gate/openstack-tox-docs/1cc6a66/html/ : SUCCESS in 4m 03s\n- openstack-tox-pep8 http://logs.openstack.org/02/672602/1/gate/openstack-tox-pep8/78a5ceb/ : SUCCESS in 2m 40s","accounts_in_message":[],"_revision_number":1},{"id":"f4a606f545b24928811e2dd9435cab5c8fd7e430","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 05:48:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cdd0f640af4a63120196cf969ccc9c55265ed223","revisions":{"cdd0f640af4a63120196cf969ccc9c55265ed223":{"kind":"REWORK","_number":1,"created":"2019-07-24 22:13:16.000000000","uploader":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"ref":"refs/changes/02/672602/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/i18n","ref":"refs/changes/02/672602/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/i18n refs/changes/02/672602/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/i18n refs/changes/02/672602/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/i18n refs/changes/02/672602/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/i18n refs/changes/02/672602/1"}}},"commit":{"parents":[{"commit":"697097398529bb659390f14f00308e7d795631ce","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/697097398529bb659390f14f00308e7d795631ce"}]}],"author":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2019-07-24 22:09:14.000000000","tz":540},"committer":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2019-07-24 22:12:07.000000000","tz":540},"subject":"Add bindep.txt","message":"Add bindep.txt\n\nFor translations, we need gettext installed. With the recent removal\nof the bindep fallback file, we need to provide our own.\n\nChange-Id: Ic0c12c29f0260873faa8c4d386fb8ad33d7e9265\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/cdd0f640af4a63120196cf969ccc9c55265ed223"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/cdd0f640af4a63120196cf969ccc9c55265ed223"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
