)]}'
{"id":"openstack%2Fi18n~673375","triplet_id":"openstack%2Fi18n~master~Iaf1780a6ae4d3e1100f8624e1c45c348166aa499","project":"openstack/i18n","branch":"master","topic":"atc-auc","hashtags":[],"change_id":"Iaf1780a6ae4d3e1100f8624e1c45c348166aa499","subject":"More explicitly documents ATCs \u0026 AUCs","status":"MERGED","created":"2019-07-29 23:49:54.000000000","updated":"2019-08-17 12:54:11.000000000","submitted":"2019-08-17 12:54:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673375-1566046451211-32243520","meta_rev_id":"d929da784a82174d88a1d392d39560a08320255d","_number":673375,"virtual_id_number":673375,"owner":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6882,"name":"Sungjin Kang","display_name":"ujuc","email":"community@ujuc.me","username":"sungjin"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":2,"date":"2019-08-17 12:54:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6882,"name":"Sungjin Kang","display_name":"ujuc","email":"community@ujuc.me","username":"sungjin"},"all":[{"value":2,"date":"2019-08-01 00:58:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6882,"name":"Sungjin Kang","display_name":"ujuc","email":"community@ujuc.me","username":"sungjin"},{"value":2,"date":"2019-08-17 12:43:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":0,"_account_id":6882,"name":"Sungjin Kang","display_name":"ujuc","email":"community@ujuc.me","username":"sungjin"},{"value":1,"date":"2019-08-17 12:43:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":6882,"name":"Sungjin Kang","display_name":"ujuc","email":"community@ujuc.me","username":"sungjin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-01 00:58:44.000000000","updated_by":{"_account_id":6882,"name":"Sungjin Kang","display_name":"ujuc","email":"community@ujuc.me","username":"sungjin"},"reviewer":{"_account_id":6882,"name":"Sungjin Kang","display_name":"ujuc","email":"community@ujuc.me","username":"sungjin"},"state":"REVIEWER"},{"updated":"2019-08-17 12:43:37.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-08-17 12:54:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f142ba0ccb40904ab31e2d38b7adb888d9a0c738","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2019-07-29 23:49:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"96dcd15f5c897fc7fa49108518e1ac8735805096","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 00:01:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/75/673375/1/check/openstack-tox-docs/005dd52/html/ : SUCCESS in 6m 21s\n- openstack-tox-pep8 http://logs.openstack.org/75/673375/1/check/openstack-tox-pep8/d6a218e/ : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":1},{"id":"d4ed4b7a969b8bae9e0e72cc92206434d8b8115d","author":{"_account_id":6882,"name":"Sungjin Kang","display_name":"ujuc","email":"community@ujuc.me","username":"sungjin"},"date":"2019-08-01 00:58:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"87c1095cb21439b60f9cecc11b53c7a4e38aa970","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-08-17 12:43:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"6cf728584022f404e9010596794563ea39f1daba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-17 12:43:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"83a5fb1cd4105ac806d59086e34dba79a6fd6005","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-17 12:54:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/204a27fb1997483382ed99ba9bcd11fe : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/187fb7c139c3427791e3db652588c393 : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":1},{"id":"9c863ff8fd9cb64b7d1a6c59f9c89080a89fcf18","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-17 12:54:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9f55c1060f3b22a040efef0696dccad941d72075","revisions":{"9f55c1060f3b22a040efef0696dccad941d72075":{"kind":"REWORK","_number":1,"created":"2019-07-29 23:49:54.000000000","uploader":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"ref":"refs/changes/75/673375/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/i18n","ref":"refs/changes/75/673375/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/i18n refs/changes/75/673375/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/i18n refs/changes/75/673375/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/i18n refs/changes/75/673375/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/i18n refs/changes/75/673375/1"}}},"commit":{"parents":[{"commit":"026b9c2042fdf9469d9e285209315b714e5de469","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/026b9c2042fdf9469d9e285209315b714e5de469"}]}],"author":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2019-07-29 23:45:12.000000000","tz":540},"committer":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2019-07-29 23:45:12.000000000","tz":540},"subject":"More explicitly documents ATCs \u0026 AUCs","message":"More explicitly documents ATCs \u0026 AUCs\n\nATCs are recognized to translators who contributed to\nOpenStack official projects. User Committee recognizes translators\nwho contributed to user and operator code and documentation\ntranslation [1].\n\n[1] http://eavesdrop.openstack.org/meetings/uc/2018/uc.2018-10-01-18.02.log.html\n\nChange-Id: Iaf1780a6ae4d3e1100f8624e1c45c348166aa499\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/9f55c1060f3b22a040efef0696dccad941d72075"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/9f55c1060f3b22a040efef0696dccad941d72075"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
