)]}'
{"id":"openstack%2Fi18n~723601","triplet_id":"openstack%2Fi18n~master~Ibf9fdcddddf616fc9386b44ef33917421804238c","project":"openstack/i18n","branch":"master","topic":"update-instructions","hashtags":[],"change_id":"Ibf9fdcddddf616fc9386b44ef33917421804238c","subject":"Update python project setup instructions","status":"MERGED","created":"2020-04-27 16:24:35.000000000","updated":"2020-05-09 12:36:51.000000000","submitted":"2020-05-09 12:35:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"723601-1589027732622-d1a6a7d1","meta_rev_id":"7e9360c7c583ebe86cbf35a0d1d6edf765cb3aaa","_number":723601,"virtual_id_number":723601,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6882,"name":"Sungjin Kang","display_name":"ujuc","email":"community@ujuc.me","username":"sungjin"},{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":17907,"name":"Robert Simai","email":"robert.simai@suse.com","username":"rsimai"},{"value":0,"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-09 12:35:32.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17566,"name":"JF Taltavull","display_name":"JFT","email":"jean-francois.taltavull@elca.ch","username":"jftalta"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":0,"_account_id":6882,"name":"Sungjin Kang","display_name":"ujuc","email":"community@ujuc.me","username":"sungjin"},{"value":2,"date":"2020-05-09 12:26:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":17907,"name":"Robert Simai","email":"robert.simai@suse.com","username":"rsimai"},{"value":0,"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},{"value":2,"date":"2020-05-01 11:10:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17566,"name":"JF Taltavull","display_name":"JFT","email":"jean-francois.taltavull@elca.ch","username":"jftalta"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":0,"date":"2020-04-30 11:40:39.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":6882,"name":"Sungjin Kang","display_name":"ujuc","email":"community@ujuc.me","username":"sungjin"},{"value":1,"date":"2020-05-09 12:26:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"date":"2020-04-30 11:40:39.000000000","_account_id":17907,"name":"Robert Simai","email":"robert.simai@suse.com","username":"rsimai"},{"value":0,"date":"2020-04-30 11:40:39.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-30 11:40:39.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":17566,"name":"JF Taltavull","display_name":"JFT","email":"jean-francois.taltavull@elca.ch","username":"jftalta"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":6882,"name":"Sungjin Kang","display_name":"ujuc","email":"community@ujuc.me","username":"sungjin"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":17566,"name":"JF Taltavull","display_name":"JFT","email":"jean-francois.taltavull@elca.ch","username":"jftalta"},{"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},{"_account_id":17907,"name":"Robert Simai","email":"robert.simai@suse.com","username":"rsimai"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-30 11:40:39.000000000","updated_by":{"_account_id":6882,"name":"Sungjin Kang","display_name":"ujuc","email":"community@ujuc.me","username":"sungjin"},"reviewer":{"_account_id":6882,"name":"Sungjin Kang","display_name":"ujuc","email":"community@ujuc.me","username":"sungjin"},"state":"REVIEWER"},{"updated":"2020-04-30 11:40:39.000000000","updated_by":{"_account_id":17566,"name":"JF Taltavull","display_name":"JFT","email":"jean-francois.taltavull@elca.ch","username":"jftalta"},"reviewer":{"_account_id":17566,"name":"JF Taltavull","display_name":"JFT","email":"jean-francois.taltavull@elca.ch","username":"jftalta"},"state":"REVIEWER"},{"updated":"2020-04-30 11:40:39.000000000","updated_by":{"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},"reviewer":{"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},"state":"REVIEWER"},{"updated":"2020-04-30 11:40:39.000000000","updated_by":{"_account_id":17907,"name":"Robert Simai","email":"robert.simai@suse.com","username":"rsimai"},"reviewer":{"_account_id":17907,"name":"Robert Simai","email":"robert.simai@suse.com","username":"rsimai"},"state":"REVIEWER"},{"updated":"2020-05-01 11:10:46.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-05-09 12:26:53.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2020-05-09 12:35:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"289c86592dab431596b1e2b75f92284cdcb4f0e5","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-27 16:24:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f475ab118e915c3390a8f31f1f955ef67d973403","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 16:47:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0df9399a5c54b1ab542a32cc4114a32 : SUCCESS in 7m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80e8ca0c1e6b48519efe7f3feb89e133 : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":1},{"id":"86d2e7b5c0f08792f364aa23e531f27847b4eead","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-05-01 11:10:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"442ddaa4f2847003e37daaa8334052e18708aa93","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2020-05-09 12:26:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"42ab3cb3ad0530fc2a53bb9a84eef47789d47bf4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-09 12:27:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"be27f3e0a2b022f52e05c691e1ea995990c78d27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-09 12:35:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/505861d238f04242b2ac9700750af406 : SUCCESS in 7m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d998fdaf62d948ab9d634dfae94c8ab0 : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":1},{"id":"bf2e8905581111276f7741e3a05541d187a802c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-09 12:35:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c7c2c85d9e4a07c38c57a16264ebbfa3de5a5de7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-09 12:36:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ca986d54c3247159e689356189f41a8 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"775319ae15ef98b84c90f6ee6abf9cf39bc3bf4c","revisions":{"775319ae15ef98b84c90f6ee6abf9cf39bc3bf4c":{"kind":"REWORK","_number":1,"created":"2020-04-27 16:24:35.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/01/723601/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/i18n","ref":"refs/changes/01/723601/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/i18n refs/changes/01/723601/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/i18n refs/changes/01/723601/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/i18n refs/changes/01/723601/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/i18n refs/changes/01/723601/1"}}},"commit":{"parents":[{"commit":"40a1f1f4f9bc28725b0df5f460d743cdb5ab79d2","subject":"Sync translation_team.yaml with Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/40a1f1f4f9bc28725b0df5f460d743cdb5ab79d2"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-27 16:22:56.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-27 16:22:56.000000000","tz":120},"subject":"Update python project setup instructions","message":"Update python project setup instructions\n\nWe don\u0027t need babel.cfg and instructions in setup.cfg anymore, see\nhttp://lists.openstack.org/pipermail/openstack-discuss/2020-April/014233.html\n\nUpdate docs for this.\n\nChange-Id: Ibf9fdcddddf616fc9386b44ef33917421804238c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/775319ae15ef98b84c90f6ee6abf9cf39bc3bf4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/775319ae15ef98b84c90f6ee6abf9cf39bc3bf4c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
