)]}'
{"id":"openstack%2Fi18n~750085","triplet_id":"openstack%2Fi18n~master~Ib7626f43ec00f1db43085ca10c705ce7febf581b","project":"openstack/i18n","branch":"master","topic":"zanata/translations","hashtags":[],"change_id":"Ib7626f43ec00f1db43085ca10c705ce7febf581b","subject":"Imported Translations from Zanata","status":"MERGED","created":"2020-09-06 06:47:02.000000000","updated":"2020-09-11 13:39:09.000000000","submitted":"2020-09-11 13:36:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"750085-1599831406600-87da3b05","meta_rev_id":"a572f98591d0d8243c5bb5ced19de2d8ec0ed09f","_number":750085,"virtual_id_number":750085,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6882,"name":"Sungjin Kang","display_name":"ujuc","email":"community@ujuc.me","username":"sungjin"},{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":17907,"name":"Robert Simai","email":"robert.simai@suse.com","username":"rsimai"},{"value":0,"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-11 13:36:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17566,"name":"JF Taltavull","display_name":"JFT","email":"jean-francois.taltavull@elca.ch","username":"jftalta"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":0,"_account_id":6882,"name":"Sungjin Kang","display_name":"ujuc","email":"community@ujuc.me","username":"sungjin"},{"value":2,"date":"2020-09-11 13:10:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":17907,"name":"Robert Simai","email":"robert.simai@suse.com","username":"rsimai"},{"value":0,"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17566,"name":"JF Taltavull","display_name":"JFT","email":"jean-francois.taltavull@elca.ch","username":"jftalta"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":0,"date":"2020-09-10 10:15:40.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":6882,"name":"Sungjin Kang","display_name":"ujuc","email":"community@ujuc.me","username":"sungjin"},{"value":1,"date":"2020-09-11 13:10:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"date":"2020-09-10 10:15:40.000000000","_account_id":17907,"name":"Robert Simai","email":"robert.simai@suse.com","username":"rsimai"},{"value":0,"date":"2020-09-10 10:15:40.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},{"value":0,"date":"2020-09-10 10:15:40.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-10 10:15:40.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":17566,"name":"JF Taltavull","display_name":"JFT","email":"jean-francois.taltavull@elca.ch","username":"jftalta"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":6882,"name":"Sungjin Kang","display_name":"ujuc","email":"community@ujuc.me","username":"sungjin"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":17566,"name":"JF Taltavull","display_name":"JFT","email":"jean-francois.taltavull@elca.ch","username":"jftalta"},{"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},{"_account_id":17907,"name":"Robert Simai","email":"robert.simai@suse.com","username":"rsimai"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-10 10:15:40.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-09-10 10:15:40.000000000","updated_by":{"_account_id":6882,"name":"Sungjin Kang","display_name":"ujuc","email":"community@ujuc.me","username":"sungjin"},"reviewer":{"_account_id":6882,"name":"Sungjin Kang","display_name":"ujuc","email":"community@ujuc.me","username":"sungjin"},"state":"REVIEWER"},{"updated":"2020-09-10 10:15:40.000000000","updated_by":{"_account_id":17566,"name":"JF Taltavull","display_name":"JFT","email":"jean-francois.taltavull@elca.ch","username":"jftalta"},"reviewer":{"_account_id":17566,"name":"JF Taltavull","display_name":"JFT","email":"jean-francois.taltavull@elca.ch","username":"jftalta"},"state":"REVIEWER"},{"updated":"2020-09-10 10:15:40.000000000","updated_by":{"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},"reviewer":{"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},"state":"REVIEWER"},{"updated":"2020-09-10 10:15:40.000000000","updated_by":{"_account_id":17907,"name":"Robert Simai","email":"robert.simai@suse.com","username":"rsimai"},"reviewer":{"_account_id":17907,"name":"Robert Simai","email":"robert.simai@suse.com","username":"rsimai"},"state":"REVIEWER"},{"updated":"2020-09-11 13:10:37.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2020-09-11 13:36:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c7b43fc44a7ab7e105eb2ec494202c5b7bc9af58","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-06 06:47:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3bd2c626e57daba341eaf0a29bf936eb25a392a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-06 06:56:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27220f58bc98404188c92e4355c56a3c : SUCCESS in 7m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b9bb90a84b8496f8fb6b231e5e93f4e : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":1},{"id":"7b786cfa97ddc14bb5cc56fe3dc9dd1b0e2ebde9","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-07 06:49:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4dfd401719704a37133da482592db2cf85a43fcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 07:07:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c31b8213645f425c8696567d5e981988 : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9082baaac7447b0a6d481841a0ae4c9 : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":2},{"id":"53b26fa498e6bef7d48727068d873e2c26290251","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2020-09-11 13:10:37.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e3de61a6d92bf3d0d652c90f3926478f3e4e425e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 13:10:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2d562a1ae1e20f0543572e4266b49455388c51bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 13:36:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ece391ee017044e1adf9fe5c2d0051c8 : SUCCESS in 24m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88eab78ea06342b2ae52e9d8101ac617 : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":2},{"id":"d179d86a27091609b4f5600e5b874a18d125d571","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 13:36:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ce322d77dc5c663c6b639dc946816e0dc746bdaf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 13:39:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7b540fcc2984d12aa35367a1503d161 : SUCCESS in 1m 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b829a532e2a03b07e5dfbefdcf42d02d069dc387","revisions":{"58da42516c834cd78803ef6de5d2081a64d44903":{"kind":"REWORK","_number":1,"created":"2020-09-06 06:47:02.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/85/750085/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/i18n","ref":"refs/changes/85/750085/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/i18n refs/changes/85/750085/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/i18n refs/changes/85/750085/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/i18n refs/changes/85/750085/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/i18n refs/changes/85/750085/1"}}},"commit":{"parents":[{"commit":"9ec33d3a93160f44b73324480708a94e1ea57617","subject":"Merge \"Stop to use the __future__ module.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/9ec33d3a93160f44b73324480708a94e1ea57617"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-06 06:46:58.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-06 06:46:58.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: Ib7626f43ec00f1db43085ca10c705ce7febf581b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/58da42516c834cd78803ef6de5d2081a64d44903"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/58da42516c834cd78803ef6de5d2081a64d44903"}]},"branch":"refs/heads/master"},"b829a532e2a03b07e5dfbefdcf42d02d069dc387":{"kind":"REWORK","_number":2,"created":"2020-09-07 06:49:10.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/85/750085/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/i18n","ref":"refs/changes/85/750085/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/i18n refs/changes/85/750085/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/i18n refs/changes/85/750085/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/i18n refs/changes/85/750085/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/i18n refs/changes/85/750085/2"}}},"commit":{"parents":[{"commit":"9ec33d3a93160f44b73324480708a94e1ea57617","subject":"Merge \"Stop to use the __future__ module.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/9ec33d3a93160f44b73324480708a94e1ea57617"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-07 06:49:07.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-07 06:49:07.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: Ib7626f43ec00f1db43085ca10c705ce7febf581b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/b829a532e2a03b07e5dfbefdcf42d02d069dc387"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/b829a532e2a03b07e5dfbefdcf42d02d069dc387"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
