)]}'
{"id":"openstack%2Fi18n~769874","triplet_id":"openstack%2Fi18n~master~Ic53539e0813be1c9801ff131520c523174ace5c0","project":"openstack/i18n","branch":"master","topic":"fix-relmgt-pip-doc","hashtags":[],"change_id":"Ic53539e0813be1c9801ff131520c523174ace5c0","subject":"Move doc dependencies to doc/requirements.txt","status":"ABANDONED","created":"2021-01-08 07:06:02.000000000","updated":"2021-01-21 09:30:23.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f7c9fb3d07efe55c96b0e3fe068befe4af450f1e","_number":769874,"virtual_id_number":769874,"owner":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-01-08 07:16:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-08 07:16:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e7c19dd0d5cc0f456f139735639cc922ebb2e7a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-01-08 07:06:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d3d31585feec0285d0bb16524f9a9d13e304a53f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 07:16:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88feb5c6282e4b489be45d8813387b05 : SUCCESS in 9m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e485990801c64b86ba8c600b2e3b5dc0 : FAILURE in 3m 13s","accounts_in_message":[],"_revision_number":1},{"id":"8178ac6e104a80e7d2d6c89b8b251359425d1ac8","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-01-21 09:30:06.000000000","message":"Patch Set 1:\n\npep8 job contains a script to check POT file generation which depends on sphinx.\ni18n repo does not provide a release deliverable and it is mainly for documentation.\nThus, doc/requirements.txt provides less value. I will abandone it.","accounts_in_message":[],"_revision_number":1},{"id":"f7c9fb3d07efe55c96b0e3fe068befe4af450f1e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-01-21 09:30:23.000000000","message":"Abandoned\n\nAbandoning per my above comment","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6896c4ab951a7dafbfca5d8af2a62ab3ee929590","revisions":{"6896c4ab951a7dafbfca5d8af2a62ab3ee929590":{"kind":"REWORK","_number":1,"created":"2021-01-08 07:06:02.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/74/769874/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/i18n","ref":"refs/changes/74/769874/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/i18n refs/changes/74/769874/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/i18n refs/changes/74/769874/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/i18n refs/changes/74/769874/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/i18n refs/changes/74/769874/1"}}},"commit":{"parents":[{"commit":"80e0fee56bdcd9d16ac560c004cc8fa1edf547c6","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/80e0fee56bdcd9d16ac560c004cc8fa1edf547c6"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2021-01-08 07:07:26.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2021-01-08 07:07:26.000000000","tz":540},"subject":"Move doc dependencies to doc/requirements.txt","message":"Move doc dependencies to doc/requirements.txt\n\nSplitting document dependencies into a separate requirements file\nis our current best practice in OpenStack.\ni18n repo has no release deliverable and does not affect the release\nprocess, but I believe it is a good idea to follow the practice.\n\nChange-Id: Ic53539e0813be1c9801ff131520c523174ace5c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/6896c4ab951a7dafbfca5d8af2a62ab3ee929590"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/6896c4ab951a7dafbfca5d8af2a62ab3ee929590"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
