)]}'
{"id":"openstack%2Fi18n~774017","triplet_id":"openstack%2Fi18n~master~I83f38fd673456e0d1179c50d9df903e16b01bd5f","project":"openstack/i18n","branch":"master","topic":"debug-job-scripts","hashtags":[],"change_id":"I83f38fd673456e0d1179c50d9df903e16b01bd5f","subject":"Document tips on debugging translation job scripts","status":"MERGED","created":"2021-02-04 00:15:08.000000000","updated":"2021-03-01 16:56:28.000000000","submitted":"2021-03-01 16:38:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"774017-debug-job-scripts","meta_rev_id":"d7526107e106b43fb95ecdbd275372d61789e6a2","_number":774017,"virtual_id_number":774017,"owner":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-01 16:37:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2021-02-27 15:06:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":1,"date":"2021-03-01 07:35:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":1,"date":"2021-03-01 16:11:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-04 00:16:38.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2021-02-04 00:37:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-12 03:42:20.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"}],"messages":[{"id":"f4c187f9b343846c72d336650233204b507bd11d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-02-04 00:15:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0032d4e1061ab775fcbba41070da5844bd399dd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 00:37:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5604b82486b04e49ac4ea2f0f8fb86f0 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28302e5fc755480da286de9b1e6d033b : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":1},{"id":"dd59507c1108be17a3069c433788f76a1ee31dbe","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-02-04 06:47:27.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"52ea890b4364b9dec41162903fb6cfe32887347c","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2021-02-20 05:29:42.000000000","message":"Patch Set 1:\n\n(4 comments)\n\nGreat documentation - adding a few minor comments (will review more later)","accounts_in_message":[],"_revision_number":1},{"id":"3f200304dcae3be41343a90b53006437321dae59","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-02-24 00:02:40.000000000","message":"Patch Set 1:\n\n(4 comments)\n\nThanks for the review.","accounts_in_message":[],"_revision_number":1},{"id":"02911351c30b8f2e9271b0927bae84d9669d01f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-02-24 00:03:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e37b2c91a908bfff43dfcd4f676540ed793b6343","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 00:14:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10161584b776479ca7d4baf2fad59e6d : SUCCESS in 8m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/349fdf1d95a349be87eb2605488677e8 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":2},{"id":"ccc49644ac2d6fd75c90bc9a298094aebbf64fc8","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2021-02-27 15:06:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"52bd49e70fffba8ed3af4697264935a1c0a96a37","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2021-02-27 19:35:33.000000000","message":"Patch Set 2:\n\n@amotoki: Since now we are on SIG, if I couldn\u0027t get proper # of reviews, can I approve? Thoughts?","accounts_in_message":[],"_revision_number":2},{"id":"aa8e051d2ccb682ad0c5eac0565075729ea0cbb0","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-03-01 07:28:22.000000000","message":"Patch Set 2:\n\n\u003e @amotoki: Since now we are on SIG, if I couldn\u0027t get proper # of reviews, can I approve? Thoughts?\n\nOn SIG review criteria, I think we can adjust the +2 rule review by review. If we have folks who can review it, let\u0027s ask them for review. Otherwise, we can merge it depending on the situation.\nRegading this, I believe +1 from Andreas can be considered as +2 as he is familiar with this area.","accounts_in_message":[],"_revision_number":2},{"id":"92ca601f5343d40a751b155c89f5330798737a92","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-03-01 07:35:43.000000000","message":"Patch Set 2: Code-Review+1\n\nThanks!","accounts_in_message":[],"_revision_number":2},{"id":"c6b4d7b3f1179078ac6b753d97a1f4dbb1f26cd5","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2021-03-01 16:11:01.000000000","message":"Patch Set 2: Workflow+1\n\nAgreed thank you!!\n\n\u003e Patch Set 2:\n\u003e \n\u003e \u003e @amotoki: Since now we are on SIG, if I couldn\u0027t get proper # of reviews, can I approve? Thoughts?\n\u003e \n\u003e On SIG review criteria, I think we can adjust the +2 rule review by review. If we have folks who can review it, let\u0027s ask them for review. Otherwise, we can merge it depending on the situation.\n\u003e Regading this, I believe +1 from Andreas can be considered as +2 as he is familiar with this area.","accounts_in_message":[],"_revision_number":2},{"id":"ae4e01841c1b843c73bba202fbe0f658c1b1a3af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 16:14:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2676bd390f3d3cd5ebbe621e84e220854afdca19","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 16:37:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40fb8419611d42818c7851f52b71f0a0 : SUCCESS in 8m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/251adfdcd0924f46b7a2a22dc5a5b4e7 : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":2},{"id":"8f8a897235061b838cd93992947f66ee0d170b8a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 16:38:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d7526107e106b43fb95ecdbd275372d61789e6a2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 16:56:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/435b479b45cf41d4b860df161eebc7be : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"78a21fa83087f39fe8deea00e872f0f2408c29a8","revisions":{"aaa5174e24e097257e51de56ae1d8cd62a93f28b":{"kind":"REWORK","_number":1,"created":"2021-02-04 00:15:08.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/17/774017/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/i18n","ref":"refs/changes/17/774017/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/i18n refs/changes/17/774017/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/i18n refs/changes/17/774017/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/i18n refs/changes/17/774017/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/i18n refs/changes/17/774017/1"}}},"commit":{"parents":[{"commit":"80e0fee56bdcd9d16ac560c004cc8fa1edf547c6","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/80e0fee56bdcd9d16ac560c004cc8fa1edf547c6"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2021-02-04 00:12:59.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2021-02-04 00:15:00.000000000","tz":540},"subject":"Document tips on debugging translation job scripts","message":"Document tips on debugging translation job scripts\n\nThe need for this information is rare but when I develop\nthe translation job scripts I always research the same thing\nso it is useful at least for me.\n\nChange-Id: I83f38fd673456e0d1179c50d9df903e16b01bd5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/aaa5174e24e097257e51de56ae1d8cd62a93f28b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/aaa5174e24e097257e51de56ae1d8cd62a93f28b"}]},"branch":"refs/heads/master"},"78a21fa83087f39fe8deea00e872f0f2408c29a8":{"kind":"REWORK","_number":2,"created":"2021-02-24 00:03:06.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/17/774017/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/i18n","ref":"refs/changes/17/774017/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/i18n refs/changes/17/774017/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/i18n refs/changes/17/774017/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/i18n refs/changes/17/774017/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/i18n refs/changes/17/774017/2"}}},"commit":{"parents":[{"commit":"80e0fee56bdcd9d16ac560c004cc8fa1edf547c6","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/80e0fee56bdcd9d16ac560c004cc8fa1edf547c6"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2021-02-04 00:12:59.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2021-02-24 00:02:27.000000000","tz":540},"subject":"Document tips on debugging translation job scripts","message":"Document tips on debugging translation job scripts\n\nThe need for this information is rare but when I develop\nthe translation job scripts I always research the same thing\nso it is useful at least for me.\n\nChange-Id: I83f38fd673456e0d1179c50d9df903e16b01bd5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/78a21fa83087f39fe8deea00e872f0f2408c29a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/78a21fa83087f39fe8deea00e872f0f2408c29a8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
