)]}'
{"id":"openstack%2Fi18n~777864","triplet_id":"openstack%2Fi18n~master~Ica09d0348863aca9822eef7d9978a36b2ff3aac6","project":"openstack/i18n","branch":"master","topic":"team-member-update","attention_set":{},"removed_from_attention_set":{"14482":{"account":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"last_update":"2023-09-07 04:25:19.000000000","reason":"Change was submitted"},"841":{"account":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"last_update":"2023-09-07 02:26:06.000000000","reason":"\u003cGERRIT_ACCOUNT_841\u003e replied on the change","reason_account":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"}}},"hashtags":[],"change_id":"Ica09d0348863aca9822eef7d9978a36b2ff3aac6","subject":"Sync translation_team.yaml with Zanata","status":"MERGED","created":"2021-02-27 15:07:11.000000000","updated":"2023-09-07 04:26:59.000000000","submitted":"2023-09-07 04:25:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"777864-team-member-update","meta_rev_id":"ba1595d1fff63a090e39ab263217190c30c1f3a4","_number":777864,"virtual_id_number":777864,"owner":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-07 04:25:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":2,"date":"2023-09-07 02:26:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-07 04:15:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"all":[{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-07 04:15:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-27 15:18:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-22 04:55:46.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2023-09-07 04:15:35.000000000","updated_by":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"reviewer":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"state":"REVIEWER"}],"messages":[{"id":"c70db8ad00adf04291efe309f15055fb84af8f82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2021-02-27 15:07:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d691b65b283462c825224cf782742ea14bacb8dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-27 15:18:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19fb7251285246239a55356d8b6edf3f : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/915bbf495f594d0abc6101563936d64f : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":1},{"id":"ca88fc22ac6b59681938dac6687b4b847e72be79","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-07-22 04:55:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"919759026e3c5cb98d44d5e59f29af72f93a9a6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 04:56:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9366db0c4522ed4093cc00495a29232644fb3bd0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 05:17:41.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a353f59d945432ab6830d729c3dca1f : SUCCESS in 8m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74c14a8ac40e4be5b8b6e85e0dedbd8f : FAILURE in 4m 12s","accounts_in_message":[],"_revision_number":1},{"id":"1cf1a9c5066ce140c233dc2832f77f862dff2289","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-09-02 02:53:26.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"d91a295899b6f81258909de8cb9956d53872380e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-02 03:05:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/af9881a4c5bb4dc782db72d71d13aeda\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a11da3a5b25441cf939bdd6b0ca4eae0 : FAILURE in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff47834256624282b3834cbfbe6ebba6 : FAILURE in 2m 45s","accounts_in_message":[],"_revision_number":2},{"id":"5d7284dfcac29ac1965eafe996ac6862dde3f22e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-09-02 06:18:08.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"900bfb79b2ca4fe7732dec834c0076fd7ffad20c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-02 06:24:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3f886d8cbb14847b6686f8f0d6f2921\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/444a1bcdbfcf473dbb3477a512e9cafc : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56e4a48e03814d838ff98d0c442e7ce3 : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":3},{"id":"cad741551e749ed8d68c3af13816cd30c7a5556f","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2023-09-07 02:26:06.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"223762d0101229a8397bc1b8b29b28c2ee4d8b47","author":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"date":"2023-09-07 04:15:35.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8c8f59829046a89710707b378f5c257e20a2ea1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 04:15:52.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"73afeee11430aec8f305e9191f8d575fa349db17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 04:25:19.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d24b3afc8784888a15ca1c921c5e2ed\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16d1af68cb7d411cb3429f907759eed6 : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee8216359a964f1bba5372ce909491b7 : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":3},{"id":"542ee2c5a327ae532b91958c07d60dc96b6a65d7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 04:25:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"ba1595d1fff63a090e39ab263217190c30c1f3a4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 04:26:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89e8f1e2478b4bb68376fe5fa85b1f1c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b98a0d3a4aae45cc8611c357f761cc40 : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c9c82d4f3e36a1164b195d9f690c7e536cebbc9a","revisions":{"222ca712b21d8659fd52f19dec806fb1fbeed4ca":{"kind":"REWORK","_number":1,"created":"2021-02-27 15:07:11.000000000","uploader":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"ref":"refs/changes/64/777864/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/i18n","ref":"refs/changes/64/777864/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/i18n refs/changes/64/777864/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/i18n refs/changes/64/777864/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/i18n refs/changes/64/777864/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/i18n refs/changes/64/777864/1"}}},"commit":{"parents":[{"commit":"27a98e3ec4ae1ed505efad8aa2be9ab4e4be2cbd","subject":"Document on renaming Chinese locales in horizon","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/27a98e3ec4ae1ed505efad8aa2be9ab4e4be2cbd"}]}],"author":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2021-02-27 15:05:42.000000000","tz":540},"committer":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2021-02-27 15:06:56.000000000","tz":540},"subject":"Sync translation_team.yaml with Zanata","message":"Sync translation_team.yaml with Zanata\n\nThis list is generated by running \"tox -e zanata-users-sync\"\ncommand to sync translation members up-to-date similar as\nI5a309564fae7229fdc097715bdbb1ce85d109753\n\nChange-Id: Ica09d0348863aca9822eef7d9978a36b2ff3aac6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/222ca712b21d8659fd52f19dec806fb1fbeed4ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/222ca712b21d8659fd52f19dec806fb1fbeed4ca"}]},"branch":"refs/heads/master"},"f7aaaaecf24c1a1f619320620bbc1532a519865a":{"kind":"REWORK","_number":2,"created":"2023-09-02 02:53:26.000000000","uploader":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"ref":"refs/changes/64/777864/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/i18n","ref":"refs/changes/64/777864/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/i18n refs/changes/64/777864/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/i18n refs/changes/64/777864/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/i18n refs/changes/64/777864/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/i18n refs/changes/64/777864/2"}}},"commit":{"parents":[{"commit":"4e8bc2dd50a595da5855463015d79d125b800edc","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/4e8bc2dd50a595da5855463015d79d125b800edc"}]}],"author":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2021-02-27 15:05:42.000000000","tz":540},"committer":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2023-09-02 02:53:16.000000000","tz":540},"subject":"Sync translation_team.yaml with Zanata","message":"Sync translation_team.yaml with Zanata\n\nThis list is generated by running \"tox -e zanata-users-sync\"\ncommand to sync translation members up-to-date similar as\nI5a309564fae7229fdc097715bdbb1ce85d109753\n\nChange-Id: Ica09d0348863aca9822eef7d9978a36b2ff3aac6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/f7aaaaecf24c1a1f619320620bbc1532a519865a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/f7aaaaecf24c1a1f619320620bbc1532a519865a"}]},"branch":"refs/heads/master"},"c9c82d4f3e36a1164b195d9f690c7e536cebbc9a":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-09-02 06:18:08.000000000","uploader":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"ref":"refs/changes/64/777864/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/i18n","ref":"refs/changes/64/777864/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/i18n refs/changes/64/777864/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/i18n refs/changes/64/777864/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/i18n refs/changes/64/777864/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/i18n refs/changes/64/777864/3"}}},"commit":{"parents":[{"commit":"9b939e1b404a96a70961b6e1099e5f0b291af4da","subject":"Update tox.ini to work with tox 4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/9b939e1b404a96a70961b6e1099e5f0b291af4da"}]}],"author":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2021-02-27 15:05:42.000000000","tz":540},"committer":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2023-09-02 06:17:49.000000000","tz":540},"subject":"Sync translation_team.yaml with Zanata","message":"Sync translation_team.yaml with Zanata\n\nThis list is generated by running \"tox -e zanata-users-sync\"\ncommand to sync translation members up-to-date similar as\nI5a309564fae7229fdc097715bdbb1ce85d109753\n\nChange-Id: Ica09d0348863aca9822eef7d9978a36b2ff3aac6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/c9c82d4f3e36a1164b195d9f690c7e536cebbc9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/c9c82d4f3e36a1164b195d9f690c7e536cebbc9a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
